)]}'
{"id":"openstack%2Fkayobe~655706","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I16b523285eb34c1cd0eaf920aa4512d0ae1b4f05","project":"openstack/kayobe","branch":"stable/rocky","topic":"opendev-stable-rocky","hashtags":[],"change_id":"I16b523285eb34c1cd0eaf920aa4512d0ae1b4f05","subject":"Update git.openstack.org to opendev.org","status":"MERGED","created":"2019-04-25 14:25:54.000000000","updated":"2019-04-26 17:57:41.000000000","submitted":"2019-04-26 17:57:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655706-1556301461516-fe66b0bc","meta_rev_id":"493a7f1a80d6141b20a58807d4ff39a283e078d7","_number":655706,"virtual_id_number":655706,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-04-26 17:57:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-04-26 12:36:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-04-26 14:22:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-26 12:36:40.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-04-26 14:22:00.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-04-26 17:57:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47e501d3d12ee6910f593aee3958b45646ae512a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-25 14:25:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78772b43b6b95a18302011406366a314d87cb96b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-25 14:37:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"32ef182366b363f54e938aba039b2968c2d54ad0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-25 15:14:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a417644017026ec6d5f46a7e2d7b275a644d5a9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 16:18:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/06/655706/1/check/test-release-openstack/29af5a4/ : SUCCESS in 3m 22s\n- build-openstack-sphinx-docs http://logs.openstack.org/06/655706/1/check/build-openstack-sphinx-docs/55bacb9/html/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/06/655706/1/check/openstack-tox-docs/a689fe0/html/ : SUCCESS in 4m 30s\n- openstack-tox-cover http://logs.openstack.org/06/655706/1/check/openstack-tox-cover/4f9b3b5/cover/ : SUCCESS in 5m 02s\n- openstack-tox-py35 http://logs.openstack.org/06/655706/1/check/openstack-tox-py35/7883613/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/06/655706/1/check/openstack-tox-pep8/9e5e546/ : SUCCESS in 6m 30s\n- openstack-tox-py27 http://logs.openstack.org/06/655706/1/check/openstack-tox-py27/4f0e079/ : SUCCESS in 6m 15s\n- build-openstack-releasenotes http://logs.openstack.org/06/655706/1/check/build-openstack-releasenotes/b3e42a3/html/ : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/06/655706/1/check/kayobe-tox-ansible-syntax/8efa81b/ : SUCCESS in 5m 38s\n- kayobe-tox-ansible http://logs.openstack.org/06/655706/1/check/kayobe-tox-ansible/095f5d1/ : SUCCESS in 9m 14s\n- kayobe-tox-molecule http://logs.openstack.org/06/655706/1/check/kayobe-tox-molecule/2c503e5/ : SUCCESS in 30m 49s\n- kayobe-overcloud-centos http://logs.openstack.org/06/655706/1/check/kayobe-overcloud-centos/27b766e/ : SUCCESS in 58m 12s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/06/655706/1/check/kayobe-overcloud-upgrade-centos/0e3ef35/ : SUCCESS in 1h 42m 51s\n- kayobe-seed-centos http://logs.openstack.org/06/655706/1/check/kayobe-seed-centos/2f9b24d/ : SUCCESS in 20m 15s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/06/655706/1/check/kayobe-seed-upgrade-centos/1ec6816/ : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bc568d58b2abd79b3161267ade224f357612f3ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 16:18:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"105e7d5f655fc32318635a5c8ad5ca37683a33fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 17:29:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/06/655706/1/gate/test-release-openstack/8f53222/ : SUCCESS in 3m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/06/655706/1/gate/build-openstack-sphinx-docs/dab1213/html/ : SUCCESS in 5m 24s\n- openstack-tox-docs http://logs.openstack.org/06/655706/1/gate/openstack-tox-docs/f91c455/html/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/06/655706/1/gate/openstack-tox-py35/53d5c24/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/06/655706/1/gate/openstack-tox-pep8/06cbfc2/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/06/655706/1/gate/openstack-tox-py27/a049901/ : SUCCESS in 4m 42s\n- build-openstack-releasenotes http://logs.openstack.org/06/655706/1/gate/build-openstack-releasenotes/7b747e9/html/ : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/06/655706/1/gate/kayobe-tox-ansible-syntax/8e89290/ : SUCCESS in 5m 43s\n- kayobe-tox-ansible http://logs.openstack.org/06/655706/1/gate/kayobe-tox-ansible/06b8426/ : SUCCESS in 10m 36s\n- kayobe-tox-molecule http://logs.openstack.org/06/655706/1/gate/kayobe-tox-molecule/c983e5b/ : SUCCESS in 33m 07s\n- kayobe-overcloud-centos http://logs.openstack.org/06/655706/1/gate/kayobe-overcloud-centos/0743934/ : SUCCESS in 1h 07m 49s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/06/655706/1/gate/kayobe-overcloud-upgrade-centos/5dc09f4/ : FAILURE in 58m 54s\n- kayobe-seed-centos http://logs.openstack.org/06/655706/1/gate/kayobe-seed-centos/e3227df/ : SUCCESS in 23m 30s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/06/655706/1/gate/kayobe-seed-upgrade-centos/f3867d2/ : SUCCESS in 41m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c59fab0bd839a5ce650f09d921870fcb9796dfe5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-25 17:31:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6aa838d5a09882aea5aeaa14d8918c61633926a9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-25 17:57:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d7cc69c20e4125e875d6133ea4662fc14d28ba7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 19:51:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/06/655706/2/check/test-release-openstack/73d7d55/ : SUCCESS in 3m 47s\n- build-openstack-sphinx-docs http://logs.openstack.org/06/655706/2/check/build-openstack-sphinx-docs/6c7f272/html/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/06/655706/2/check/openstack-tox-docs/2e0aba6/html/ : SUCCESS in 4m 39s\n- openstack-tox-cover http://logs.openstack.org/06/655706/2/check/openstack-tox-cover/97f37e2/cover/ : SUCCESS in 4m 51s\n- openstack-tox-py35 http://logs.openstack.org/06/655706/2/check/openstack-tox-py35/ae00f8c/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/06/655706/2/check/openstack-tox-pep8/57ab3b7/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/06/655706/2/check/openstack-tox-py27/f599099/ : SUCCESS in 4m 47s\n- build-openstack-releasenotes http://logs.openstack.org/06/655706/2/check/build-openstack-releasenotes/0386dbe/html/ : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/06/655706/2/check/kayobe-tox-ansible-syntax/effb4e1/ : SUCCESS in 5m 24s\n- kayobe-tox-ansible http://logs.openstack.org/06/655706/2/check/kayobe-tox-ansible/971a203/ : SUCCESS in 8m 36s\n- kayobe-tox-molecule http://logs.openstack.org/06/655706/2/check/kayobe-tox-molecule/f7b5bc9/ : SUCCESS in 29m 14s\n- kayobe-overcloud-centos http://logs.openstack.org/06/655706/2/check/kayobe-overcloud-centos/98809f4/ : SUCCESS in 54m 57s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/06/655706/2/check/kayobe-overcloud-upgrade-centos/d89cc83/ : SUCCESS in 1h 32m 34s\n- kayobe-seed-centos http://logs.openstack.org/06/655706/2/check/kayobe-seed-centos/0c0d176/ : SUCCESS in 23m 03s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/06/655706/2/check/kayobe-seed-upgrade-centos/a22f094/ : SUCCESS in 41m 43s","accounts_in_message":[],"_revision_number":2},{"id":"cbf19350f0dc058f4f13963e03c0a680a93b2ba7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-26 12:36:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7ca606acc5ef4e2fe0e9991962ab6b6017237c32","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-04-26 14:22:00.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"175e9bda160825f21bfe067118ae5b66b63e698a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 14:22:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"026e3ee80fc8e3527b4fae3d5fbe877c2ad4ca32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 16:05:33.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ce08862aa73462ee95101526f90e0a7c8f5c2f60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 17:57:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"98c85c6768568dbe38fc3322fc11d759f9845a23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 17:57:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/06/655706/2/gate/test-release-openstack/d0b50dc/ : SUCCESS in 4m 12s\n- build-openstack-sphinx-docs http://logs.openstack.org/06/655706/2/gate/build-openstack-sphinx-docs/5b7a192/html/ : SUCCESS in 4m 51s\n- openstack-tox-docs http://logs.openstack.org/06/655706/2/gate/openstack-tox-docs/2c5e7d7/html/ : SUCCESS in 4m 16s\n- openstack-tox-py35 http://logs.openstack.org/06/655706/2/gate/openstack-tox-py35/6a805df/ : SUCCESS in 7m 18s\n- openstack-tox-pep8 http://logs.openstack.org/06/655706/2/gate/openstack-tox-pep8/53dbb59/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/06/655706/2/gate/openstack-tox-py27/2413beb/ : SUCCESS in 4m 33s\n- build-openstack-releasenotes http://logs.openstack.org/06/655706/2/gate/build-openstack-releasenotes/dd3c8be/html/ : SUCCESS in 3m 48s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/06/655706/2/gate/kayobe-tox-ansible-syntax/c5359d5/ : SUCCESS in 5m 31s\n- kayobe-tox-ansible http://logs.openstack.org/06/655706/2/gate/kayobe-tox-ansible/c96aefe/ : SUCCESS in 9m 32s\n- kayobe-tox-molecule http://logs.openstack.org/06/655706/2/gate/kayobe-tox-molecule/3511e6f/ : SUCCESS in 25m 37s\n- kayobe-overcloud-centos http://logs.openstack.org/06/655706/2/gate/kayobe-overcloud-centos/ec2b083/ : SUCCESS in 59m 53s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/06/655706/2/gate/kayobe-overcloud-upgrade-centos/e58e580/ : SUCCESS in 1h 31m 47s\n- kayobe-seed-centos http://logs.openstack.org/06/655706/2/gate/kayobe-seed-centos/fc42ca3/ : SUCCESS in 20m 58s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/06/655706/2/gate/kayobe-seed-upgrade-centos/7721019/ : SUCCESS in 48m 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"517008741112b75c8f25561d896c39757885bb76","revisions":{"bf585c64f46f898991d379638c1867337bd31af3":{"kind":"REWORK","_number":1,"created":"2019-04-25 14:25:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/655706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/655706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/655706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/655706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/655706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/655706/1"}}},"commit":{"parents":[{"commit":"7da955f006dd53a6e7b4e1f0fc0cbdf04673ef34","subject":"Always remove temporary file containing passwords","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7da955f006dd53a6e7b4e1f0fc0cbdf04673ef34"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-25 14:25:00.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-25 14:25:36.000000000","tz":60},"subject":"Update git.openstack.org to opendev.org","message":"Update git.openstack.org to opendev.org\n\nThe project infrastructure is now hosted at opendev.org.\n\nKayobe projects have been moved from the openstack/ namespace to x/.\n\nChange-Id: I16b523285eb34c1cd0eaf920aa4512d0ae1b4f05\n(cherry picked from commit 5b9f85d715dd18b7a748bbee52980452a432a6d0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf585c64f46f898991d379638c1867337bd31af3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf585c64f46f898991d379638c1867337bd31af3"}]},"branch":"refs/heads/stable/rocky"},"517008741112b75c8f25561d896c39757885bb76":{"kind":"REWORK","_number":2,"created":"2019-04-25 17:57:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/655706/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/655706/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/655706/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/655706/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/655706/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/655706/2"}}},"commit":{"parents":[{"commit":"7da955f006dd53a6e7b4e1f0fc0cbdf04673ef34","subject":"Always remove temporary file containing passwords","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7da955f006dd53a6e7b4e1f0fc0cbdf04673ef34"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-25 14:25:00.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-25 17:57:25.000000000","tz":60},"subject":"Update git.openstack.org to opendev.org","message":"Update git.openstack.org to opendev.org\n\nThe project infrastructure is now hosted at opendev.org.\n\nKayobe projects have been moved from the openstack/ namespace to x/.\n\nChange-Id: I16b523285eb34c1cd0eaf920aa4512d0ae1b4f05\n(cherry picked from commit 5b9f85d715dd18b7a748bbee52980452a432a6d0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/517008741112b75c8f25561d896c39757885bb76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/517008741112b75c8f25561d896c39757885bb76"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
