)]}'
{"id":"openstack%2Fkayobe~655747","triplet_id":"openstack%2Fkayobe~stable%2Fqueens~I16b523285eb34c1cd0eaf920aa4512d0ae1b4f05","project":"openstack/kayobe","branch":"stable/queens","topic":"opendev-stable-queens","hashtags":[],"change_id":"I16b523285eb34c1cd0eaf920aa4512d0ae1b4f05","subject":"Update git.openstack.org to opendev.org","status":"MERGED","created":"2019-04-25 17:38:14.000000000","updated":"2019-05-09 18:55:25.000000000","submitted":"2019-05-09 18:55:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655747-1557428125901-866de8cb","meta_rev_id":"7e453c4604ed54abc4f20406d9e469924cdc4412","_number":655747,"virtual_id_number":655747,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"date":"2019-04-29 08:02:18.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-09 18:55:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-04-29 09:43:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-09 17:32:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-09 17:32:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-29 09:43:16.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-05-09 17:32:19.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-05-09 18:55:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d788259a38f87eeb2e968443b0f325916f9db691","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-25 17:38:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aef1b033ea1909d20688a65102943fae1ceef81a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-25 17:57:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"13b7bbd02e761a54ac9181ae0ff39efaefe5811f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 19:23:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/47/655747/2/check/test-release-openstack/64ad193/ : SUCCESS in 2m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/47/655747/2/check/build-openstack-sphinx-docs/1dc9224/html/ : SUCCESS in 4m 16s\n- openstack-tox-cover http://logs.openstack.org/47/655747/2/check/openstack-tox-cover/4ccfa22/cover/ : SUCCESS in 8m 05s\n- openstack-tox-py35 http://logs.openstack.org/47/655747/2/check/openstack-tox-py35/139e4d6/ : SUCCESS in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/47/655747/2/check/openstack-tox-pep8/716aa15/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/47/655747/2/check/openstack-tox-py27/4ab3b6e/ : SUCCESS in 4m 36s\n- build-openstack-releasenotes http://logs.openstack.org/47/655747/2/check/build-openstack-releasenotes/55d3f17/html/ : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/47/655747/2/check/kayobe-tox-ansible-syntax/f44c315/ : SUCCESS in 4m 56s\n- kayobe-tox-ansible http://logs.openstack.org/47/655747/2/check/kayobe-tox-ansible/cf61854/ : SUCCESS in 8m 57s\n- kayobe-tox-molecule http://logs.openstack.org/47/655747/2/check/kayobe-tox-molecule/04e7770/ : SUCCESS in 29m 07s\n- kayobe-overcloud-centos http://logs.openstack.org/47/655747/2/check/kayobe-overcloud-centos/21d749d/ : SUCCESS in 45m 35s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/47/655747/2/check/kayobe-overcloud-upgrade-centos/1239d82/ : FAILURE in 33m 02s\n- kayobe-seed-centos http://logs.openstack.org/47/655747/2/check/kayobe-seed-centos/cf9a38f/ : SUCCESS in 25m 48s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/47/655747/2/check/kayobe-seed-upgrade-centos/ef16937/ : SUCCESS in 34m 37s","accounts_in_message":[],"_revision_number":2},{"id":"c356940ddab30341a0bac966f36d1fb45dd0a78f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-26 14:14:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"99f4cbb1fe24f7c9a344e3c8a9caefd5065d4508","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 18:18:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/47/655747/2/check/test-release-openstack/ed352fb/ : SUCCESS in 4m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/47/655747/2/check/build-openstack-sphinx-docs/16ec294/html/ : SUCCESS in 4m 36s\n- openstack-tox-cover http://logs.openstack.org/47/655747/2/check/openstack-tox-cover/5257332/cover/ : SUCCESS in 5m 35s\n- openstack-tox-py35 http://logs.openstack.org/47/655747/2/check/openstack-tox-py35/f671b2b/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/47/655747/2/check/openstack-tox-pep8/8084ef5/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/47/655747/2/check/openstack-tox-py27/4002889/ : SUCCESS in 6m 41s\n- build-openstack-releasenotes http://logs.openstack.org/47/655747/2/check/build-openstack-releasenotes/3fc8cdb/html/ : SUCCESS in 5m 10s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/47/655747/2/check/kayobe-tox-ansible-syntax/70b2c8b/ : SUCCESS in 6m 04s\n- kayobe-tox-ansible http://logs.openstack.org/47/655747/2/check/kayobe-tox-ansible/b97bf96/ : SUCCESS in 9m 24s\n- kayobe-tox-molecule http://logs.openstack.org/47/655747/2/check/kayobe-tox-molecule/354b4c0/ : SUCCESS in 30m 45s\n- kayobe-overcloud-centos http://logs.openstack.org/47/655747/2/check/kayobe-overcloud-centos/d4d3589/ : SUCCESS in 1h 21m 43s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/47/655747/2/check/kayobe-overcloud-upgrade-centos/bb53514/ : TIMED_OUT in 2h 01m 08s\n- kayobe-seed-centos http://logs.openstack.org/47/655747/2/check/kayobe-seed-centos/7b515b9/ : SUCCESS in 26m 08s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/47/655747/2/check/kayobe-seed-upgrade-centos/10abca9/ : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":2},{"id":"d349effeb5297698d97215138f6a085da398034d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-29 08:02:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0d1c8e230fef1a328a2bd670aa27658622497f7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 09:29:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/655747/2/check/test-release-openstack/fdb04e3/ : SUCCESS in 3m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/47/655747/2/check/build-openstack-sphinx-docs/5899c02/html/ : SUCCESS in 4m 24s\n- openstack-tox-cover http://logs.openstack.org/47/655747/2/check/openstack-tox-cover/2d96f84/cover/ : SUCCESS in 4m 58s\n- openstack-tox-py35 http://logs.openstack.org/47/655747/2/check/openstack-tox-py35/8fa0787/ : SUCCESS in 5m 53s\n- openstack-tox-pep8 http://logs.openstack.org/47/655747/2/check/openstack-tox-pep8/50fab84/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/47/655747/2/check/openstack-tox-py27/c3b4b0a/ : SUCCESS in 4m 37s\n- build-openstack-releasenotes http://logs.openstack.org/47/655747/2/check/build-openstack-releasenotes/9933903/html/ : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/47/655747/2/check/kayobe-tox-ansible-syntax/dbb5ed7/ : SUCCESS in 5m 15s\n- kayobe-tox-ansible http://logs.openstack.org/47/655747/2/check/kayobe-tox-ansible/1aa4ccd/ : SUCCESS in 8m 20s\n- kayobe-tox-molecule http://logs.openstack.org/47/655747/2/check/kayobe-tox-molecule/facbe36/ : SUCCESS in 30m 00s\n- kayobe-overcloud-centos http://logs.openstack.org/47/655747/2/check/kayobe-overcloud-centos/8b83ee1/ : SUCCESS in 46m 59s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/47/655747/2/check/kayobe-overcloud-upgrade-centos/c03e0d6/ : SUCCESS in 1h 21m 56s\n- kayobe-seed-centos http://logs.openstack.org/47/655747/2/check/kayobe-seed-centos/5441f09/ : SUCCESS in 25m 19s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/47/655747/2/check/kayobe-seed-upgrade-centos/21b336c/ : SUCCESS in 41m 15s","accounts_in_message":[],"_revision_number":2},{"id":"7f9296cc7f46b1e77fbc99f8df957ca5b427ff27","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-29 09:43:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8bac5e6e6feb3d2a53659e94c6ddba1c0abbfa0a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-09 17:32:06.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good, although we may have to change the upper constraints links again, see Tony Breeds\u0027 comment at https://review.opendev.org/#/c/657209/1/tox.ini","accounts_in_message":[],"_revision_number":2},{"id":"6a866e933f1d24add5d6353fad2ff30ac4deaee3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-09 17:32:19.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ab7ba66bf1dcd3ff4fee0b168ed9427ec90bdef6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 17:32:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"10a721bb3679cede3b9c2ea2f36150f288a0b5e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 18:55:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"69a7673d9d53f9a97e26f32fa995a79f0ce8ebfa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 18:55:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/47/655747/2/gate/test-release-openstack/10aad0b/ : SUCCESS in 3m 50s\n- build-openstack-sphinx-docs http://logs.openstack.org/47/655747/2/gate/build-openstack-sphinx-docs/fa6cfc7/html/ : SUCCESS in 6m 35s\n- openstack-tox-py35 http://logs.openstack.org/47/655747/2/gate/openstack-tox-py35/4d43e89/ : SUCCESS in 7m 36s\n- openstack-tox-pep8 http://logs.openstack.org/47/655747/2/gate/openstack-tox-pep8/2b09210/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/47/655747/2/gate/openstack-tox-py27/6d65c47/ : SUCCESS in 5m 01s\n- build-openstack-releasenotes http://logs.openstack.org/47/655747/2/gate/build-openstack-releasenotes/2fd8200/html/ : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/47/655747/2/gate/kayobe-tox-ansible-syntax/1b738e7/ : SUCCESS in 8m 05s\n- kayobe-tox-ansible http://logs.openstack.org/47/655747/2/gate/kayobe-tox-ansible/29f6274/ : SUCCESS in 11m 35s\n- kayobe-tox-molecule http://logs.openstack.org/47/655747/2/gate/kayobe-tox-molecule/685009a/ : SUCCESS in 32m 26s\n- kayobe-overcloud-centos http://logs.openstack.org/47/655747/2/gate/kayobe-overcloud-centos/d4f9622/ : SUCCESS in 45m 14s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/47/655747/2/gate/kayobe-overcloud-upgrade-centos/b687d54/ : SUCCESS in 1h 16m 06s\n- kayobe-seed-centos http://logs.openstack.org/47/655747/2/gate/kayobe-seed-centos/3eec763/ : SUCCESS in 16m 21s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/47/655747/2/gate/kayobe-seed-upgrade-centos/dba8932/ : SUCCESS in 24m 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2e9d5cfeb006b120b2cf678cc9d086822716e32d","revisions":{"20714fd709a0aa085ff3067cca62a56d65252fd4":{"kind":"REWORK","_number":1,"created":"2019-04-25 17:38:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/47/655747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/655747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/655747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/655747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/655747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/655747/1"}}},"commit":{"parents":[{"commit":"9d4da943b3297d9616c6ecbb9b837b23b45050a2","subject":"Always remove temporary file containing passwords","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d4da943b3297d9616c6ecbb9b837b23b45050a2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-25 14:25:00.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-25 17:38:01.000000000","tz":60},"subject":"Update git.openstack.org to opendev.org","message":"Update git.openstack.org to opendev.org\n\nThe project infrastructure is now hosted at opendev.org.\n\nKayobe projects have been moved from the openstack/ namespace to x/.\n\nChange-Id: I16b523285eb34c1cd0eaf920aa4512d0ae1b4f05\n(cherry picked from commit 5b9f85d715dd18b7a748bbee52980452a432a6d0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20714fd709a0aa085ff3067cca62a56d65252fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20714fd709a0aa085ff3067cca62a56d65252fd4"}]},"branch":"refs/heads/stable/queens"},"2e9d5cfeb006b120b2cf678cc9d086822716e32d":{"kind":"REWORK","_number":2,"created":"2019-04-25 17:57:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/47/655747/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/655747/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/655747/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/655747/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/655747/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/655747/2"}}},"commit":{"parents":[{"commit":"9d4da943b3297d9616c6ecbb9b837b23b45050a2","subject":"Always remove temporary file containing passwords","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d4da943b3297d9616c6ecbb9b837b23b45050a2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-25 14:25:00.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-25 17:56:55.000000000","tz":60},"subject":"Update git.openstack.org to opendev.org","message":"Update git.openstack.org to opendev.org\n\nThe project infrastructure is now hosted at opendev.org.\n\nKayobe projects have been moved from the openstack/ namespace to x/.\n\nChange-Id: I16b523285eb34c1cd0eaf920aa4512d0ae1b4f05\n(cherry picked from commit 5b9f85d715dd18b7a748bbee52980452a432a6d0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e9d5cfeb006b120b2cf678cc9d086822716e32d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e9d5cfeb006b120b2cf678cc9d086822716e32d"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
