)]}'
{"id":"openstack%2Fkayobe~656712","triplet_id":"openstack%2Fkayobe~master~I38fce0a6c155e4b7d2ff7a2535f18e8adc2566e9","project":"openstack/kayobe","branch":"master","topic":"bug/2005546","hashtags":[],"change_id":"I38fce0a6c155e4b7d2ff7a2535f18e8adc2566e9","subject":"Fix setup.cfg syntax error, not allowed comment in files section","status":"MERGED","created":"2019-05-01 21:40:01.000000000","updated":"2019-05-02 18:09:44.000000000","submitted":"2019-05-02 17:46:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656712-1556819203645-0b0d46a4","meta_rev_id":"a7872fe62d45463734259ace3fc1f0c14108759a","_number":656712,"virtual_id_number":656712,"owner":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":27875,"name":"Gábor Szitás","email":"gabor.szitas@nokia.com","username":"szitas"},{"value":0,"date":"2019-05-02 03:35:39.000000000","_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-02 17:46:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-05-01 21:46:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-05-01 21:46:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27875,"name":"Gábor Szitás","email":"gabor.szitas@nokia.com","username":"szitas"},{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":2,"date":"2019-05-02 17:28:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":27875,"name":"Gábor Szitás","email":"gabor.szitas@nokia.com","username":"szitas"},{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":1,"date":"2019-05-02 17:28:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":27875,"name":"Gábor Szitás","email":"gabor.szitas@nokia.com","username":"szitas"},{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":27875,"name":"Gábor Szitás","email":"gabor.szitas@nokia.com","username":"szitas"},{"value":0,"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27875,"name":"Gábor Szitás","email":"gabor.szitas@nokia.com","username":"szitas"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-01 21:46:21.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-05-01 21:46:56.000000000","updated_by":{"_account_id":27875,"name":"Gábor Szitás","email":"gabor.szitas@nokia.com","username":"szitas"},"reviewer":{"_account_id":27875,"name":"Gábor Szitás","email":"gabor.szitas@nokia.com","username":"szitas"},"state":"REVIEWER"},{"updated":"2019-05-02 17:28:13.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-02 17:46:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d3cdcffff882adea23bd4f61c5dcee358b526bbe","author":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"date":"2019-05-01 21:40:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b869ae81350d71618c4fc9f82159f81a2cae034","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-05-01 21:46:21.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks for the fix. I wonder why we I haven\u0027t hit this before.","accounts_in_message":[],"_revision_number":1},{"id":"e599bae220f9fa6392e24ecca4ba4d497a006031","author":{"_account_id":27875,"name":"Gábor Szitás","email":"gabor.szitas@nokia.com","username":"szitas"},"date":"2019-05-01 21:46:56.000000000","message":"Patch Set 1: Code-Review+1\n\nLNTM","accounts_in_message":[],"_revision_number":1},{"id":"d6762ba61896ce62fcafba9129f4d5d1720b6d42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 21:50:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/12/656712/1/check/test-release-openstack/2068278/ : SUCCESS in 3m 37s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/656712/1/check/build-openstack-sphinx-docs/31c8d63/html/ : SUCCESS in 6m 17s\n- openstack-tox-docs http://logs.openstack.org/12/656712/1/check/openstack-tox-docs/eeefd52/html/ : SUCCESS in 4m 01s\n- openstack-tox-cover http://logs.openstack.org/12/656712/1/check/openstack-tox-cover/cc41e16/cover/ : SUCCESS in 5m 59s\n- openstack-tox-py35 http://logs.openstack.org/12/656712/1/check/openstack-tox-py35/625adaf/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/12/656712/1/check/openstack-tox-py36/a8591da/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/12/656712/1/check/openstack-tox-pep8/36d5b95/ : SUCCESS in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/12/656712/1/check/openstack-tox-py27/484144b/ : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"cdebaf9aa6c218d493440d9cf497a68101cef177","author":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"date":"2019-05-02 03:35:39.000000000","message":"Patch Set 1:\n\n\u003e Thanks for the fix. I wonder why we I haven\u0027t hit this before.\n\nSo I have setuptools-41.0.1 and on:\n1. CentOS 7.5.1804 it fails\n2. Fedora 29 it works","accounts_in_message":[],"_revision_number":1},{"id":"0781a5029ac3d9e680ae1d2cecce1bea1e656a32","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-02 17:28:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks for the fix!\n\nOdd that this doesn\u0027t fail with pip install, but hey.","accounts_in_message":[],"_revision_number":1},{"id":"02b2b21e791e44b24b315224a7ee01f7efbe1ab7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 17:28:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e29ee449656ada7f539734a43d4552424117b26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 17:46:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6dac129b4d8650da039c4e1f7520a1e480def814","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 17:46:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/12/656712/1/gate/test-release-openstack/653fb25/ : SUCCESS in 3m 34s\n- build-openstack-sphinx-docs http://logs.openstack.org/12/656712/1/gate/build-openstack-sphinx-docs/4ae18b8/html/ : SUCCESS in 12m 21s\n- openstack-tox-docs http://logs.openstack.org/12/656712/1/gate/openstack-tox-docs/0431a07/html/ : SUCCESS in 4m 15s\n- openstack-tox-py35 http://logs.openstack.org/12/656712/1/gate/openstack-tox-py35/9bb50b6/ : SUCCESS in 6m 27s\n- openstack-tox-py36 http://logs.openstack.org/12/656712/1/gate/openstack-tox-py36/50c5b2e/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/12/656712/1/gate/openstack-tox-pep8/148b1b9/ : SUCCESS in 13m 34s\n- openstack-tox-py27 http://logs.openstack.org/12/656712/1/gate/openstack-tox-py27/b088a64/ : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"566203c6588c4e31a7e791b58008b1025cf65f1d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-02 18:09:44.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit dbf69664dbc5c0a64f109cb20be13103691e4d24","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e40f8498ae223ba1e71a9e2a3899ef66c822efe","revisions":{"1e40f8498ae223ba1e71a9e2a3899ef66c822efe":{"kind":"REWORK","_number":1,"created":"2019-05-01 21:40:01.000000000","uploader":{"_account_id":11444,"name":"Radoslaw Smigielski","display_name":"Radek","email":"rsmigiel@redhat.com","username":"radek"},"ref":"refs/changes/12/656712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/12/656712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/656712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/656712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/656712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/12/656712/1"}}},"commit":{"parents":[{"commit":"272709d54aff0ff731feb227ea51627c18ede423","subject":"Merge \"Add openstack_release and openstack_branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/272709d54aff0ff731feb227ea51627c18ede423"}]}],"author":{"name":"Radoslaw Smigielski","email":"radoslaw.smigielski@nokia.com","date":"2019-05-01 21:28:15.000000000","tz":-360},"committer":{"name":"Radoslaw Smigielski","email":"radoslaw.smigielski@nokia.com","date":"2019-05-01 21:39:44.000000000","tz":-360},"subject":"Fix setup.cfg syntax error, not allowed comment in files section","message":"Fix setup.cfg syntax error, not allowed comment in files section\n\nIn [files] section of setup.cfg there is not allowed comment, which\nbreaks installation with setup.py.\n\nChange-Id: I38fce0a6c155e4b7d2ff7a2535f18e8adc2566e9\nStory: 2005546\nTask: 30692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e40f8498ae223ba1e71a9e2a3899ef66c822efe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e40f8498ae223ba1e71a9e2a3899ef66c822efe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
