)]}'
{"id":"openstack%2Fkayobe~660087","triplet_id":"openstack%2Fkayobe~master~I2495c97455e40cf1f0573cee6ae845da14d762b5","project":"openstack/kayobe","branch":"master","topic":"docs-supported-centos","hashtags":[],"change_id":"I2495c97455e40cf1f0573cee6ae845da14d762b5","subject":"Update supported CentOS version","status":"MERGED","created":"2019-05-20 10:45:44.000000000","updated":"2019-05-20 11:41:26.000000000","submitted":"2019-05-20 11:41:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660087-1558352486438-4f1731f4","meta_rev_id":"91af82575543ad1d22c21fbaebb331c428763663","_number":660087,"virtual_id_number":660087,"owner":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-20 11:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-05-20 11:19:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-20 11:15:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-05-20 11:19:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 11:15:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-05-20 11:19:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-20 11:41:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4aa64f405163ef2d1ee091095373d9787206536","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-05-20 10:45:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98fce574c442e96d547a8ae20dc35494ea532c4a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-20 10:57:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nI think I wanted to do the same change shortly after I started reading the Kayobe documentation :-)\n\nThe change looks good, I just have minor comments about the commit message.","accounts_in_message":[],"_revision_number":1},{"id":"63cf0fee0db1cac83de57ae1c1115071deeb6d9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:08:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/87/660087/1/check/build-openstack-sphinx-docs/54fe7fd/html/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/87/660087/1/check/openstack-tox-docs/1d12a73/html/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/87/660087/1/check/openstack-tox-pep8/ab3e3c8/ : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9737d98f599956c954a740ad3b7b2674e56ac317","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-05-20 11:13:53.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d4fb00c3bfa408192a6d20bd5b505aace60b5722","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-20 11:15:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"211721c7ddc859c18ce0e3036335830f21a8cf32","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 11:19:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"7e9eb970500262a6f2a0b5acee328b2484b4c833","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:27:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/87/660087/2/check/build-openstack-sphinx-docs/29d0ad7/html/ : SUCCESS in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/87/660087/2/check/openstack-tox-docs/146e632/html/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/87/660087/2/check/openstack-tox-pep8/95b7caf/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":2},{"id":"1e85f5a2e83ae82747789cb59bd27c42be3670ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:27:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76eafc831921b4c118171cd6a666bda08d98dd4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:41:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/87/660087/2/gate/build-openstack-sphinx-docs/1994200/html/ : SUCCESS in 5m 24s\n- openstack-tox-docs http://logs.openstack.org/87/660087/2/gate/openstack-tox-docs/6fdcb3d/html/ : SUCCESS in 6m 50s\n- openstack-tox-pep8 http://logs.openstack.org/87/660087/2/gate/openstack-tox-pep8/675539c/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"a5082cc12b14caa05acddfe03e62a631c9935dd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 11:41:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac85dc6064b517c1c0e7b496ec7b47a1c3a78b38","revisions":{"e69687a9c7020c2b215242053855fcace757c060":{"kind":"REWORK","_number":1,"created":"2019-05-20 10:45:44.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/87/660087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/660087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/660087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/660087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/660087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/660087/1"}}},"commit":{"parents":[{"commit":"9f92afc38397d3a0573fc127c2050c75b9bf87e5","subject":"Fix table generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f92afc38397d3a0573fc127c2050c75b9bf87e5"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-05-20 10:44:54.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-05-20 10:44:54.000000000","tz":60},"subject":"Updated supported Centos version","message":"Updated supported Centos version\n\nChange-Id: I2495c97455e40cf1f0573cee6ae845da14d762b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e69687a9c7020c2b215242053855fcace757c060"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e69687a9c7020c2b215242053855fcace757c060"}]},"branch":"refs/heads/master"},"ac85dc6064b517c1c0e7b496ec7b47a1c3a78b38":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-20 11:13:53.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/87/660087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/660087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/660087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/660087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/660087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/660087/2"}}},"commit":{"parents":[{"commit":"9f92afc38397d3a0573fc127c2050c75b9bf87e5","subject":"Fix table generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f92afc38397d3a0573fc127c2050c75b9bf87e5"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-05-20 10:44:54.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-05-20 11:12:39.000000000","tz":60},"subject":"Update supported CentOS version","message":"Update supported CentOS version\n\nChange-Id: I2495c97455e40cf1f0573cee6ae845da14d762b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac85dc6064b517c1c0e7b496ec7b47a1c3a78b38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac85dc6064b517c1c0e7b496ec7b47a1c3a78b38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
