)]}'
{"id":"openstack%2Fkayobe~660398","triplet_id":"openstack%2Fkayobe~master~I4ddf077913ddc620c60c78ca491e70b2e8f59489","project":"openstack/kayobe","branch":"master","topic":"docs-prune","hashtags":[],"change_id":"I4ddf077913ddc620c60c78ca491e70b2e8f59489","subject":"Remove out of date statements","status":"MERGED","created":"2019-05-21 13:53:28.000000000","updated":"2019-05-21 16:36:42.000000000","submitted":"2019-05-21 16:36:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660398-1558456602944-f9d4f00a","meta_rev_id":"bc7132cfc60218cc4011b8000577de26bf2d1014","_number":660398,"virtual_id_number":660398,"owner":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-21 16:36:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-05-21 15:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-21 16:25:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-21 16:25:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 15:57:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-05-21 16:25:31.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-05-21 16:36:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e480be99b3421dd0a35f1f616b0da68d8c17d6b9","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-05-21 13:53:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99cff53379bf99b0af33ccc841259a418c562030","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-21 13:57:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"51d18bde0fe85599eb1711eeb498c9ac53e033af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 14:03:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/98/660398/1/check/test-release-openstack/a42da73/ : SUCCESS in 3m 07s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/660398/1/check/build-openstack-sphinx-docs/e45383e/html/ : SUCCESS in 4m 07s\n- openstack-tox-docs http://logs.openstack.org/98/660398/1/check/openstack-tox-docs/ea0be28/html/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/98/660398/1/check/openstack-tox-pep8/9ff6dbc/ : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d375abe7d524bfdd2fcfc7b997fa0c56c1562ebc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-21 15:51:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ba60504a77f5fb48d770dd42ade742c51a2cb24","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-05-21 15:56:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7ceb4718efd71cb605ef689aa540b32f2f36ba57","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-21 15:57:38.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"58130f218f8aa7b50fc822f95732abba9d64da18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 16:09:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/98/660398/2/check/test-release-openstack/a8006c2/ : SUCCESS in 3m 57s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/660398/2/check/build-openstack-sphinx-docs/fb1ea79/html/ : SUCCESS in 6m 25s\n- openstack-tox-docs http://logs.openstack.org/98/660398/2/check/openstack-tox-docs/e6f34bb/html/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/98/660398/2/check/openstack-tox-pep8/c989921/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"e41745203a441c016f8dd2c1cf6fb4d234cfe8c5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-21 16:23:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a7161a7d4a19044ae4132faef3d87e45e3fc344","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-21 16:25:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d223093e2bb700edba5b89302074ef90d065487b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-21 16:25:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"da540121429f2b1ef8e3ceac152cc637b9055e8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 16:25:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4ad6d719f929177d41b9d619cb7d97254801691e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 16:36:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"da553e808da5833643aca4faa4e01f5af95f241a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 16:36:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/98/660398/2/gate/test-release-openstack/dcad3a1/ : SUCCESS in 3m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/98/660398/2/gate/build-openstack-sphinx-docs/c08c769/html/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/98/660398/2/gate/openstack-tox-docs/310ddcc/html/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/98/660398/2/gate/openstack-tox-pep8/eab2439/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"863fe9e261cc555b40592c2d2022ee05d2de79f1","revisions":{"571ba8074d1dd327e3bfe9c4de8f12ef85470533":{"kind":"REWORK","_number":1,"created":"2019-05-21 13:53:28.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/98/660398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/660398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/660398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/660398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/660398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/660398/1"}}},"commit":{"parents":[{"commit":"bfc19d3ee6c3f1611a0a8e5812ca78cb1030f18d","subject":"Merge \"Multiple updates for Kayobe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bfc19d3ee6c3f1611a0a8e5812ca78cb1030f18d"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-05-21 13:53:17.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-05-21 13:53:17.000000000","tz":60},"subject":"Remove out of date statements","message":"Remove out of date statements\n\nChange-Id: I4ddf077913ddc620c60c78ca491e70b2e8f59489\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/571ba8074d1dd327e3bfe9c4de8f12ef85470533"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/571ba8074d1dd327e3bfe9c4de8f12ef85470533"}]},"branch":"refs/heads/master"},"863fe9e261cc555b40592c2d2022ee05d2de79f1":{"kind":"REWORK","_number":2,"created":"2019-05-21 15:56:06.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/98/660398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/660398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/660398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/660398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/660398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/660398/2"}}},"commit":{"parents":[{"commit":"bfc19d3ee6c3f1611a0a8e5812ca78cb1030f18d","subject":"Merge \"Multiple updates for Kayobe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bfc19d3ee6c3f1611a0a8e5812ca78cb1030f18d"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-05-21 13:53:17.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-05-21 15:55:47.000000000","tz":60},"subject":"Remove out of date statements","message":"Remove out of date statements\n\nChange-Id: I4ddf077913ddc620c60c78ca491e70b2e8f59489\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/863fe9e261cc555b40592c2d2022ee05d2de79f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/863fe9e261cc555b40592c2d2022ee05d2de79f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
