)]}'
{"id":"openstack%2Fkayobe~662485","triplet_id":"openstack%2Fkayobe~master~I00ee10bdc7f8c1b6ab8576115c757e27b6c2c397","project":"openstack/kayobe","branch":"master","topic":"blazar-custom-config","hashtags":[],"change_id":"I00ee10bdc7f8c1b6ab8576115c757e27b6c2c397","subject":"Add support for custom Blazar configuration","status":"MERGED","created":"2019-05-31 14:34:20.000000000","updated":"2019-06-24 16:16:31.000000000","submitted":"2019-06-24 16:16:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662485-1561392991020-230b38e8","meta_rev_id":"b0961eef1287f7f0953b234d6a8dc74d9390f975","_number":662485,"virtual_id_number":662485,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-24 16:16:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"recommended":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"all":[{"value":1,"date":"2019-06-13 07:19:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":2,"date":"2019-06-24 14:31:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-06-13 07:23:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-06-03 16:04:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2019-06-13 09:15:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-06-24 14:31:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 16:04:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-13 07:19:27.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-06-13 07:23:13.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-06-13 09:15:18.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-06-24 14:31:56.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-06-24 16:16:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b6c16c55284344f2fa416a2e0c1b7ddacba00cc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 14:34:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3a7daf27e8f023535b31e69fc7fe362286a2d9a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 14:34:40.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"22ba1e4a61ea97ff80bf0a213124eecd5394cd93","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 15:44:48.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"86b8fab89117d6fda22c45cdd0c25f490ad787da","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 15:55:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3280b57327c983002b6d68a98fb7eba5e2640c4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 17:45:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/85/662485/3/check/openstack-tox-docs/b532db0/html/ : SUCCESS in 4m 31s\n- openstack-tox-cover http://logs.openstack.org/85/662485/3/check/openstack-tox-cover/426d55d/cover/ : SUCCESS in 4m 56s\n- openstack-tox-py35 http://logs.openstack.org/85/662485/3/check/openstack-tox-py35/cdf8d13/ : SUCCESS in 4m 52s\n- openstack-tox-py36 http://logs.openstack.org/85/662485/3/check/openstack-tox-py36/88444ff/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/85/662485/3/check/openstack-tox-pep8/f2987eb/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/85/662485/3/check/openstack-tox-py27/26d4c74/ : SUCCESS in 5m 12s\n- build-openstack-releasenotes http://logs.openstack.org/85/662485/3/check/build-openstack-releasenotes/ac34039/html/ : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/85/662485/3/check/kayobe-tox-ansible-syntax/122b83b/ : SUCCESS in 5m 02s\n- kayobe-tox-ansible http://logs.openstack.org/85/662485/3/check/kayobe-tox-ansible/c642d16/ : SUCCESS in 9m 03s\n- kayobe-tox-molecule http://logs.openstack.org/85/662485/3/check/kayobe-tox-molecule/5469a65/ : SUCCESS in 32m 03s\n- kayobe-overcloud-centos http://logs.openstack.org/85/662485/3/check/kayobe-overcloud-centos/90465c8/ : SUCCESS in 1h 10m 15s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/85/662485/3/check/kayobe-overcloud-upgrade-centos/9d8b858/ : SUCCESS in 1h 44m 05s\n- kayobe-seed-centos http://logs.openstack.org/85/662485/3/check/kayobe-seed-centos/7788102/ : SUCCESS in 22m 35s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/85/662485/3/check/kayobe-seed-upgrade-centos/e13c473/ : SUCCESS in 27m 44s","accounts_in_message":[],"_revision_number":3},{"id":"ba6781df48324b95ecf4ab74ef9fe8ea00d2f8a7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-03 16:04:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f0c45e9ea82dde4533408eec181a4a6f988c09f7","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-06-13 07:19:27.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c6f0a3f1231e93ee18835f4a011574573f1fad67","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-06-13 07:23:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"71b77697e42de6dccbd4d0b2e14cab7017f0216a","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-06-13 09:15:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5a3d91e0eff53aa7623f52710cee10095c886f1e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-24 14:11:05.000000000","message":"Patch Set 3:\n\nShould we have a policy on the kolla_extra stuff? It seems like a lot of boiler plate to copy around and is only required if Kayobe is providing some configuration itself right?","accounts_in_message":[],"_revision_number":3},{"id":"65d6b417f2bd0fe1bc38a1da7ce70b86ac2955e1","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-24 14:31:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c36d1dfdafc5b2dbe1b5961fe2d7bf6c02b2f61a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:32:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e3e71db15a25086d63f0f44221ea515a4382e435","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 16:16:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/85/662485/3/gate/openstack-tox-docs/2a63359/html/ : SUCCESS in 5m 45s\n- openstack-tox-py35 http://logs.openstack.org/85/662485/3/gate/openstack-tox-py35/ee9e0b8/ : SUCCESS in 5m 30s\n- openstack-tox-py36 http://logs.openstack.org/85/662485/3/gate/openstack-tox-py36/e18941c/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/85/662485/3/gate/openstack-tox-pep8/593736e/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/85/662485/3/gate/openstack-tox-py27/4b49c57/ : SUCCESS in 6m 45s\n- build-openstack-releasenotes http://logs.openstack.org/85/662485/3/gate/build-openstack-releasenotes/cb02fd6/html/ : SUCCESS in 4m 44s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/85/662485/3/gate/kayobe-tox-ansible-syntax/1c3be3e/ : SUCCESS in 6m 55s\n- kayobe-tox-ansible http://logs.openstack.org/85/662485/3/gate/kayobe-tox-ansible/b65a075/ : SUCCESS in 9m 20s\n- kayobe-tox-molecule http://logs.openstack.org/85/662485/3/gate/kayobe-tox-molecule/ae32057/ : SUCCESS in 53m 33s\n- kayobe-overcloud-centos http://logs.openstack.org/85/662485/3/gate/kayobe-overcloud-centos/b2b22c5/ : SUCCESS in 1h 02m 35s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/85/662485/3/gate/kayobe-overcloud-upgrade-centos/7a3a21d/ : SUCCESS in 1h 34m 57s\n- kayobe-seed-centos http://logs.openstack.org/85/662485/3/gate/kayobe-seed-centos/2c03e9f/ : SUCCESS in 38m 10s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/85/662485/3/gate/kayobe-seed-upgrade-centos/6721a99/ : SUCCESS in 23m 52s","accounts_in_message":[],"_revision_number":3},{"id":"0adf390daa926f012b3034edc798c0106bd49b4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 16:16:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"49d4d7be79bf39d32e4456e528dd424cb29b8195","revisions":{"91f88a5215b80b0fbf2ee7eabc117fb2ac61dd52":{"kind":"REWORK","_number":1,"created":"2019-05-31 14:34:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/85/662485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/662485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/662485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/662485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/662485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/662485/1"}}},"commit":{"parents":[{"commit":"4daebe5c8c1f0ce5da2811a582851351cca78635","subject":"Merge \"Add commands for database backup and recovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4daebe5c8c1f0ce5da2811a582851351cca78635"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 14:32:43.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 14:32:43.000000000","tz":60},"subject":"Add support for custom Blazar configuration","message":"Add support for custom Blazar configuration\n\nChange-Id: I00ee10bdc7f8c1b6ab8576115c757e27b6c2c397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91f88a5215b80b0fbf2ee7eabc117fb2ac61dd52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91f88a5215b80b0fbf2ee7eabc117fb2ac61dd52"}]},"branch":"refs/heads/master"},"5604072aaa80bafebacaee3e8483fa3a2453f736":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-31 14:34:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/85/662485/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/662485/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/662485/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/662485/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/662485/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/662485/2"}}},"commit":{"parents":[{"commit":"4daebe5c8c1f0ce5da2811a582851351cca78635","subject":"Merge \"Add commands for database backup and recovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4daebe5c8c1f0ce5da2811a582851351cca78635"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 14:34:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 14:34:21.000000000","tz":60},"subject":"Add support for custom Blazar configuration","message":"Add support for custom Blazar configuration\n\nChange-Id: I00ee10bdc7f8c1b6ab8576115c757e27b6c2c397\nStory: 2005809\nTask: 33549\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5604072aaa80bafebacaee3e8483fa3a2453f736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5604072aaa80bafebacaee3e8483fa3a2453f736"}]},"branch":"refs/heads/master"},"49d4d7be79bf39d32e4456e528dd424cb29b8195":{"kind":"REWORK","_number":3,"created":"2019-05-31 15:55:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/85/662485/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/662485/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/662485/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/662485/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/662485/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/662485/3"}}},"commit":{"parents":[{"commit":"4daebe5c8c1f0ce5da2811a582851351cca78635","subject":"Merge \"Add commands for database backup and recovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4daebe5c8c1f0ce5da2811a582851351cca78635"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 15:55:10.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 15:55:10.000000000","tz":60},"subject":"Add support for custom Blazar configuration","message":"Add support for custom Blazar configuration\n\nChange-Id: I00ee10bdc7f8c1b6ab8576115c757e27b6c2c397\nStory: 2005809\nTask: 33549\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49d4d7be79bf39d32e4456e528dd424cb29b8195"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49d4d7be79bf39d32e4456e528dd424cb29b8195"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
