)]}'
{"id":"openstack%2Fkayobe~662491","triplet_id":"openstack%2Fkayobe~master~I06c07108693766fd436a60a39f270c116931941e","project":"openstack/kayobe","branch":"master","topic":"gnocchi-custom-config","hashtags":[],"change_id":"I06c07108693766fd436a60a39f270c116931941e","subject":"Add support for custom Gnocchi configuration","status":"MERGED","created":"2019-05-31 14:49:21.000000000","updated":"2019-07-03 19:13:07.000000000","submitted":"2019-07-03 19:13:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662491-1562181187670-dfe5b5ea","meta_rev_id":"df5f9c6b700e9acff3c3d16f530bc20e0824a223","_number":662491,"virtual_id_number":662491,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-03 19:13:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"recommended":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"all":[{"value":1,"date":"2019-06-13 07:19:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":2,"date":"2019-06-24 15:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-06-13 07:23:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-06-03 16:04:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-03 14:46:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2019-06-13 09:16:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-06-24 15:19:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-03 14:46:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 16:04:53.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-13 07:19:23.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-06-13 07:23:46.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-06-13 09:16:28.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-06-24 15:19:53.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-03 19:13:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"296a1c27095963c4ef94b07f106ab4b422d070e6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 14:49:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00e97e0959c70a1f72c9f361dc0abb0074105d7b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 14:49:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e5e5c93fa9a0d647a974647c0b3ad94c8819dfbf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 15:06:51.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"c1463e7a3205becd8ebb3a905186126d5d18fef5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 15:08:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"116e5d53e450f9a390fb2316536134b5a27bc9d5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 15:08:55.000000000","message":"Topic changed from gnocchi to gnocchi-custom-config","accounts_in_message":[],"_revision_number":3},{"id":"4186a7f9fbb3151d8512dfbdb6bd6f8827f89302","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 15:44:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b28b9319bf12432242a294bf83c0f60742edfa6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 18:05:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/662491/4/check/openstack-tox-docs/056c9a8/html/ : SUCCESS in 4m 54s\n- openstack-tox-cover http://logs.openstack.org/91/662491/4/check/openstack-tox-cover/e5a426d/cover/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/91/662491/4/check/openstack-tox-py35/d15525b/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/91/662491/4/check/openstack-tox-py36/6354a46/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/91/662491/4/check/openstack-tox-pep8/4ec0190/ : SUCCESS in 5m 05s\n- openstack-tox-py27 http://logs.openstack.org/91/662491/4/check/openstack-tox-py27/c90a6c8/ : SUCCESS in 5m 08s\n- build-openstack-releasenotes http://logs.openstack.org/91/662491/4/check/build-openstack-releasenotes/632ef89/html/ : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/91/662491/4/check/kayobe-tox-ansible-syntax/3ab6f5f/ : SUCCESS in 6m 05s\n- kayobe-tox-ansible http://logs.openstack.org/91/662491/4/check/kayobe-tox-ansible/b4634e2/ : SUCCESS in 9m 24s\n- kayobe-tox-molecule http://logs.openstack.org/91/662491/4/check/kayobe-tox-molecule/3c8e59a/ : SUCCESS in 32m 23s\n- kayobe-overcloud-centos http://logs.openstack.org/91/662491/4/check/kayobe-overcloud-centos/b87bc62/ : SUCCESS in 59m 27s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/91/662491/4/check/kayobe-overcloud-upgrade-centos/299b65d/ : SUCCESS in 1h 55m 54s\n- kayobe-seed-centos http://logs.openstack.org/91/662491/4/check/kayobe-seed-centos/f9cf3c3/ : SUCCESS in 24m 22s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/91/662491/4/check/kayobe-seed-upgrade-centos/661ade9/ : FAILURE in 24m 01s","accounts_in_message":[],"_revision_number":4},{"id":"10b6e4ca6e799baf11bd038ad52b71d96a97a4fa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 18:17:31.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f634e2c18a323a59aeb22e6be14071c208511b40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 20:15:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/662491/4/check/openstack-tox-docs/110a15a/html/ : SUCCESS in 3m 57s\n- openstack-tox-cover http://logs.openstack.org/91/662491/4/check/openstack-tox-cover/15bda25/cover/ : SUCCESS in 4m 58s\n- openstack-tox-py35 http://logs.openstack.org/91/662491/4/check/openstack-tox-py35/e9deb20/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/91/662491/4/check/openstack-tox-py36/4eef824/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/91/662491/4/check/openstack-tox-pep8/992121c/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/91/662491/4/check/openstack-tox-py27/ad35fde/ : SUCCESS in 4m 48s\n- build-openstack-releasenotes http://logs.openstack.org/91/662491/4/check/build-openstack-releasenotes/d56b865/html/ : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/91/662491/4/check/kayobe-tox-ansible-syntax/df77373/ : SUCCESS in 4m 47s\n- kayobe-tox-ansible http://logs.openstack.org/91/662491/4/check/kayobe-tox-ansible/f59a97c/ : SUCCESS in 9m 05s\n- kayobe-tox-molecule http://logs.openstack.org/91/662491/4/check/kayobe-tox-molecule/29ebe95/ : SUCCESS in 33m 27s\n- kayobe-overcloud-centos http://logs.openstack.org/91/662491/4/check/kayobe-overcloud-centos/6983723/ : SUCCESS in 1h 07m 53s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/91/662491/4/check/kayobe-overcloud-upgrade-centos/8959997/ : SUCCESS in 1h 50m 20s\n- kayobe-seed-centos http://logs.openstack.org/91/662491/4/check/kayobe-seed-centos/247cfbe/ : SUCCESS in 20m 11s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/91/662491/4/check/kayobe-seed-upgrade-centos/1a4cb95/ : SUCCESS in 25m 55s","accounts_in_message":[],"_revision_number":4},{"id":"4db805d4deeb84d694299309223bfb0c3d2870e8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-03 16:04:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c0ad941e34c710c9b9925c26f952de84a9161b5c","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-06-13 07:19:23.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2b3d35ab0f94f743b7dba05dbb7141fb4ef58374","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-06-13 07:23:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"929de11019758b58e947b8555408d80344943cc4","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-06-13 09:16:28.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9b3e5584aaf4ca31046acc89d761d3030739bb79","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-24 15:19:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLooks good to me. I wonder what will happen to Gnocchi longer term..","accounts_in_message":[],"_revision_number":4},{"id":"58bd71d6f11b9634b3e343e3fa35448b9c2b8491","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 15:20:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a4a01a868ed0d145ef9ad89ec92f4c9e638f2f78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 20:13:41.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/662491/4/gate/openstack-tox-docs/cf50615/html/ : SUCCESS in 2m 53s\n- openstack-tox-py35 http://logs.openstack.org/91/662491/4/gate/openstack-tox-py35/a3e5efb/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/91/662491/4/gate/openstack-tox-py36/306a831/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/91/662491/4/gate/openstack-tox-pep8/4c68662/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/91/662491/4/gate/openstack-tox-py27/5c497b1/ : SUCCESS in 4m 20s\n- build-openstack-releasenotes http://logs.openstack.org/91/662491/4/gate/build-openstack-releasenotes/9138a1b/html/ : SUCCESS in 3m 15s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/91/662491/4/gate/kayobe-tox-ansible-syntax/ea7c25a/ : SUCCESS in 4m 42s\n- kayobe-tox-ansible http://logs.openstack.org/91/662491/4/gate/kayobe-tox-ansible/45ab69d/ : SUCCESS in 7m 43s\n- kayobe-tox-molecule http://logs.openstack.org/91/662491/4/gate/kayobe-tox-molecule/d8c14f5/ : SUCCESS in 51m 50s\n- kayobe-overcloud-centos http://logs.openstack.org/91/662491/4/gate/kayobe-overcloud-centos/5b16e40/ : SUCCESS in 1h 05m 49s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/91/662491/4/gate/kayobe-overcloud-upgrade-centos/bf65d43/ : SUCCESS in 1h 47m 05s\n- kayobe-seed-centos http://logs.openstack.org/91/662491/4/gate/kayobe-seed-centos/f772c22/ : SUCCESS in 21m 25s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/91/662491/4/gate/kayobe-seed-upgrade-centos/4f4dcff/ : FAILURE in 26m 03s","accounts_in_message":[],"_revision_number":4},{"id":"efe959fc044467ab14d4e3140a43cd731b34e547","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-25 07:49:48.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ec332c257dcfa224c6172d0296d091807744a956","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 09:07:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/662491/4/check/openstack-tox-docs/980fd57/html/ : SUCCESS in 3m 16s\n- openstack-tox-cover http://logs.openstack.org/91/662491/4/check/openstack-tox-cover/abcea1a/cover/ : SUCCESS in 3m 35s\n- openstack-tox-py35 http://logs.openstack.org/91/662491/4/check/openstack-tox-py35/6493132/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/91/662491/4/check/openstack-tox-py36/3f36e48/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/91/662491/4/check/openstack-tox-pep8/ce03b3c/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/91/662491/4/check/openstack-tox-py27/3ffe2b4/ : SUCCESS in 4m 19s\n- build-openstack-releasenotes http://logs.openstack.org/91/662491/4/check/build-openstack-releasenotes/f737a85/html/ : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/91/662491/4/check/kayobe-tox-ansible-syntax/12f2851/ : SUCCESS in 3m 58s\n- kayobe-tox-ansible http://logs.openstack.org/91/662491/4/check/kayobe-tox-ansible/f56a5e1/ : SUCCESS in 7m 51s\n- kayobe-tox-molecule http://logs.openstack.org/91/662491/4/check/kayobe-tox-molecule/9326b86/ : SUCCESS in 52m 53s\n- kayobe-overcloud-centos finger://ze04.openstack.org/056b9cdb1f664d87966be6405c09a49e : RETRY_LIMIT in 1h 05m 40s\n- kayobe-overcloud-upgrade-centos finger://ze03.openstack.org/4f9a2712aa454ed0a44d73c3fa6e5994 : RETRY_LIMIT in 1h 08m 22s\n- kayobe-seed-centos http://logs.openstack.org/91/662491/4/check/kayobe-seed-centos/901a8d0/ : SUCCESS in 21m 55s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/91/662491/4/check/kayobe-seed-upgrade-centos/491abfb/ : SUCCESS in 25m 40s","accounts_in_message":[],"_revision_number":4},{"id":"bec42163a4d0659978594041c3cad324a580eb73","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-25 09:07:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7467e603772cf1ca6227a535e1c75de36bebbdea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 11:09:47.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/662491/4/check/openstack-tox-docs/5b63a2f/ : POST_FAILURE in 7m 00s\n- openstack-tox-cover http://logs.openstack.org/91/662491/4/check/openstack-tox-cover/c453a0e/cover/ : SUCCESS in 5m 17s\n- openstack-tox-py35 http://logs.openstack.org/91/662491/4/check/openstack-tox-py35/1d3970f/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/91/662491/4/check/openstack-tox-py36/11e0d78/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/91/662491/4/check/openstack-tox-pep8/737764e/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/91/662491/4/check/openstack-tox-py27/062b2c4/ : SUCCESS in 3m 44s\n- build-openstack-releasenotes http://logs.openstack.org/91/662491/4/check/build-openstack-releasenotes/3d82704/html/ : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/91/662491/4/check/kayobe-tox-ansible-syntax/9aded3e/ : SUCCESS in 4m 07s\n- kayobe-tox-ansible http://logs.openstack.org/91/662491/4/check/kayobe-tox-ansible/41bdc41/ : SUCCESS in 14m 15s\n- kayobe-tox-molecule http://logs.openstack.org/91/662491/4/check/kayobe-tox-molecule/c2e5e09/ : SUCCESS in 51m 13s\n- kayobe-overcloud-centos finger://ze09.openstack.org/b34a2a3858dc49aa8844ee8bc888bfcd : RETRY_LIMIT in 1h 00m 08s\n- kayobe-overcloud-upgrade-centos finger://ze08.openstack.org/d79ce960c09d4adfaa3a917467ec6475 : RETRY_LIMIT in 59m 30s\n- kayobe-seed-centos http://logs.openstack.org/91/662491/4/check/kayobe-seed-centos/7e33236/ : SUCCESS in 19m 42s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/91/662491/4/check/kayobe-seed-upgrade-centos/d02e6cb/ : SUCCESS in 28m 35s","accounts_in_message":[],"_revision_number":4},{"id":"1cf0b7e0054547aa036453a0fb34806a28dfeefc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-03 10:34:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d941eec95ce305395e21af08c681feafa11214bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 12:43:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/662491/4/check/openstack-tox-docs/3ee5eb6/html/ : SUCCESS in 2m 43s\n- openstack-tox-cover http://logs.openstack.org/91/662491/4/check/openstack-tox-cover/5f0bb31/cover/ : SUCCESS in 3m 45s\n- openstack-tox-py35 http://logs.openstack.org/91/662491/4/check/openstack-tox-py35/aa92ecf/ : SUCCESS in 4m 00s\n- openstack-tox-py36 http://logs.openstack.org/91/662491/4/check/openstack-tox-py36/36f42d9/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/91/662491/4/check/openstack-tox-pep8/83e7fb3/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/91/662491/4/check/openstack-tox-py27/311ee52/ : SUCCESS in 5m 13s\n- build-openstack-releasenotes http://logs.openstack.org/91/662491/4/check/build-openstack-releasenotes/0206197/html/ : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/91/662491/4/check/kayobe-tox-ansible-syntax/f21c80e/ : SUCCESS in 3m 57s\n- kayobe-tox-ansible http://logs.openstack.org/91/662491/4/check/kayobe-tox-ansible/a522b45/ : SUCCESS in 8m 27s\n- kayobe-tox-molecule http://logs.openstack.org/91/662491/4/check/kayobe-tox-molecule/b11a906/ : SUCCESS in 50m 31s\n- kayobe-overcloud-centos http://logs.openstack.org/91/662491/4/check/kayobe-overcloud-centos/26bed8d/ : SUCCESS in 1h 04m 33s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/91/662491/4/check/kayobe-overcloud-upgrade-centos/226c85a/ : SUCCESS in 1h 51m 41s\n- kayobe-seed-centos http://logs.openstack.org/91/662491/4/check/kayobe-seed-centos/d1301d1/ : SUCCESS in 50m 16s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/91/662491/4/check/kayobe-seed-upgrade-centos/0a40796/ : SUCCESS in 1h 16m 33s","accounts_in_message":[],"_revision_number":4},{"id":"56015759a7d5c0ab3b6e4b95975481ac7e705528","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 12:43:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a9205649cf389d462c54d49630b8e5cc87173d7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 14:45:08.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/91/662491/4/gate/openstack-tox-docs/f3387ae/html/ : SUCCESS in 2m 53s\n- openstack-tox-py35 http://logs.openstack.org/91/662491/4/gate/openstack-tox-py35/9905b84/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/91/662491/4/gate/openstack-tox-py36/08de691/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/91/662491/4/gate/openstack-tox-pep8/a181978/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/91/662491/4/gate/openstack-tox-py27/10cb3e4/ : SUCCESS in 3m 35s\n- build-openstack-releasenotes http://logs.openstack.org/91/662491/4/gate/build-openstack-releasenotes/120d538/html/ : SUCCESS in 2m 56s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/91/662491/4/gate/kayobe-tox-ansible-syntax/0c97e2a/ : SUCCESS in 5m 13s\n- kayobe-tox-ansible http://logs.openstack.org/91/662491/4/gate/kayobe-tox-ansible/acdab2b/ : SUCCESS in 7m 05s\n- kayobe-tox-molecule http://logs.openstack.org/91/662491/4/gate/kayobe-tox-molecule/9572453/ : SUCCESS in 51m 53s\n- kayobe-overcloud-centos http://logs.openstack.org/91/662491/4/gate/kayobe-overcloud-centos/e4da2d0/ : SUCCESS in 1h 01m 19s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/91/662491/4/gate/kayobe-overcloud-upgrade-centos/c263e9f/ : FAILURE in 15m 32s\n- kayobe-seed-centos http://logs.openstack.org/91/662491/4/gate/kayobe-seed-centos/9695056/ : SUCCESS in 22m 11s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/91/662491/4/gate/kayobe-seed-upgrade-centos/ef33e3d/ : SUCCESS in 31m 54s","accounts_in_message":[],"_revision_number":4},{"id":"e2b460395222f4b924efaa370dce987299795cf3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-03 14:46:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1f1692070175a3fcbe97a1d4ff707b0fe3251494","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 17:15:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/662491/4/check/openstack-tox-docs/825e44a/html/ : SUCCESS in 3m 24s\n- openstack-tox-cover http://logs.openstack.org/91/662491/4/check/openstack-tox-cover/c5f990e/cover/ : SUCCESS in 4m 24s\n- openstack-tox-py35 http://logs.openstack.org/91/662491/4/check/openstack-tox-py35/32b768b/ : SUCCESS in 4m 00s\n- openstack-tox-py36 http://logs.openstack.org/91/662491/4/check/openstack-tox-py36/98db92c/ : SUCCESS in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/91/662491/4/check/openstack-tox-pep8/09782d8/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/91/662491/4/check/openstack-tox-py27/068b14b/ : SUCCESS in 3m 24s\n- build-openstack-releasenotes http://logs.openstack.org/91/662491/4/check/build-openstack-releasenotes/4fdafba/html/ : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/91/662491/4/check/kayobe-tox-ansible-syntax/ac4624d/ : SUCCESS in 4m 07s\n- kayobe-tox-ansible http://logs.openstack.org/91/662491/4/check/kayobe-tox-ansible/0e194c6/ : SUCCESS in 8m 38s\n- kayobe-tox-molecule http://logs.openstack.org/91/662491/4/check/kayobe-tox-molecule/58a415d/ : SUCCESS in 51m 13s\n- kayobe-overcloud-centos http://logs.openstack.org/91/662491/4/check/kayobe-overcloud-centos/1a76ef0/ : SUCCESS in 1h 09m 06s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/91/662491/4/check/kayobe-overcloud-upgrade-centos/5a6f0e6/ : SUCCESS in 1h 58m 34s\n- kayobe-seed-centos http://logs.openstack.org/91/662491/4/check/kayobe-seed-centos/6af3632/ : SUCCESS in 21m 04s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/91/662491/4/check/kayobe-seed-upgrade-centos/90d081a/ : SUCCESS in 28m 41s","accounts_in_message":[],"_revision_number":4},{"id":"a54630c19e123883e9d18d0b462ae68316a61d2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 17:15:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7571124ee29c77327e7d4b2b050e12cc7632ae7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 19:13:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"df5f9c6b700e9acff3c3d16f530bc20e0824a223","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 19:13:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/662491/4/gate/openstack-tox-docs/9ed50cb/html/ : SUCCESS in 3m 36s\n- openstack-tox-py35 http://logs.openstack.org/91/662491/4/gate/openstack-tox-py35/1dd78ef/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/91/662491/4/gate/openstack-tox-py36/e08cfc1/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/91/662491/4/gate/openstack-tox-pep8/bfc4228/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/91/662491/4/gate/openstack-tox-py27/d7b3545/ : SUCCESS in 4m 13s\n- build-openstack-releasenotes http://logs.openstack.org/91/662491/4/gate/build-openstack-releasenotes/e80e7b8/html/ : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/91/662491/4/gate/kayobe-tox-ansible-syntax/fbc1595/ : SUCCESS in 4m 45s\n- kayobe-tox-ansible http://logs.openstack.org/91/662491/4/gate/kayobe-tox-ansible/b3c7a7b/ : SUCCESS in 8m 07s\n- kayobe-tox-molecule http://logs.openstack.org/91/662491/4/gate/kayobe-tox-molecule/669c98d/ : SUCCESS in 53m 36s\n- kayobe-overcloud-centos http://logs.openstack.org/91/662491/4/gate/kayobe-overcloud-centos/8249894/ : SUCCESS in 1h 13m 28s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/91/662491/4/gate/kayobe-overcloud-upgrade-centos/b4768ad/ : SUCCESS in 1h 49m 14s\n- kayobe-seed-centos http://logs.openstack.org/91/662491/4/gate/kayobe-seed-centos/377ca5b/ : SUCCESS in 24m 44s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/91/662491/4/gate/kayobe-seed-upgrade-centos/685b878/ : SUCCESS in 30m 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"603527c215b08c47e213cd5115bc1637c9706ade","revisions":{"c829da567d6490110375aca9d6ec8a984a6b7c80":{"kind":"REWORK","_number":1,"created":"2019-05-31 14:49:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/91/662491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/662491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/662491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/662491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/662491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/662491/1"}}},"commit":{"parents":[{"commit":"4daebe5c8c1f0ce5da2811a582851351cca78635","subject":"Merge \"Add commands for database backup and recovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4daebe5c8c1f0ce5da2811a582851351cca78635"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 14:48:53.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 14:48:53.000000000","tz":60},"subject":"Add support for custom Gnocchi configuration","message":"Add support for custom Gnocchi configuration\n\nChange-Id: I06c07108693766fd436a60a39f270c116931941e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c829da567d6490110375aca9d6ec8a984a6b7c80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c829da567d6490110375aca9d6ec8a984a6b7c80"}]},"branch":"refs/heads/master"},"722bb819c6b34e9ca0087d85481ad82c7bd1e6e5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-31 14:49:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/91/662491/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/662491/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/662491/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/662491/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/662491/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/662491/2"}}},"commit":{"parents":[{"commit":"4daebe5c8c1f0ce5da2811a582851351cca78635","subject":"Merge \"Add commands for database backup and recovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4daebe5c8c1f0ce5da2811a582851351cca78635"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 14:49:26.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 14:49:26.000000000","tz":60},"subject":"Add support for custom Gnocchi configuration","message":"Add support for custom Gnocchi configuration\n\nChange-Id: I06c07108693766fd436a60a39f270c116931941e\nStory: 2005810\nTask: 33550\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/722bb819c6b34e9ca0087d85481ad82c7bd1e6e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/722bb819c6b34e9ca0087d85481ad82c7bd1e6e5"}]},"branch":"refs/heads/master"},"f313c1870d3796fe581e1045825c0d88be22f7ad":{"kind":"REWORK","_number":3,"created":"2019-05-31 15:08:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/91/662491/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/662491/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/662491/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/662491/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/662491/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/662491/3"}}},"commit":{"parents":[{"commit":"4daebe5c8c1f0ce5da2811a582851351cca78635","subject":"Merge \"Add commands for database backup and recovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4daebe5c8c1f0ce5da2811a582851351cca78635"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 15:08:04.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 15:08:04.000000000","tz":60},"subject":"Add support for custom Gnocchi configuration","message":"Add support for custom Gnocchi configuration\n\nChange-Id: I06c07108693766fd436a60a39f270c116931941e\nStory: 2005810\nTask: 33550\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f313c1870d3796fe581e1045825c0d88be22f7ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f313c1870d3796fe581e1045825c0d88be22f7ad"}]},"branch":"refs/heads/master"},"603527c215b08c47e213cd5115bc1637c9706ade":{"kind":"REWORK","_number":4,"created":"2019-05-31 15:44:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/91/662491/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/662491/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/662491/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/662491/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/662491/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/662491/4"}}},"commit":{"parents":[{"commit":"4daebe5c8c1f0ce5da2811a582851351cca78635","subject":"Merge \"Add commands for database backup and recovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4daebe5c8c1f0ce5da2811a582851351cca78635"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 15:29:54.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 15:29:54.000000000","tz":60},"subject":"Add support for custom Gnocchi configuration","message":"Add support for custom Gnocchi configuration\n\nChange-Id: I06c07108693766fd436a60a39f270c116931941e\nStory: 2005810\nTask: 33550\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/603527c215b08c47e213cd5115bc1637c9706ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/603527c215b08c47e213cd5115bc1637c9706ade"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
