)]}'
{"id":"openstack%2Fkayobe~662495","triplet_id":"openstack%2Fkayobe~master~I0a9e1a3b878231f9ce7bf295b5526c9cfabbbf9e","project":"openstack/kayobe","branch":"master","topic":"ceilometer-custom-config","hashtags":[],"change_id":"I0a9e1a3b878231f9ce7bf295b5526c9cfabbbf9e","subject":"Add support for custom Ceilometer configuration","status":"MERGED","created":"2019-05-31 15:04:40.000000000","updated":"2019-07-12 21:28:40.000000000","submitted":"2019-07-12 21:28:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662495-1562966920789-fd40a304","meta_rev_id":"334c3628fd9a2f084cb180b27f456d76336b97c4","_number":662495,"virtual_id_number":662495,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-12 21:28:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-03 14:46:05.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":2,"date":"2019-07-12 16:35:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-07-03 18:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-07-12 16:35:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 07:19:30.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-06-13 07:20:54.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-06-13 09:16:16.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-07-03 18:31:43.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-12 16:35:08.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-12 21:28:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a7ca5f51c6938bd06a870cc21837c39af116437","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 15:04:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"265f293aa21e5b356972ba562768078f372b0a27","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 15:06:48.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"481e07dffdfc9a46fbcc64fe10f6b4324db23f33","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 15:08:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3958fad8a8c15c7c20fb3aa8ded514298a463b86","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 15:08:49.000000000","message":"Topic changed from ceilometer to ceilometer-custom-config","accounts_in_message":[],"_revision_number":2},{"id":"85c86abf532f3facded69dac6a4bd25000537ec3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-05-31 15:44:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f83409847a8303c237bc18b0ac0eb2c34bf0a217","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 17:53:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/662495/3/check/openstack-tox-docs/37ce32f/html/ : SUCCESS in 4m 48s\n- openstack-tox-cover http://logs.openstack.org/95/662495/3/check/openstack-tox-cover/4a0dcd0/cover/ : SUCCESS in 5m 22s\n- openstack-tox-py35 http://logs.openstack.org/95/662495/3/check/openstack-tox-py35/82523b0/ : SUCCESS in 5m 28s\n- openstack-tox-py36 http://logs.openstack.org/95/662495/3/check/openstack-tox-py36/89f075b/ : SUCCESS in 5m 11s\n- openstack-tox-pep8 http://logs.openstack.org/95/662495/3/check/openstack-tox-pep8/e7ae89b/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/95/662495/3/check/openstack-tox-py27/ab52907/ : SUCCESS in 4m 36s\n- build-openstack-releasenotes http://logs.openstack.org/95/662495/3/check/build-openstack-releasenotes/c764d9a/html/ : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/95/662495/3/check/kayobe-tox-ansible-syntax/233913c/ : SUCCESS in 5m 55s\n- kayobe-tox-ansible http://logs.openstack.org/95/662495/3/check/kayobe-tox-ansible/33dab4a/ : SUCCESS in 10m 17s\n- kayobe-tox-molecule http://logs.openstack.org/95/662495/3/check/kayobe-tox-molecule/436445d/ : SUCCESS in 34m 55s\n- kayobe-overcloud-centos http://logs.openstack.org/95/662495/3/check/kayobe-overcloud-centos/1c6fb5a/ : SUCCESS in 1h 04m 48s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/95/662495/3/check/kayobe-overcloud-upgrade-centos/fcc0ccb/ : SUCCESS in 1h 52m 42s\n- kayobe-seed-centos http://logs.openstack.org/95/662495/3/check/kayobe-seed-centos/1358913/ : SUCCESS in 19m 47s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/95/662495/3/check/kayobe-seed-upgrade-centos/7d54634/ : SUCCESS in 25m 47s","accounts_in_message":[],"_revision_number":3},{"id":"28e16c7a9ad779e180050094968b2fefb68f6e75","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-03 16:03:14.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":3},{"id":"98ed28bd8c2e6ce467cdc38037948742b5c8f837","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-06-13 07:19:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3032aedd6f1f481cef76350a3f4e5f537e67a284","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-06-13 07:20:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"95f5c56796e9e4715ed580786550a670c5e0bc7e","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-06-13 09:16:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8bd85932288fec96067753e938c37b7ff7835304","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-24 15:21:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nFollows the pattern..","accounts_in_message":[],"_revision_number":3},{"id":"3c110d3e903d3010f530cca3f3fc557a559078b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 20:08:47.000000000","message":"Patch Set 3: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"7f24bd8511ba502b939184cd57840454085e2ecf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-25 09:45:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0cd94881eeaf6cf437f6361cda990752883c748f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-25 09:47:59.000000000","message":"Patch Set 4:\n\nResolved merged conflicts","accounts_in_message":[],"_revision_number":4},{"id":"5a97dbfa32e1068a28c5eeff356bafff6d45b50d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 13:13:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/662495/4/check/openstack-tox-docs/79bea07/html/ : SUCCESS in 2m 42s\n- openstack-tox-cover http://logs.openstack.org/95/662495/4/check/openstack-tox-cover/050f326/cover/ : SUCCESS in 4m 06s\n- openstack-tox-py35 http://logs.openstack.org/95/662495/4/check/openstack-tox-py35/3b81ed6/ : SUCCESS in 6m 02s\n- openstack-tox-py36 http://logs.openstack.org/95/662495/4/check/openstack-tox-py36/c1fda42/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/95/662495/4/check/openstack-tox-pep8/a245958/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/95/662495/4/check/openstack-tox-py27/0fb160b/ : SUCCESS in 5m 35s\n- build-openstack-releasenotes http://logs.openstack.org/95/662495/4/check/build-openstack-releasenotes/4b8f2fb/html/ : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/95/662495/4/check/kayobe-tox-ansible-syntax/ce88476/ : SUCCESS in 5m 14s\n- kayobe-tox-ansible http://logs.openstack.org/95/662495/4/check/kayobe-tox-ansible/7b8711a/ : SUCCESS in 8m 44s\n- kayobe-tox-molecule finger://ze01.openstack.org/d4e121ae28a749f295cca39a07ded322 : RETRY_LIMIT in 49m 19s\n- kayobe-overcloud-centos finger://ze05.openstack.org/42c8d522662f450f8eef7656c65aa0d6 : RETRY_LIMIT in 58m 22s\n- kayobe-overcloud-upgrade-centos finger://ze05.openstack.org/df2cca27937e480b9960dee21cbf73f7 : RETRY_LIMIT in 56m 02s\n- kayobe-seed-centos http://logs.openstack.org/95/662495/4/check/kayobe-seed-centos/e28d07e/ : SUCCESS in 21m 35s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/95/662495/4/check/kayobe-seed-upgrade-centos/d836d60/ : SUCCESS in 25m 30s","accounts_in_message":[],"_revision_number":4},{"id":"f20033c6375e7f0df2c6fc158fe993d780b7fa1e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-25 17:38:08.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5afc497d93b1d00503ffec5bfa414914873c7fe5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 19:24:52.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/662495/4/check/openstack-tox-docs/53e7c81/html/ : SUCCESS in 2m 43s\n- openstack-tox-cover http://logs.openstack.org/95/662495/4/check/openstack-tox-cover/9becd0b/cover/ : SUCCESS in 3m 35s\n- openstack-tox-py35 http://logs.openstack.org/95/662495/4/check/openstack-tox-py35/2e2d10d/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/95/662495/4/check/openstack-tox-py36/157453e/ : SUCCESS in 3m 15s\n- openstack-tox-pep8 http://logs.openstack.org/95/662495/4/check/openstack-tox-pep8/a947b03/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/95/662495/4/check/openstack-tox-py27/16fa11e/ : SUCCESS in 3m 23s\n- build-openstack-releasenotes http://logs.openstack.org/95/662495/4/check/build-openstack-releasenotes/0d0e329/html/ : SUCCESS in 2m 31s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/95/662495/4/check/kayobe-tox-ansible-syntax/3d4177f/ : SUCCESS in 3m 53s\n- kayobe-tox-ansible http://logs.openstack.org/95/662495/4/check/kayobe-tox-ansible/5e85e49/ : SUCCESS in 7m 49s\n- kayobe-tox-molecule http://logs.openstack.org/95/662495/4/check/kayobe-tox-molecule/314b2c5/ : SUCCESS in 50m 51s\n- kayobe-overcloud-centos http://logs.openstack.org/95/662495/4/check/kayobe-overcloud-centos/68aa4fe/ : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/95/662495/4/check/kayobe-overcloud-upgrade-centos/e1983f2/ : SUCCESS in 1h 40m 14s\n- kayobe-seed-centos http://logs.openstack.org/95/662495/4/check/kayobe-seed-centos/ed4079a/ : SUCCESS in 20m 30s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/95/662495/4/check/kayobe-seed-upgrade-centos/5a0a13b/ : FAILURE in 22m 31s","accounts_in_message":[],"_revision_number":4},{"id":"68299f662f562486d4ed529a8abad0c52260be54","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-26 08:50:20.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7274692a7ea0db3cdfc9a924dc4560b4a7c2ad60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 10:10:06.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/95/662495/4/check/openstack-tox-docs/f90f79b/html/ : SUCCESS in 3m 01s\n- openstack-tox-cover http://logs.openstack.org/95/662495/4/check/openstack-tox-cover/5d63ee7/ : FAILURE in 15m 14s\n- openstack-tox-py35 http://logs.openstack.org/95/662495/4/check/openstack-tox-py35/6746723/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/95/662495/4/check/openstack-tox-py36/6daf971/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/95/662495/4/check/openstack-tox-pep8/47707a9/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/95/662495/4/check/openstack-tox-py27/9ab5bfb/ : SUCCESS in 3m 29s\n- build-openstack-releasenotes http://logs.openstack.org/95/662495/4/check/build-openstack-releasenotes/3d728eb/html/ : SUCCESS in 2m 49s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/95/662495/4/check/kayobe-tox-ansible-syntax/999cc70/ : FAILURE in 15m 12s\n- kayobe-tox-ansible http://logs.openstack.org/95/662495/4/check/kayobe-tox-ansible/959260f/ : FAILURE in 15m 07s\n- kayobe-tox-molecule http://logs.openstack.org/95/662495/4/check/kayobe-tox-molecule/7d6c057/ : SUCCESS in 53m 43s\n- kayobe-overcloud-centos http://logs.openstack.org/95/662495/4/check/kayobe-overcloud-centos/9a894ec/ : RETRY_LIMIT in 7m 13s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/95/662495/4/check/kayobe-overcloud-upgrade-centos/42772c4/ : RETRY_LIMIT in 7m 00s\n- kayobe-seed-centos http://logs.openstack.org/95/662495/4/check/kayobe-seed-centos/a464549/ : SUCCESS in 21m 56s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/95/662495/4/check/kayobe-seed-upgrade-centos/e89fa69/ : SUCCESS in 24m 44s","accounts_in_message":[],"_revision_number":4},{"id":"cc260eef46d5959f2063981a326bce7c92432376","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-03 14:46:05.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"47c437fd38151b1c42166bc746588b58ff7be53e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 16:46:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/662495/4/check/openstack-tox-docs/9c83a7b/html/ : SUCCESS in 3m 59s\n- openstack-tox-cover http://logs.openstack.org/95/662495/4/check/openstack-tox-cover/34b0d6b/cover/ : SUCCESS in 3m 41s\n- openstack-tox-py35 http://logs.openstack.org/95/662495/4/check/openstack-tox-py35/a32a29c/ : SUCCESS in 4m 17s\n- openstack-tox-py36 http://logs.openstack.org/95/662495/4/check/openstack-tox-py36/53d3ac9/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/95/662495/4/check/openstack-tox-pep8/c81e9a4/ : SUCCESS in 4m 47s\n- openstack-tox-py27 http://logs.openstack.org/95/662495/4/check/openstack-tox-py27/d2e1653/ : SUCCESS in 4m 33s\n- build-openstack-releasenotes http://logs.openstack.org/95/662495/4/check/build-openstack-releasenotes/88e7728/html/ : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/95/662495/4/check/kayobe-tox-ansible-syntax/6dd19bc/ : SUCCESS in 4m 13s\n- kayobe-tox-ansible http://logs.openstack.org/95/662495/4/check/kayobe-tox-ansible/2a33a81/ : SUCCESS in 8m 53s\n- kayobe-tox-molecule http://logs.openstack.org/95/662495/4/check/kayobe-tox-molecule/e57e7ce/ : SUCCESS in 50m 04s\n- kayobe-overcloud-centos http://logs.openstack.org/95/662495/4/check/kayobe-overcloud-centos/55aca65/ : SUCCESS in 1h 07m 06s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/95/662495/4/check/kayobe-overcloud-upgrade-centos/953a6e1/ : SUCCESS in 1h 47m 08s\n- kayobe-seed-centos http://logs.openstack.org/95/662495/4/check/kayobe-seed-centos/bbf3f12/ : SUCCESS in 24m 01s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/95/662495/4/check/kayobe-seed-upgrade-centos/41f0e1e/ : SUCCESS in 27m 18s","accounts_in_message":[],"_revision_number":4},{"id":"bae747065c54657d573cf85345f851652f49fd1f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-03 18:31:43.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"86e13a73fa73019208fa2afec1bab55ebe4613ae","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-12 16:35:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4a038bc9e041ccedd49443839867169ed361d21c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 16:35:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"941c4a136183415dcd8ef58762df6cf4f739ca55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 21:28:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"334c3628fd9a2f084cb180b27f456d76336b97c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 21:28:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/95/662495/4/gate/openstack-tox-docs/2d79aa9/html/ : SUCCESS in 3m 29s\n- openstack-tox-py35 http://logs.openstack.org/95/662495/4/gate/openstack-tox-py35/9e6a152/ : SUCCESS in 3m 51s\n- openstack-tox-py36 http://logs.openstack.org/95/662495/4/gate/openstack-tox-py36/393722b/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 http://logs.openstack.org/95/662495/4/gate/openstack-tox-pep8/514df4f/ : SUCCESS in 5m 16s\n- openstack-tox-py27 http://logs.openstack.org/95/662495/4/gate/openstack-tox-py27/4e4ac7a/ : SUCCESS in 3m 35s\n- build-openstack-releasenotes http://logs.openstack.org/95/662495/4/gate/build-openstack-releasenotes/289b6b3/html/ : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/95/662495/4/gate/kayobe-tox-ansible-syntax/4e29e11/ : SUCCESS in 6m 54s\n- kayobe-tox-ansible http://logs.openstack.org/95/662495/4/gate/kayobe-tox-ansible/41449a5/ : SUCCESS in 8m 26s\n- kayobe-tox-molecule http://logs.openstack.org/95/662495/4/gate/kayobe-tox-molecule/63ecb52/ : SUCCESS in 53m 18s\n- kayobe-overcloud-centos http://logs.openstack.org/95/662495/4/gate/kayobe-overcloud-centos/cd11749/ : SUCCESS in 1h 07m 05s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/95/662495/4/gate/kayobe-overcloud-upgrade-centos/0584e6c/ : SUCCESS in 1h 47m 04s\n- kayobe-seed-centos http://logs.openstack.org/95/662495/4/gate/kayobe-seed-centos/bae35e7/ : SUCCESS in 24m 04s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/95/662495/4/gate/kayobe-seed-upgrade-centos/d953b0b/ : SUCCESS in 29m 08s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"58f856fd00d4ef291d9043f008cb32fba6b2fda1","revisions":{"875fd53dc6a798d81622ade64d365871412d5719":{"kind":"REWORK","_number":1,"created":"2019-05-31 15:04:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/95/662495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/662495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/662495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/662495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/662495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/662495/1"}}},"commit":{"parents":[{"commit":"4daebe5c8c1f0ce5da2811a582851351cca78635","subject":"Merge \"Add commands for database backup and recovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4daebe5c8c1f0ce5da2811a582851351cca78635"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 15:04:04.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 15:04:04.000000000","tz":60},"subject":"Add support for custom Ceilometer configuration","message":"Add support for custom Ceilometer configuration\n\nChange-Id: I0a9e1a3b878231f9ce7bf295b5526c9cfabbbf9e\nStory: 2005811\nTask: 33551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/875fd53dc6a798d81622ade64d365871412d5719"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/875fd53dc6a798d81622ade64d365871412d5719"}]},"branch":"refs/heads/master"},"620c8ef5b96de8d60fa2bd719c21a77b5e739b20":{"kind":"REWORK","_number":2,"created":"2019-05-31 15:08:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/95/662495/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/662495/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/662495/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/662495/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/662495/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/662495/2"}}},"commit":{"parents":[{"commit":"4daebe5c8c1f0ce5da2811a582851351cca78635","subject":"Merge \"Add commands for database backup and recovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4daebe5c8c1f0ce5da2811a582851351cca78635"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 15:08:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 15:08:27.000000000","tz":60},"subject":"Add support for custom Ceilometer configuration","message":"Add support for custom Ceilometer configuration\n\nChange-Id: I0a9e1a3b878231f9ce7bf295b5526c9cfabbbf9e\nStory: 2005811\nTask: 33551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/620c8ef5b96de8d60fa2bd719c21a77b5e739b20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/620c8ef5b96de8d60fa2bd719c21a77b5e739b20"}]},"branch":"refs/heads/master"},"92b838c3daf612a31076af62c0f6fccc05641b38":{"kind":"REWORK","_number":3,"created":"2019-05-31 15:44:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/95/662495/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/662495/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/662495/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/662495/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/662495/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/662495/3"}}},"commit":{"parents":[{"commit":"4daebe5c8c1f0ce5da2811a582851351cca78635","subject":"Merge \"Add commands for database backup and recovery\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4daebe5c8c1f0ce5da2811a582851351cca78635"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 15:32:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 15:32:16.000000000","tz":60},"subject":"Add support for custom Ceilometer configuration","message":"Add support for custom Ceilometer configuration\n\nChange-Id: I0a9e1a3b878231f9ce7bf295b5526c9cfabbbf9e\nStory: 2005811\nTask: 33551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/92b838c3daf612a31076af62c0f6fccc05641b38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/92b838c3daf612a31076af62c0f6fccc05641b38"}]},"branch":"refs/heads/master"},"58f856fd00d4ef291d9043f008cb32fba6b2fda1":{"kind":"REWORK","_number":4,"created":"2019-06-25 09:45:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/95/662495/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/662495/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/662495/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/662495/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/662495/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/662495/4"}}},"commit":{"parents":[{"commit":"ab1e5321a96fd119d46c3c70c013ce53fdca8de4","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab1e5321a96fd119d46c3c70c013ce53fdca8de4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-05-31 15:32:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-25 09:44:57.000000000","tz":60},"subject":"Add support for custom Ceilometer configuration","message":"Add support for custom Ceilometer configuration\n\nChange-Id: I0a9e1a3b878231f9ce7bf295b5526c9cfabbbf9e\nStory: 2005811\nTask: 33551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/58f856fd00d4ef291d9043f008cb32fba6b2fda1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/58f856fd00d4ef291d9043f008cb32fba6b2fda1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
