)]}'
{"id":"openstack%2Fkayobe~664533","triplet_id":"openstack%2Fkayobe~master~Ieddf039483649cc7b575019409257b6ce5b3fb99","project":"openstack/kayobe","branch":"master","topic":"molecule-job","hashtags":[],"change_id":"Ieddf039483649cc7b575019409257b6ce5b3fb99","subject":"Cap paramiko to releases below 2.5.0","status":"MERGED","created":"2019-06-11 09:29:21.000000000","updated":"2019-06-13 08:54:11.000000000","submitted":"2019-06-11 19:43:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664533-1560282199974-d90c8b36","meta_rev_id":"942e81ff5bbc4aff4a966244fa2f8db09b79f6ee","_number":664533,"virtual_id_number":664533,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-11 19:43:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-06-11 17:59:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-06-11 17:59:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 17:59:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-06-11 19:43:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c36f16a9af3b50acb9f8f111dbcc02a4c2d41344","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-11 09:29:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd90b8fa00c46e562437156571af64ab7de70d41","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-11 10:57:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"43d10b3e318b803b8e90ba8be75f7fd150e64456","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 13:04:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/664533/2/check/openstack-tox-docs/d4249a4/html/ : SUCCESS in 4m 07s\n- openstack-tox-cover http://logs.openstack.org/33/664533/2/check/openstack-tox-cover/6134e6d/cover/ : SUCCESS in 5m 20s\n- openstack-tox-py35 http://logs.openstack.org/33/664533/2/check/openstack-tox-py35/41a7263/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/33/664533/2/check/openstack-tox-py36/b34277f/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/33/664533/2/check/openstack-tox-pep8/3822d9d/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/33/664533/2/check/openstack-tox-py27/6ee6908/ : SUCCESS in 5m 05s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/33/664533/2/check/kayobe-tox-ansible-syntax/0f7e7a8/ : SUCCESS in 6m 31s\n- kayobe-tox-ansible http://logs.openstack.org/33/664533/2/check/kayobe-tox-ansible/accdb31/ : SUCCESS in 9m 12s\n- kayobe-tox-molecule http://logs.openstack.org/33/664533/2/check/kayobe-tox-molecule/093b6b9/ : FAILURE in 7m 20s\n- kayobe-overcloud-centos http://logs.openstack.org/33/664533/2/check/kayobe-overcloud-centos/323bcc7/ : SUCCESS in 1h 06m 56s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/33/664533/2/check/kayobe-overcloud-upgrade-centos/5698c50/ : SUCCESS in 1h 59m 44s\n- kayobe-seed-centos http://logs.openstack.org/33/664533/2/check/kayobe-seed-centos/5713102/ : SUCCESS in 20m 17s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/33/664533/2/check/kayobe-seed-upgrade-centos/0df1ad3/ : FAILURE in 23m 45s","accounts_in_message":[],"_revision_number":2},{"id":"4046914dbf203407ff535228963278f536cefd18","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-11 13:53:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f723c692ad686d7fa795eaac130d060e0ba97f02","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-11 14:10:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c6244991b4f6054e9488560c4e57139e5eb1140f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 16:28:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/664533/3/check/openstack-tox-docs/360bd9a/html/ : SUCCESS in 4m 55s\n- openstack-tox-cover http://logs.openstack.org/33/664533/3/check/openstack-tox-cover/ca9195f/cover/ : SUCCESS in 6m 43s\n- openstack-tox-py35 http://logs.openstack.org/33/664533/3/check/openstack-tox-py35/645a308/ : SUCCESS in 4m 40s\n- openstack-tox-py36 http://logs.openstack.org/33/664533/3/check/openstack-tox-py36/31896fa/ : SUCCESS in 5m 11s\n- openstack-tox-pep8 http://logs.openstack.org/33/664533/3/check/openstack-tox-pep8/8c07a49/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/33/664533/3/check/openstack-tox-py27/ea36331/ : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/33/664533/3/check/kayobe-tox-ansible-syntax/1d0170b/ : SUCCESS in 5m 08s\n- kayobe-tox-ansible http://logs.openstack.org/33/664533/3/check/kayobe-tox-ansible/ccd2a56/ : SUCCESS in 8m 53s\n- kayobe-tox-molecule http://logs.openstack.org/33/664533/3/check/kayobe-tox-molecule/220161a/ : SUCCESS in 37m 50s\n- kayobe-overcloud-centos http://logs.openstack.org/33/664533/3/check/kayobe-overcloud-centos/cb033dc/ : SUCCESS in 1h 38m 31s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/33/664533/3/check/kayobe-overcloud-upgrade-centos/ca9bfc2/ : SUCCESS in 2h 30m 57s\n- kayobe-seed-centos http://logs.openstack.org/33/664533/3/check/kayobe-seed-centos/1b6ba75/ : SUCCESS in 25m 29s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/33/664533/3/check/kayobe-seed-upgrade-centos/d327d75/ : SUCCESS in 25m 04s","accounts_in_message":[],"_revision_number":3},{"id":"95fd179ac6dd1067f748d483171d33d24b6ac2f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-11 17:59:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nUnblocking CI.","accounts_in_message":[],"_revision_number":3},{"id":"a7fcb02ed701ec366bb31a1a68d05e4bc8e91e11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 17:59:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b322eed4350e0b25639942180daaf5cd7c81e37c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 19:43:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8d7d80d168d4fb18d43af97e567324b796376046","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 19:43:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/664533/3/gate/openstack-tox-docs/83aef04/html/ : SUCCESS in 6m 23s\n- openstack-tox-py35 http://logs.openstack.org/33/664533/3/gate/openstack-tox-py35/5d1aba3/ : SUCCESS in 7m 50s\n- openstack-tox-py36 http://logs.openstack.org/33/664533/3/gate/openstack-tox-py36/4fe896f/ : SUCCESS in 8m 03s\n- openstack-tox-pep8 http://logs.openstack.org/33/664533/3/gate/openstack-tox-pep8/843b4a4/ : SUCCESS in 7m 20s\n- openstack-tox-py27 http://logs.openstack.org/33/664533/3/gate/openstack-tox-py27/bf16590/ : SUCCESS in 5m 30s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/33/664533/3/gate/kayobe-tox-ansible-syntax/4486a4d/ : SUCCESS in 5m 05s\n- kayobe-tox-ansible http://logs.openstack.org/33/664533/3/gate/kayobe-tox-ansible/7ecc66c/ : SUCCESS in 8m 56s\n- kayobe-tox-molecule http://logs.openstack.org/33/664533/3/gate/kayobe-tox-molecule/6b4e5f4/ : SUCCESS in 54m 34s\n- kayobe-overcloud-centos http://logs.openstack.org/33/664533/3/gate/kayobe-overcloud-centos/e977c11/ : SUCCESS in 1h 09m 19s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/33/664533/3/gate/kayobe-overcloud-upgrade-centos/9741c7e/ : SUCCESS in 1h 37m 21s\n- kayobe-seed-centos http://logs.openstack.org/33/664533/3/gate/kayobe-seed-centos/859d376/ : SUCCESS in 21m 27s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/33/664533/3/gate/kayobe-seed-upgrade-centos/a30cb49/ : SUCCESS in 28m 39s","accounts_in_message":[],"_revision_number":3},{"id":"c31c786188c5eadb8acc41bee97120eb9a3bd0a0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-12 13:47:46.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 559b69e765f267cf58841fd7ee6dbf93e4ce1519","accounts_in_message":[],"_revision_number":3},{"id":"ae5ac2c4dbdf529b4c8fc7f56d7df3642d341f9d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-13 08:28:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"54fb2105286f9e5b071ad9d0659df75d41b0b67f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-13 08:54:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"686cf8d5419d20499e824d155232ef23bde4535b","revisions":{"0d7468788ab10e2d21d441509d5300164054a9b7":{"kind":"REWORK","_number":1,"created":"2019-06-11 09:29:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/33/664533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/664533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/664533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/664533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/664533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/664533/1"}}},"commit":{"parents":[{"commit":"e884faab61e6f062153dd207f63b768d74f2d779","subject":"Docs: improve package update admin docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e884faab61e6f062153dd207f63b768d74f2d779"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-11 09:29:02.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-11 09:29:02.000000000","tz":60},"subject":"DNM: Test if paramiko 2.5.0 is triggering job failures","message":"DNM: Test if paramiko 2.5.0 is triggering job failures\n\nChange-Id: Ieddf039483649cc7b575019409257b6ce5b3fb99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d7468788ab10e2d21d441509d5300164054a9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d7468788ab10e2d21d441509d5300164054a9b7"}]},"branch":"refs/heads/master"},"e232e5327035527eb525e326995cc7dd52bdb06e":{"kind":"REWORK","_number":2,"created":"2019-06-11 10:57:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/33/664533/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/664533/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/664533/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/664533/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/664533/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/664533/2"}}},"commit":{"parents":[{"commit":"e884faab61e6f062153dd207f63b768d74f2d779","subject":"Docs: improve package update admin docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e884faab61e6f062153dd207f63b768d74f2d779"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-11 10:57:40.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-11 10:57:40.000000000","tz":60},"subject":"DNM: Test if requiring pyasn1 solves job failures","message":"DNM: Test if requiring pyasn1 solves job failures\n\nChange-Id: Ieddf039483649cc7b575019409257b6ce5b3fb99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e232e5327035527eb525e326995cc7dd52bdb06e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e232e5327035527eb525e326995cc7dd52bdb06e"}]},"branch":"refs/heads/master"},"686cf8d5419d20499e824d155232ef23bde4535b":{"kind":"REWORK","_number":3,"created":"2019-06-11 13:53:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/33/664533/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/664533/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/664533/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/664533/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/664533/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/664533/3"}}},"commit":{"parents":[{"commit":"e884faab61e6f062153dd207f63b768d74f2d779","subject":"Docs: improve package update admin docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e884faab61e6f062153dd207f63b768d74f2d779"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-11 13:48:11.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-11 13:48:11.000000000","tz":60},"subject":"Cap paramiko to releases below 2.5.0","message":"Cap paramiko to releases below 2.5.0\n\nThis new version makes the kayobe-tox-molecule job fail with:\n\n    ImportError: cannot import name py31compat\n\nChange-Id: Ieddf039483649cc7b575019409257b6ce5b3fb99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/686cf8d5419d20499e824d155232ef23bde4535b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/686cf8d5419d20499e824d155232ef23bde4535b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
