)]}'
{"id":"openstack%2Fkayobe~664930","triplet_id":"openstack%2Fkayobe~stable%2Frocky~Ieddf039483649cc7b575019409257b6ce5b3fb99","project":"openstack/kayobe","branch":"stable/rocky","topic":"molecule-job-stable/rocky","hashtags":[],"change_id":"Ieddf039483649cc7b575019409257b6ce5b3fb99","subject":"Cap paramiko to releases below 2.5.0","status":"MERGED","created":"2019-06-12 13:47:46.000000000","updated":"2019-06-13 10:25:35.000000000","submitted":"2019-06-13 10:25:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664930-1560421535275-38ffb1f0","meta_rev_id":"d860a71daf384c7e0735e41783b919e32cb0724c","_number":664930,"virtual_id_number":664930,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-13 10:25:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-06-13 08:23:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-12 14:04:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-06-13 08:23:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 14:04:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-06-13 10:25:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3e1513f56c7f5467eff5bf60447373b91ea3620","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-12 13:47:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"745feb38288f165f77fb38bed85e9c9dd1964d69","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-12 14:04:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5b97a8ac17074872dbd1223352a7d11aac1eeab2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 15:52:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/30/664930/1/check/build-openstack-sphinx-docs/de8abd6/html/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/30/664930/1/check/openstack-tox-docs/0d3af03/html/ : SUCCESS in 5m 21s\n- openstack-tox-cover http://logs.openstack.org/30/664930/1/check/openstack-tox-cover/7f4daef/cover/ : SUCCESS in 5m 27s\n- openstack-tox-py35 http://logs.openstack.org/30/664930/1/check/openstack-tox-py35/c74985e/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/30/664930/1/check/openstack-tox-pep8/d4e6b20/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/30/664930/1/check/openstack-tox-py27/f5c1647/ : SUCCESS in 5m 13s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/30/664930/1/check/kayobe-tox-ansible-syntax/c873bb8/ : SUCCESS in 6m 40s\n- kayobe-tox-ansible http://logs.openstack.org/30/664930/1/check/kayobe-tox-ansible/fbc3b76/ : SUCCESS in 11m 07s\n- kayobe-tox-molecule http://logs.openstack.org/30/664930/1/check/kayobe-tox-molecule/70f32eb/ : SUCCESS in 34m 44s\n- kayobe-overcloud-centos http://logs.openstack.org/30/664930/1/check/kayobe-overcloud-centos/868722d/ : SUCCESS in 1h 02m 24s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/30/664930/1/check/kayobe-overcloud-upgrade-centos/42dbe52/ : SUCCESS in 1h 39m 35s\n- kayobe-seed-centos http://logs.openstack.org/30/664930/1/check/kayobe-seed-centos/18f57e1/ : SUCCESS in 15m 46s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/30/664930/1/check/kayobe-seed-upgrade-centos/4385044/ : SUCCESS in 27m 49s","accounts_in_message":[],"_revision_number":1},{"id":"78468f8907ca3284a5e9bdc99c4d0e4a90449de1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-13 08:23:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b249ead50d11789a2d0d861badaabb1ddd8995db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 08:24:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a6bae54ff5f8e93938db11d79c1b44b90d192b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 10:25:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/30/664930/1/gate/build-openstack-sphinx-docs/11934fe/html/ : SUCCESS in 7m 55s\n- openstack-tox-docs http://logs.openstack.org/30/664930/1/gate/openstack-tox-docs/9cb5879/html/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/30/664930/1/gate/openstack-tox-py35/76b9f76/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/30/664930/1/gate/openstack-tox-pep8/064e453/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/30/664930/1/gate/openstack-tox-py27/4a87c19/ : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/30/664930/1/gate/kayobe-tox-ansible-syntax/fa30e90/ : SUCCESS in 5m 08s\n- kayobe-tox-ansible http://logs.openstack.org/30/664930/1/gate/kayobe-tox-ansible/72e17b4/ : SUCCESS in 7m 58s\n- kayobe-tox-molecule http://logs.openstack.org/30/664930/1/gate/kayobe-tox-molecule/b456a27/ : SUCCESS in 48m 00s\n- kayobe-overcloud-centos http://logs.openstack.org/30/664930/1/gate/kayobe-overcloud-centos/722ebd0/ : SUCCESS in 55m 59s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/30/664930/1/gate/kayobe-overcloud-upgrade-centos/c0647c6/ : SUCCESS in 1h 36m 42s\n- kayobe-seed-centos http://logs.openstack.org/30/664930/1/gate/kayobe-seed-centos/6755016/ : SUCCESS in 37m 09s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/30/664930/1/gate/kayobe-seed-upgrade-centos/6ee30d9/ : SUCCESS in 27m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7cd308997737c5023779269fe6fee32b458363f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 10:25:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"559b69e765f267cf58841fd7ee6dbf93e4ce1519","revisions":{"559b69e765f267cf58841fd7ee6dbf93e4ce1519":{"kind":"REWORK","_number":1,"created":"2019-06-12 13:47:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/664930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/664930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/664930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/664930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/664930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/664930/1"}}},"commit":{"parents":[{"commit":"2e7cec1b5afe9e4a69d48d4d0ad95154e295439b","subject":"Rocky: Add missing monasca groups to inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e7cec1b5afe9e4a69d48d4d0ad95154e295439b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-11 13:48:11.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-12 13:47:46.000000000","tz":0},"subject":"Cap paramiko to releases below 2.5.0","message":"Cap paramiko to releases below 2.5.0\n\nThis new version makes the kayobe-tox-molecule job fail with:\n\n    ImportError: cannot import name py31compat\n\nChange-Id: Ieddf039483649cc7b575019409257b6ce5b3fb99\n(cherry picked from commit 686cf8d5419d20499e824d155232ef23bde4535b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/559b69e765f267cf58841fd7ee6dbf93e4ce1519"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/559b69e765f267cf58841fd7ee6dbf93e4ce1519"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
