)]}'
{"id":"openstack%2Fkayobe~665093","triplet_id":"openstack%2Fkayobe~master~Ia7d069bc49037c9961bce2a7814fae5745ae8a91","project":"openstack/kayobe","branch":"master","topic":"molecule-job","hashtags":[],"change_id":"Ia7d069bc49037c9961bce2a7814fae5745ae8a91","subject":"Move paramiko cap to test-requirements.txt","status":"MERGED","created":"2019-06-13 08:55:08.000000000","updated":"2019-07-12 21:28:36.000000000","submitted":"2019-07-12 21:28:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665093-1562966916747-42ebc4e7","meta_rev_id":"de9e2c508388cddce8a36d1a38df7f3eeeeb8816","_number":665093,"virtual_id_number":665093,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-12 21:28:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-07-12 16:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-07-08 16:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-07-12 16:32:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 16:37:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-12 16:32:11.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-12 21:28:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba9884fa05e14fc669ddedfc0dce5a4a8d786bd5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-13 08:55:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55fb231c9b7a234f213e83808f7b3bfb8b695973","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 10:43:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/93/665093/1/check/openstack-tox-docs/3a5439b/html/ : SUCCESS in 4m 37s\n- openstack-tox-cover http://logs.openstack.org/93/665093/1/check/openstack-tox-cover/860ae73/cover/ : SUCCESS in 7m 07s\n- openstack-tox-py35 http://logs.openstack.org/93/665093/1/check/openstack-tox-py35/1794054/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/93/665093/1/check/openstack-tox-py36/3ce4fdb/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/93/665093/1/check/openstack-tox-pep8/4df3812/ : SUCCESS in 4m 50s\n- openstack-tox-py27 http://logs.openstack.org/93/665093/1/check/openstack-tox-py27/1f245ec/ : SUCCESS in 5m 29s\n- build-openstack-releasenotes http://logs.openstack.org/93/665093/1/check/build-openstack-releasenotes/b9d8089/html/ : SUCCESS in 5m 09s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/93/665093/1/check/kayobe-tox-ansible-syntax/19baf27/ : SUCCESS in 5m 09s\n- kayobe-tox-ansible http://logs.openstack.org/93/665093/1/check/kayobe-tox-ansible/8d32e2a/ : SUCCESS in 9m 29s\n- kayobe-tox-molecule http://logs.openstack.org/93/665093/1/check/kayobe-tox-molecule/bc5685b/ : SUCCESS in 53m 07s\n- kayobe-overcloud-centos http://logs.openstack.org/93/665093/1/check/kayobe-overcloud-centos/71e5ac6/ : SUCCESS in 1h 05m 38s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/93/665093/1/check/kayobe-overcloud-upgrade-centos/febe7ac/ : SUCCESS in 1h 40m 04s\n- kayobe-seed-centos http://logs.openstack.org/93/665093/1/check/kayobe-seed-centos/5696a54/ : SUCCESS in 24m 33s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/93/665093/1/check/kayobe-seed-upgrade-centos/05864e0/ : FAILURE in 22m 51s","accounts_in_message":[],"_revision_number":1},{"id":"aaef616bfa9d307d5cc5927c543a6fa7330b3626","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 16:13:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5adaa15a135d1ab10eb25a525126f89a23e537cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 17:24:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/93/665093/1/check/openstack-tox-docs/7666034/html/ : SUCCESS in 4m 53s\n- openstack-tox-cover http://logs.openstack.org/93/665093/1/check/openstack-tox-cover/40c43a0/ : FAILURE in 5m 37s\n- openstack-tox-py35 http://logs.openstack.org/93/665093/1/check/openstack-tox-py35/c9ad980/ : FAILURE in 5m 12s\n- openstack-tox-py36 http://logs.openstack.org/93/665093/1/check/openstack-tox-py36/b599d8e/ : FAILURE in 5m 35s\n- openstack-tox-pep8 http://logs.openstack.org/93/665093/1/check/openstack-tox-pep8/a0db776/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/93/665093/1/check/openstack-tox-py27/d29493b/ : FAILURE in 5m 17s\n- build-openstack-releasenotes http://logs.openstack.org/93/665093/1/check/build-openstack-releasenotes/d1e984f/html/ : SUCCESS in 5m 58s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/93/665093/1/check/kayobe-tox-ansible-syntax/7f17f14/ : SUCCESS in 5m 52s\n- kayobe-tox-ansible http://logs.openstack.org/93/665093/1/check/kayobe-tox-ansible/d1a6191/ : SUCCESS in 10m 45s\n- kayobe-tox-molecule http://logs.openstack.org/93/665093/1/check/kayobe-tox-molecule/fdfbf77/ : SUCCESS in 54m 01s\n- kayobe-overcloud-centos http://logs.openstack.org/93/665093/1/check/kayobe-overcloud-centos/fa05066/ : RETRY_LIMIT in 5m 01s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/93/665093/1/check/kayobe-overcloud-upgrade-centos/ce52c0c/ : FAILURE in 3m 44s\n- kayobe-seed-centos http://logs.openstack.org/93/665093/1/check/kayobe-seed-centos/90a2479/ : FAILURE in 5m 50s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/93/665093/1/check/kayobe-seed-upgrade-centos/cba52dc/ : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a9262d7e0bafc2ff23b423040cf0733b9ad5b3fc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-20 11:53:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"917af08d41b97e1059437e2d4e060504e0dbae23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 13:55:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/93/665093/1/check/openstack-tox-docs/ab32100/html/ : SUCCESS in 4m 41s\n- openstack-tox-cover http://logs.openstack.org/93/665093/1/check/openstack-tox-cover/c25da71/cover/ : SUCCESS in 7m 08s\n- openstack-tox-py35 http://logs.openstack.org/93/665093/1/check/openstack-tox-py35/b9b2bb8/ : SUCCESS in 18m 17s\n- openstack-tox-py36 http://logs.openstack.org/93/665093/1/check/openstack-tox-py36/c671cb4/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/93/665093/1/check/openstack-tox-pep8/59b007a/ : SUCCESS in 5m 27s\n- openstack-tox-py27 http://logs.openstack.org/93/665093/1/check/openstack-tox-py27/d9775b1/ : SUCCESS in 5m 25s\n- build-openstack-releasenotes http://logs.openstack.org/93/665093/1/check/build-openstack-releasenotes/14aa73a/html/ : SUCCESS in 6m 21s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/93/665093/1/check/kayobe-tox-ansible-syntax/f5d0366/ : SUCCESS in 5m 17s\n- kayobe-tox-ansible http://logs.openstack.org/93/665093/1/check/kayobe-tox-ansible/43f2f1a/ : SUCCESS in 9m 32s\n- kayobe-tox-molecule http://logs.openstack.org/93/665093/1/check/kayobe-tox-molecule/bf5f85a/ : SUCCESS in 55m 27s\n- kayobe-overcloud-centos http://logs.openstack.org/93/665093/1/check/kayobe-overcloud-centos/b5017df/ : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/93/665093/1/check/kayobe-overcloud-upgrade-centos/b8e3ec3/ : SUCCESS in 1h 41m 26s\n- kayobe-seed-centos http://logs.openstack.org/93/665093/1/check/kayobe-seed-centos/829342c/ : SUCCESS in 20m 46s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/93/665093/1/check/kayobe-seed-upgrade-centos/1149d6b/ : FAILURE in 26m 21s","accounts_in_message":[],"_revision_number":1},{"id":"64f78c569bab8b0507ba90bedb71f17cb1a86048","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-08 14:15:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"302d48c3bf7ee7a9b5195157d5ca8937eaa912ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:31:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/93/665093/2/check/openstack-tox-docs/ebf2ac3/html/ : SUCCESS in 4m 48s\n- openstack-tox-cover http://logs.openstack.org/93/665093/2/check/openstack-tox-cover/f7fc72e/cover/ : SUCCESS in 5m 24s\n- openstack-tox-py35 http://logs.openstack.org/93/665093/2/check/openstack-tox-py35/373a4e5/ : SUCCESS in 5m 09s\n- openstack-tox-py36 http://logs.openstack.org/93/665093/2/check/openstack-tox-py36/32f550b/ : SUCCESS in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/93/665093/2/check/openstack-tox-pep8/59dfdb5/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/93/665093/2/check/openstack-tox-py27/322682c/ : SUCCESS in 5m 40s\n- build-openstack-releasenotes http://logs.openstack.org/93/665093/2/check/build-openstack-releasenotes/3c07192/html/ : SUCCESS in 4m 12s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/93/665093/2/check/kayobe-tox-ansible-syntax/8e464d2/ : SUCCESS in 5m 24s\n- kayobe-tox-ansible http://logs.openstack.org/93/665093/2/check/kayobe-tox-ansible/443166c/ : SUCCESS in 10m 20s\n- kayobe-tox-molecule http://logs.openstack.org/93/665093/2/check/kayobe-tox-molecule/f9931c3/ : SUCCESS in 54m 22s\n- kayobe-overcloud-centos http://logs.openstack.org/93/665093/2/check/kayobe-overcloud-centos/0201cf6/ : SUCCESS in 1h 25m 21s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/93/665093/2/check/kayobe-overcloud-upgrade-centos/7c55179/ : SUCCESS in 2h 08m 13s\n- kayobe-seed-centos http://logs.openstack.org/93/665093/2/check/kayobe-seed-centos/f9bcf84/ : SUCCESS in 17m 54s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/93/665093/2/check/kayobe-seed-upgrade-centos/7271226/ : SUCCESS in 27m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4cde100474363bcaf7992cd428f8c21640fbac6b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 16:37:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d0b16e2cd3a4ed177ed3f029a10129303b481f2a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-12 16:32:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"74e8a47cf37ecfa86666f958ff734fd49b9074fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 16:32:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1505c4516bc4990711cd3a375139388c1a0818c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 21:28:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2bb5331f67cd86235591e73c60e18607250ca7c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 21:28:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/93/665093/2/gate/openstack-tox-docs/945fa33/html/ : SUCCESS in 2m 50s\n- openstack-tox-py35 http://logs.openstack.org/93/665093/2/gate/openstack-tox-py35/c324427/ : SUCCESS in 6m 01s\n- openstack-tox-py36 http://logs.openstack.org/93/665093/2/gate/openstack-tox-py36/7636342/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/93/665093/2/gate/openstack-tox-pep8/a2e73bb/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/93/665093/2/gate/openstack-tox-py27/1a4b233/ : SUCCESS in 5m 52s\n- build-openstack-releasenotes http://logs.openstack.org/93/665093/2/gate/build-openstack-releasenotes/0e4cf7c/html/ : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/93/665093/2/gate/kayobe-tox-ansible-syntax/89c9b76/ : SUCCESS in 4m 01s\n- kayobe-tox-ansible http://logs.openstack.org/93/665093/2/gate/kayobe-tox-ansible/ecc342c/ : SUCCESS in 10m 57s\n- kayobe-tox-molecule http://logs.openstack.org/93/665093/2/gate/kayobe-tox-molecule/b0ef429/ : SUCCESS in 51m 57s\n- kayobe-overcloud-centos http://logs.openstack.org/93/665093/2/gate/kayobe-overcloud-centos/7dda4d7/ : SUCCESS in 1h 08m 49s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/93/665093/2/gate/kayobe-overcloud-upgrade-centos/41ce6ac/ : SUCCESS in 1h 59m 08s\n- kayobe-seed-centos http://logs.openstack.org/93/665093/2/gate/kayobe-seed-centos/5b1fde5/ : SUCCESS in 24m 00s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/93/665093/2/gate/kayobe-seed-upgrade-centos/fcac989/ : SUCCESS in 23m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2863c738614ee8a47b41d6e81d02a67b1a20749b","revisions":{"b029375476c1018dc5b75fbf4bdc45a06f4a5217":{"kind":"REWORK","_number":1,"created":"2019-06-13 08:55:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/93/665093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/665093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/665093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/665093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/665093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/665093/1"}}},"commit":{"parents":[{"commit":"4a45d75560502df695090a66ad36dfd58ab4dfbb","subject":"Merge \"Build logstash when monasca is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a45d75560502df695090a66ad36dfd58ab4dfbb"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-13 08:54:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-13 08:54:52.000000000","tz":60},"subject":"Move paramiko cap to test-requirements.txt","message":"Move paramiko cap to test-requirements.txt\n\nChange-Id: Ia7d069bc49037c9961bce2a7814fae5745ae8a91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b029375476c1018dc5b75fbf4bdc45a06f4a5217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b029375476c1018dc5b75fbf4bdc45a06f4a5217"}]},"branch":"refs/heads/master"},"2863c738614ee8a47b41d6e81d02a67b1a20749b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-08 14:15:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/93/665093/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/665093/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/665093/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/665093/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/665093/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/665093/2"}}},"commit":{"parents":[{"commit":"b1f54eabe13988b7091f85ebcd98d304d314798e","subject":"Merge \"Add retries to ansible galaxy install for all envs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1f54eabe13988b7091f85ebcd98d304d314798e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-13 08:54:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-08 14:15:19.000000000","tz":0},"subject":"Move paramiko cap to test-requirements.txt","message":"Move paramiko cap to test-requirements.txt\n\nChange-Id: Ia7d069bc49037c9961bce2a7814fae5745ae8a91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2863c738614ee8a47b41d6e81d02a67b1a20749b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2863c738614ee8a47b41d6e81d02a67b1a20749b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
