)]}'
{"id":"openstack%2Fkayobe~665705","triplet_id":"openstack%2Fkayobe~stable%2Fqueens~I7966cf33f6133090d83232504b8c1da4c7817e0c","project":"openstack/kayobe","branch":"stable/queens","topic":"story/2005891-stable-queens","hashtags":[],"change_id":"I7966cf33f6133090d83232504b8c1da4c7817e0c","subject":"Restrict cliff to \u003c2.15.0","status":"MERGED","created":"2019-06-17 15:14:27.000000000","updated":"2019-06-18 19:28:27.000000000","submitted":"2019-06-18 19:28:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665705-1560886107678-d3b36feb","meta_rev_id":"2816880743b699190a4dce80f2583bf005df68a5","_number":665705,"virtual_id_number":665705,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-06-18 19:28:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-06-18 18:10:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-18 16:27:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-06-18 18:10:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 16:27:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-06-18 19:28:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdb667dd31f3073614e023148a458443b0e75ec6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 15:14:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3c4c9e4f44c931145434c5e955cdf3c18479664","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-17 15:15:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"095b6116ee1bb199013cb15c3bdb2d6533f7f032","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 15:18:05.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0e749d5bc18ba152445f973d57bf2dcd7af088ad","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-17 15:45:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"664c2ca5186856ef0e16ecbcddb68e30975bcbef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 16:22:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/05/665705/2/check/build-openstack-sphinx-docs/c032150/html/ : SUCCESS in 5m 37s\n- openstack-tox-cover http://logs.openstack.org/05/665705/2/check/openstack-tox-cover/81bdcdc/cover/ : SUCCESS in 5m 29s\n- openstack-tox-py35 http://logs.openstack.org/05/665705/2/check/openstack-tox-py35/15a8978/ : SUCCESS in 5m 28s\n- openstack-tox-pep8 http://logs.openstack.org/05/665705/2/check/openstack-tox-pep8/bbe40fb/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/05/665705/2/check/openstack-tox-py27/b2468d9/ : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/05/665705/2/check/kayobe-tox-ansible-syntax/c40ed53/ : SUCCESS in 6m 21s\n- kayobe-tox-ansible http://logs.openstack.org/05/665705/2/check/kayobe-tox-ansible/dd26ff4/ : SUCCESS in 9m 07s\n- kayobe-tox-molecule http://logs.openstack.org/05/665705/2/check/kayobe-tox-molecule/744ae19/ : FAILURE in 8m 59s\n- kayobe-overcloud-centos http://logs.openstack.org/05/665705/2/check/kayobe-overcloud-centos/414afaf/ : SUCCESS in 54m 12s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/05/665705/2/check/kayobe-overcloud-upgrade-centos/356c3ae/ : FAILURE in 3m 27s\n- kayobe-seed-centos http://logs.openstack.org/05/665705/2/check/kayobe-seed-centos/9003792/ : SUCCESS in 20m 52s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/05/665705/2/check/kayobe-seed-upgrade-centos/5c590ff/ : FAILURE in 5m 37s","accounts_in_message":[],"_revision_number":2},{"id":"56e92b706a85c6866617471d2159a6af26e03591","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-17 18:27:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ce889083d8de807e1d4757f02040d9658c10ee40","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-17 18:51:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9ca71e17a71c220e4868f65ea9a206561a80e274","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 19:56:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/05/665705/3/check/build-openstack-sphinx-docs/a16feb1/html/ : SUCCESS in 5m 55s\n- openstack-tox-cover http://logs.openstack.org/05/665705/3/check/openstack-tox-cover/eb9b650/cover/ : SUCCESS in 6m 44s\n- openstack-tox-py35 http://logs.openstack.org/05/665705/3/check/openstack-tox-py35/b4b28c5/ : SUCCESS in 4m 41s\n- openstack-tox-pep8 http://logs.openstack.org/05/665705/3/check/openstack-tox-pep8/22e7349/ : SUCCESS in 6m 15s\n- openstack-tox-py27 http://logs.openstack.org/05/665705/3/check/openstack-tox-py27/b0bca7f/ : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/05/665705/3/check/kayobe-tox-ansible-syntax/7ba9bd8/ : SUCCESS in 6m 08s\n- kayobe-tox-ansible http://logs.openstack.org/05/665705/3/check/kayobe-tox-ansible/92f370f/ : SUCCESS in 9m 58s\n- kayobe-tox-molecule http://logs.openstack.org/05/665705/3/check/kayobe-tox-molecule/efd246d/ : SUCCESS in 46m 35s\n- kayobe-overcloud-centos http://logs.openstack.org/05/665705/3/check/kayobe-overcloud-centos/f1fc1b2/ : SUCCESS in 48m 17s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/05/665705/3/check/kayobe-overcloud-upgrade-centos/038af16/ : SUCCESS in 1h 22m 06s\n- kayobe-seed-centos http://logs.openstack.org/05/665705/3/check/kayobe-seed-centos/84e8698/ : SUCCESS in 17m 53s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/05/665705/3/check/kayobe-seed-upgrade-centos/14440ba/ : RETRY_LIMIT in 2m 23s","accounts_in_message":[],"_revision_number":3},{"id":"5e7da2dee5bcd4d2b48329efac78e5dc9027d559","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-18 09:02:12.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e28ee42601ec2b25cc6b69ba4af2acfc55cae07c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-18 09:23:54.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"32f3796a6a61973809d56e91cddad4ba4aa1fa0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 11:03:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/05/665705/4/check/build-openstack-sphinx-docs/5b0b5d5/html/ : SUCCESS in 4m 28s\n- openstack-tox-cover http://logs.openstack.org/05/665705/4/check/openstack-tox-cover/4ab7cc5/cover/ : SUCCESS in 5m 24s\n- openstack-tox-py35 http://logs.openstack.org/05/665705/4/check/openstack-tox-py35/35a1162/ : SUCCESS in 5m 57s\n- openstack-tox-pep8 http://logs.openstack.org/05/665705/4/check/openstack-tox-pep8/e8d6805/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/05/665705/4/check/openstack-tox-py27/bb4b183/ : SUCCESS in 4m 55s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/05/665705/4/check/kayobe-tox-ansible-syntax/93726c1/ : SUCCESS in 6m 38s\n- kayobe-tox-ansible http://logs.openstack.org/05/665705/4/check/kayobe-tox-ansible/0367f30/ : SUCCESS in 8m 14s\n- kayobe-tox-molecule http://logs.openstack.org/05/665705/4/check/kayobe-tox-molecule/1a12cd2/ : SUCCESS in 45m 39s\n- kayobe-overcloud-centos http://logs.openstack.org/05/665705/4/check/kayobe-overcloud-centos/d6f4a67/ : TIMED_OUT in 1h 31m 00s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/05/665705/4/check/kayobe-overcloud-upgrade-centos/2591627/ : SUCCESS in 1h 17m 37s\n- kayobe-seed-centos http://logs.openstack.org/05/665705/4/check/kayobe-seed-centos/ef01649/ : SUCCESS in 16m 54s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/05/665705/4/check/kayobe-seed-upgrade-centos/24551ef/ : SUCCESS in 26m 35s","accounts_in_message":[],"_revision_number":4},{"id":"a67bb46a73e1eacc23eeef0fa86b2b26da5c87bd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-18 11:21:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c1c3cfb59d2d04ca5ff5e9be0d071255d2188c79","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-18 12:32:34.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f41721fa86d6bce0a997ffd539db9d10fc24ab09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 12:55:47.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/05/665705/4/check/build-openstack-sphinx-docs/aed7819/html/ : SUCCESS in 5m 50s\n- openstack-tox-cover http://logs.openstack.org/05/665705/4/check/openstack-tox-cover/7c936db/cover/ : SUCCESS in 5m 42s\n- openstack-tox-py35 http://logs.openstack.org/05/665705/4/check/openstack-tox-py35/3e0d0be/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/05/665705/4/check/openstack-tox-pep8/4d784a1/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/05/665705/4/check/openstack-tox-py27/dcbf712/ : SUCCESS in 5m 21s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/05/665705/4/check/kayobe-tox-ansible-syntax/2720b9b/ : SUCCESS in 5m 40s\n- kayobe-tox-ansible http://logs.openstack.org/05/665705/4/check/kayobe-tox-ansible/f6084cc/ : SUCCESS in 11m 41s\n- kayobe-tox-molecule http://logs.openstack.org/05/665705/4/check/kayobe-tox-molecule/f111e5c/ : SUCCESS in 48m 19s\n- kayobe-overcloud-centos http://logs.openstack.org/05/665705/4/check/kayobe-overcloud-centos/8be4b26/ : RETRY_LIMIT in 51m 05s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/05/665705/4/check/kayobe-overcloud-upgrade-centos/fb55a20/ : SUCCESS in 1h 20m 00s\n- kayobe-seed-centos http://logs.openstack.org/05/665705/4/check/kayobe-seed-centos/363fad0/ : SUCCESS in 17m 59s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/05/665705/4/check/kayobe-seed-upgrade-centos/712c811/ : SUCCESS in 27m 17s","accounts_in_message":[],"_revision_number":4},{"id":"2dae47f5802156fa382829830bcc286a48f1f6a4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-18 13:19:22.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"617e8506791cf85b28960cf499367800c10adc87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 14:34:55.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/05/665705/4/check/build-openstack-sphinx-docs/1c55f1c/html/ : SUCCESS in 5m 09s\n- openstack-tox-cover http://logs.openstack.org/05/665705/4/check/openstack-tox-cover/c98f4eb/cover/ : SUCCESS in 4m 52s\n- openstack-tox-py35 http://logs.openstack.org/05/665705/4/check/openstack-tox-py35/e316788/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/05/665705/4/check/openstack-tox-pep8/49622d7/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/05/665705/4/check/openstack-tox-py27/d59c9ce/ : SUCCESS in 4m 57s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/05/665705/4/check/kayobe-tox-ansible-syntax/ebcbba7/ : SUCCESS in 5m 51s\n- kayobe-tox-ansible http://logs.openstack.org/05/665705/4/check/kayobe-tox-ansible/6e8b73d/ : SUCCESS in 9m 07s\n- kayobe-tox-molecule http://logs.openstack.org/05/665705/4/check/kayobe-tox-molecule/a91b4cd/ : SUCCESS in 31m 49s\n- kayobe-overcloud-centos http://logs.openstack.org/05/665705/4/check/kayobe-overcloud-centos/6d7190c/ : RETRY_LIMIT in 42m 04s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/05/665705/4/check/kayobe-overcloud-upgrade-centos/5c8ec4d/ : FAILURE in 49m 54s\n- kayobe-seed-centos http://logs.openstack.org/05/665705/4/check/kayobe-seed-centos/b8ac04e/ : RETRY_LIMIT in 44m 49s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/05/665705/4/check/kayobe-seed-upgrade-centos/50325bd/ : SUCCESS in 1h 11m 38s","accounts_in_message":[],"_revision_number":4},{"id":"5e600097c6b867aa369571ce2c86d71e721cbe05","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-18 14:43:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6c95ea8c8308da032a30d934b637e888da75fdde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 16:03:11.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/05/665705/4/check/build-openstack-sphinx-docs/ab18c80/html/ : SUCCESS in 5m 58s\n- openstack-tox-cover http://logs.openstack.org/05/665705/4/check/openstack-tox-cover/4efdb06/cover/ : SUCCESS in 5m 28s\n- openstack-tox-py35 http://logs.openstack.org/05/665705/4/check/openstack-tox-py35/17ed084/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/05/665705/4/check/openstack-tox-pep8/2e6c3d9/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/05/665705/4/check/openstack-tox-py27/4ec8f4d/ : SUCCESS in 4m 49s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/05/665705/4/check/kayobe-tox-ansible-syntax/de1c4fc/ : FAILURE in 5m 03s\n- kayobe-tox-ansible http://logs.openstack.org/05/665705/4/check/kayobe-tox-ansible/d053538/ : SUCCESS in 9m 10s\n- kayobe-tox-molecule http://logs.openstack.org/05/665705/4/check/kayobe-tox-molecule/07a7e8b/ : SUCCESS in 47m 06s\n- kayobe-overcloud-centos http://logs.openstack.org/05/665705/4/check/kayobe-overcloud-centos/7df9c6e/ : SUCCESS in 49m 22s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/05/665705/4/check/kayobe-overcloud-upgrade-centos/980402c/ : SUCCESS in 1h 13m 33s\n- kayobe-seed-centos http://logs.openstack.org/05/665705/4/check/kayobe-seed-centos/886218c/ : SUCCESS in 18m 29s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/05/665705/4/check/kayobe-seed-upgrade-centos/c8f1216/ : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":4},{"id":"80a3d2157b57db02bf906c08cc5e3d93ade12975","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-06-18 16:27:32.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck\n\n- downloading role from https://github.com/mrlesmithjr/ansible-manage-lvm/archive/v0.1.1.tar.gz\n [ERROR]: failed to download the file: HTTP Error 502: Bad Gateway","accounts_in_message":[],"_revision_number":4},{"id":"568b053fe0e0f0643056b9f4612c238fa05855eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 18:02:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/05/665705/4/check/build-openstack-sphinx-docs/9855d93/html/ : SUCCESS in 5m 02s\n- openstack-tox-cover http://logs.openstack.org/05/665705/4/check/openstack-tox-cover/17b9eba/cover/ : SUCCESS in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/05/665705/4/check/openstack-tox-py35/130a863/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/05/665705/4/check/openstack-tox-pep8/9d1f56c/ : SUCCESS in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/05/665705/4/check/openstack-tox-py27/dc9e29e/ : SUCCESS in 5m 17s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/05/665705/4/check/kayobe-tox-ansible-syntax/f2c9077/ : SUCCESS in 5m 50s\n- kayobe-tox-ansible http://logs.openstack.org/05/665705/4/check/kayobe-tox-ansible/f141657/ : SUCCESS in 8m 19s\n- kayobe-tox-molecule http://logs.openstack.org/05/665705/4/check/kayobe-tox-molecule/e0c542a/ : SUCCESS in 39m 26s\n- kayobe-overcloud-centos http://logs.openstack.org/05/665705/4/check/kayobe-overcloud-centos/816a32b/ : SUCCESS in 43m 20s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/05/665705/4/check/kayobe-overcloud-upgrade-centos/66d63ff/ : SUCCESS in 1h 10m 08s\n- kayobe-seed-centos http://logs.openstack.org/05/665705/4/check/kayobe-seed-centos/a8d634c/ : SUCCESS in 19m 19s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/05/665705/4/check/kayobe-seed-upgrade-centos/0dacf29/ : SUCCESS in 25m 11s","accounts_in_message":[],"_revision_number":4},{"id":"17db676481a985bad25f16d8f9e685de2daa8afa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-18 18:10:57.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8bdd02c3443c2f23124eb751813c85ab78135b6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 18:11:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"25de55bb65efa857df91414422395d46fdbe4746","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 19:28:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"dd96644a484f709bd1f91758f9a0055b9c7d4a79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 19:28:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/05/665705/4/gate/build-openstack-sphinx-docs/e5b044c/html/ : SUCCESS in 4m 47s\n- openstack-tox-py35 http://logs.openstack.org/05/665705/4/gate/openstack-tox-py35/1d8bddb/ : SUCCESS in 5m 18s\n- openstack-tox-pep8 http://logs.openstack.org/05/665705/4/gate/openstack-tox-pep8/26ad2e8/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/05/665705/4/gate/openstack-tox-py27/a4efb12/ : SUCCESS in 5m 03s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/05/665705/4/gate/kayobe-tox-ansible-syntax/237185c/ : SUCCESS in 5m 36s\n- kayobe-tox-ansible http://logs.openstack.org/05/665705/4/gate/kayobe-tox-ansible/ef4f47c/ : SUCCESS in 8m 51s\n- kayobe-tox-molecule http://logs.openstack.org/05/665705/4/gate/kayobe-tox-molecule/d351298/ : SUCCESS in 48m 12s\n- kayobe-overcloud-centos http://logs.openstack.org/05/665705/4/gate/kayobe-overcloud-centos/9a424f2/ : SUCCESS in 49m 05s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/05/665705/4/gate/kayobe-overcloud-upgrade-centos/2034e5f/ : SUCCESS in 1h 11m 17s\n- kayobe-seed-centos http://logs.openstack.org/05/665705/4/gate/kayobe-seed-centos/d818d66/ : SUCCESS in 18m 42s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/05/665705/4/gate/kayobe-seed-upgrade-centos/e1522d7/ : SUCCESS in 25m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c5e6de8eb1787b722fb66c11a7dbab622748a967","revisions":{"d85e65a454b2510afd2ddf15b8c6314cafbc76cc":{"kind":"REWORK","_number":1,"created":"2019-06-17 15:14:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/665705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/665705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/665705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/665705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/665705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/665705/1"}}},"commit":{"parents":[{"commit":"2ebec65132147550915ef2a0ea46b03169cfc2bd","subject":"Merge \"Update git.openstack.org to opendev.org\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ebec65132147550915ef2a0ea46b03169cfc2bd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-14 13:28:46.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-17 15:13:58.000000000","tz":60},"subject":"Restrict cliff to \u003c2.15.0","message":"Restrict cliff to \u003c2.15.0\n\nUsing Kayobe with cliff 2.15.0 fails with the following error:\n\nkayobe playbook run playbook.yml\ninvalid conflict_resolution value: \u0027ignore\u0027\n\nThis change restricts cliff to \u003c2.15.0.\n\nChange-Id: I7966cf33f6133090d83232504b8c1da4c7817e0c\nStory: 2005891\nTask: 33748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d85e65a454b2510afd2ddf15b8c6314cafbc76cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d85e65a454b2510afd2ddf15b8c6314cafbc76cc"}]},"branch":"refs/heads/stable/queens"},"00e08ae39433131004e3b767c2093626c4e58621":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-17 15:18:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/665705/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/665705/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/665705/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/665705/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/665705/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/665705/2"}}},"commit":{"parents":[{"commit":"2ebec65132147550915ef2a0ea46b03169cfc2bd","subject":"Merge \"Update git.openstack.org to opendev.org\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ebec65132147550915ef2a0ea46b03169cfc2bd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-14 13:28:46.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-17 15:17:07.000000000","tz":0},"subject":"Restrict cliff to \u003c2.15.0","message":"Restrict cliff to \u003c2.15.0\n\nUsing Kayobe with cliff 2.15.0 fails with the following error:\n\nkayobe playbook run playbook.yml\ninvalid conflict_resolution value: \u0027ignore\u0027\n\nThis change restricts cliff to \u003c2.15.0.\n\nChange-Id: I7966cf33f6133090d83232504b8c1da4c7817e0c\nDepends-On: https://review.opendev.org/665706/\nStory: 2005891\nTask: 33748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/00e08ae39433131004e3b767c2093626c4e58621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/00e08ae39433131004e3b767c2093626c4e58621"}]},"branch":"refs/heads/stable/queens"},"ac64c0a3e5002c8d3f8f10185277bd262c61c4fe":{"kind":"REWORK","_number":3,"created":"2019-06-17 18:27:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/665705/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/665705/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/665705/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/665705/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/665705/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/665705/3"}}},"commit":{"parents":[{"commit":"2ebec65132147550915ef2a0ea46b03169cfc2bd","subject":"Merge \"Update git.openstack.org to opendev.org\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ebec65132147550915ef2a0ea46b03169cfc2bd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-14 13:28:46.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-17 18:25:00.000000000","tz":60},"subject":"Restrict cliff to \u003c2.15.0","message":"Restrict cliff to \u003c2.15.0\n\nUsing Kayobe with cliff 2.15.0 fails with the following error:\n\nkayobe playbook run playbook.yml\ninvalid conflict_resolution value: \u0027ignore\u0027\n\nThis change restricts cliff to \u003c2.15.0.\n\nAlso, fix git cloning previous release repos in CI to allow us to merge\nthis change.\n\nZuul provides repositories with all branches containing all prospective\nchanges, not just the current branch. We can use this in the upgrade\njobs to include changes to the previous release branch when a patch\ndepends on them.\n\nChange-Id: I7966cf33f6133090d83232504b8c1da4c7817e0c\nDepends-On: https://review.opendev.org/665706/\nStory: 2005891\nTask: 33748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac64c0a3e5002c8d3f8f10185277bd262c61c4fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac64c0a3e5002c8d3f8f10185277bd262c61c4fe"}]},"branch":"refs/heads/stable/queens"},"c5e6de8eb1787b722fb66c11a7dbab622748a967":{"kind":"REWORK","_number":4,"created":"2019-06-18 09:23:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/05/665705/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/665705/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/665705/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/665705/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/665705/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/665705/4"}}},"commit":{"parents":[{"commit":"2ebec65132147550915ef2a0ea46b03169cfc2bd","subject":"Merge \"Update git.openstack.org to opendev.org\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ebec65132147550915ef2a0ea46b03169cfc2bd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-14 13:28:46.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-18 09:23:52.000000000","tz":0},"subject":"Restrict cliff to \u003c2.15.0","message":"Restrict cliff to \u003c2.15.0\n\nUsing Kayobe with cliff 2.15.0 fails with the following error:\n\nkayobe playbook run playbook.yml\ninvalid conflict_resolution value: \u0027ignore\u0027\n\nThis change restricts cliff to \u003c2.15.0.\n\nAlso, fix git cloning previous release repos in CI to allow us to merge\nthis change.\n\nZuul provides repositories with all branches containing all prospective\nchanges, not just the current branch. We can use this in the upgrade\njobs to include changes to the previous release branch when a patch\ndepends on them.\n\nChange-Id: I7966cf33f6133090d83232504b8c1da4c7817e0c\nDepends-On: https://review.opendev.org/665706/\nStory: 2005891\nTask: 33748\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c5e6de8eb1787b722fb66c11a7dbab622748a967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c5e6de8eb1787b722fb66c11a7dbab622748a967"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
