)]}'
{"id":"openstack%2Fkayobe~666812","triplet_id":"openstack%2Fkayobe~master~I10c2b872887a47fab1a373fc0525dcd6fc78f7e2","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I10c2b872887a47fab1a373fc0525dcd6fc78f7e2","subject":"Execute kayobe in verbose mode in CI","status":"MERGED","created":"2019-06-21 10:38:33.000000000","updated":"2019-08-30 11:44:26.000000000","submitted":"2019-08-30 11:41:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666812-1567165279644-fec2d007","meta_rev_id":"4cd6c8a78aeff9bdbfd2665f04d7382435ce3b89","_number":666812,"virtual_id_number":666812,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2019-08-27 10:37:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-30 11:41:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-08-27 12:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-30 09:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-30 09:43:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-21 12:21:18.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-08-27 12:50:37.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-08-30 09:43:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-08-30 11:41:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f4bbe72bda36d4dc37c0c3ea6034f833f7b991b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-21 10:38:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b55e7eba86181d1c3389e017dbf1309dc0bcd1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 11:56:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/12/666812/1/check/openstack-tox-docs/d4ba2f3/html/ : SUCCESS in 6m 54s\n- openstack-tox-cover http://logs.openstack.org/12/666812/1/check/openstack-tox-cover/a2b1ecd/cover/ : SUCCESS in 7m 24s\n- openstack-tox-py35 http://logs.openstack.org/12/666812/1/check/openstack-tox-py35/1ffc3af/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/12/666812/1/check/openstack-tox-py36/3bdbf09/ : SUCCESS in 6m 33s\n- openstack-tox-pep8 http://logs.openstack.org/12/666812/1/check/openstack-tox-pep8/da73914/ : SUCCESS in 6m 30s\n- openstack-tox-py27 http://logs.openstack.org/12/666812/1/check/openstack-tox-py27/14ddb02/ : SUCCESS in 6m 56s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/12/666812/1/check/kayobe-tox-ansible-syntax/336ebd2/ : SUCCESS in 5m 40s\n- kayobe-tox-ansible http://logs.openstack.org/12/666812/1/check/kayobe-tox-ansible/5960c93/ : SUCCESS in 10m 49s\n- kayobe-tox-molecule http://logs.openstack.org/12/666812/1/check/kayobe-tox-molecule/148e368/ : SUCCESS in 51m 29s\n- kayobe-overcloud-centos http://logs.openstack.org/12/666812/1/check/kayobe-overcloud-centos/a2e93be/ : SUCCESS in 1h 13m 45s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/12/666812/1/check/kayobe-overcloud-upgrade-centos/51ada58/ : RETRY_LIMIT in 2m 14s\n- kayobe-seed-centos http://logs.openstack.org/12/666812/1/check/kayobe-seed-centos/9bc1830/ : SUCCESS in 22m 59s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/12/666812/1/check/kayobe-seed-upgrade-centos/eae222b/ : RETRY_LIMIT in 2m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b6c087f1f523f6c0890d98537e6005ce9afa144b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-06-21 12:21:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e24dbd98c8e30ed96cd213becfcf8fda08a1a53","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-21 13:34:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d7a2f2d32a8da5a7fc28bb74382f09e2db83cbc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-21 14:10:31.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"4f437b922315eac3c66c338675a419e81c99e268","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-21 14:18:15.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"4e30a40142ba173337f3968f135e4aff3384caf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 16:27:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/12/666812/4/check/openstack-tox-docs/7d3c5f2/html/ : SUCCESS in 5m 29s\n- openstack-tox-cover http://logs.openstack.org/12/666812/4/check/openstack-tox-cover/e8b9f3a/cover/ : SUCCESS in 6m 46s\n- openstack-tox-py35 http://logs.openstack.org/12/666812/4/check/openstack-tox-py35/fde7575/ : SUCCESS in 5m 06s\n- openstack-tox-py36 http://logs.openstack.org/12/666812/4/check/openstack-tox-py36/41e0c48/ : SUCCESS in 6m 54s\n- openstack-tox-pep8 http://logs.openstack.org/12/666812/4/check/openstack-tox-pep8/8153cfb/ : SUCCESS in 6m 37s\n- openstack-tox-py27 http://logs.openstack.org/12/666812/4/check/openstack-tox-py27/4870613/ : SUCCESS in 6m 42s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/12/666812/4/check/kayobe-tox-ansible-syntax/6b25c52/ : SUCCESS in 7m 02s\n- kayobe-tox-ansible http://logs.openstack.org/12/666812/4/check/kayobe-tox-ansible/43663e3/ : SUCCESS in 8m 29s\n- kayobe-tox-molecule http://logs.openstack.org/12/666812/4/check/kayobe-tox-molecule/9e3fb66/ : SUCCESS in 54m 06s\n- kayobe-overcloud-centos http://logs.openstack.org/12/666812/4/check/kayobe-overcloud-centos/8c439ec/ : SUCCESS in 1h 01m 35s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/12/666812/4/check/kayobe-overcloud-upgrade-centos/0e91663/ : SUCCESS in 1h 54m 08s\n- kayobe-seed-centos http://logs.openstack.org/12/666812/4/check/kayobe-seed-centos/0d09a16/ : SUCCESS in 25m 12s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/12/666812/4/check/kayobe-seed-upgrade-centos/d56ab53/ : SUCCESS in 25m 59s","accounts_in_message":[],"_revision_number":4},{"id":"2c1ce93c462b71e5bcbeea57fb8b4192b2955d68","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-22 14:36:01.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"72de93235e6d6eda54c57389d355cd1e9408a533","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 16:43:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe177b48b97c4c61a132db9cae82b180 : SUCCESS in 5m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1e64627dbdef4cb7b8257d411027e571 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2771b188f384a378149f7dcdf9cc87d : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8a9072fdd0429c9c5ee51e92e70b84 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/997fdadc014e48d3a1d352407474c530 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88fad991730b4a70a676eda88fe8f5a8 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a170ab40b9e47a7955cfbc84d81b235 : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ce0d9bb8977449da692694301f1c3cd : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8483e0c10cf24ba7b542c6f35733f136 : SUCCESS in 49m 31s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/90130dd0b0464e6ba46bba913d67574a : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/06d44f804b274e30bb24a9ec322d1eb9 : SUCCESS in 2h 00m 19s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/7be3399dc49343c8b61d98a362802d3e : FAILURE in 11m 41s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c9305e6b0d984a5eb5a2985edc52da35 : SUCCESS in 34m 45s","accounts_in_message":[],"_revision_number":5},{"id":"b7028a9029e27e15987bf3cf52adfe16c7404669","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-08-23 11:06:22.000000000","message":"Patch Set 5:\n\nrecheck: https://review.opendev.org/#/c/678082 now landed","accounts_in_message":[],"_revision_number":5},{"id":"6d36760bd9ef4b37e688a2b9b8a601e0adfdd8ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 13:18:49.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e27b3ec904d4c8a8dfdf55ddab91d87 : SUCCESS in 4m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8fd7356651074d4f82daf6c53360fc05 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3799641af3e649cd93eaca42925d1e07 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7052c55a8b1f4bf4acf7647cfb3a0588 : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a62a286507664c889485d8bd29cb2d9d : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a6168b95cd140b8b1fc3e5ca8583ad6 : SUCCESS in 4m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75d06efa66044b558051e1cff600029e : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cfe59384a8fc433d9628d07f47010223 : SUCCESS in 10m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4cc6c165b279404b909421543338696e : SUCCESS in 46m 32s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d92252f937554a6e90080204769dd621 : TIMED_OUT in 2h 02m 44s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/aba460ce8a0e4ee19c7c41fff04bd6d9 : SUCCESS in 2h 04m 19s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1d8f583d3dff4745b515963104b84156 : SUCCESS in 27m 07s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/362831c12dc24efdbd556a978471e573 : SUCCESS in 31m 23s","accounts_in_message":[],"_revision_number":5},{"id":"6fbe82a53a86cd0e906b7e9bc5f8b228f1a0e02f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-08-23 13:51:18.000000000","message":"Patch Set 5:\n\ntenks nodes appear to be hanging after kernel exec (kernel and ramdisk are successfully downloaded by iPXE)","accounts_in_message":[],"_revision_number":5},{"id":"0cf5125cdceb601b22edcea9b88e648028a2d412","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-27 10:37:46.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"27b87091b6083de59bf6ec5cf8c688439adecea7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 12:37:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2993d4a5b20247c08d8eefbf20c79030 : SUCCESS in 5m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8448d6fe3e804867950da9838048cec1 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9478608d7df741cdb535f7cd029d7318 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/551e1456e8b84e7a9fecdb49b8085f4a : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/520a61b55507464b9ea0e66096e819ca : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35abd387c64b4cbc866cf6aad8d66433 : SUCCESS in 5m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d2c4b265fc94719898ba78e32f7c868 : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0beb9cc3372548ea84a1d0c960095f4b : SUCCESS in 8m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/789d14612932447391ff721458a29a75 : SUCCESS in 48m 08s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/2ef7348ad6e04adc997ec64e6a3a7e34 : SUCCESS in 1h 20m 08s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5608de03a6d04f2ebf662c3a37167e0b : SUCCESS in 1h 56m 39s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/7fb707dffe3249a0a0643a79762705ac : SUCCESS in 26m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3f0b7544f05f47b08410ea68316df4c7 : SUCCESS in 31m 21s","accounts_in_message":[],"_revision_number":5},{"id":"79dfb8d3648d6291f6c63aaa905d38355b3d79a8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-08-27 12:50:37.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9908158c5da90ff7430fd88afe886f5ad9a75057","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-30 09:43:43.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"406c6bb41c44b3cc6588ddc7a4a6864ec60faafd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 09:43:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ad0b905d8374ad82b8c4b77d6dfa6fb096b3cac5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 11:41:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"b80233051c35840efd9a0977c939ef7d6bbee185","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 11:41:19.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4d2d6cf13db74fcb8c2847206761c5ca : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fc644975cfb42c7ad977ddd8c63c893 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/329b45498aa9467c958a8868fa803cd9 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ddbdc1f68ce42ee8d111be6fbd2b867 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac770ce3d4d43a1a2bb68bdd24ab0f3 : SUCCESS in 5m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d09801b8c71549aa9fc3253b640cd177 : SUCCESS in 7m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/66c1796840f6422bb8cca1af1e1b2b55 : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0e16e37cb9bb496182feeffceaf48b7e : SUCCESS in 43m 41s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1151e366d1474c388f7e3ff8db68ff9d : SUCCESS in 1h 10m 11s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c40312c63f9745a5b91201eddf1fa42e : SUCCESS in 1h 51m 06s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/602e0a11979547648235fd216dddd637 : SUCCESS in 22m 56s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e2588029d09343a0880e3969389c3f1f : SUCCESS in 27m 16s","accounts_in_message":[],"_revision_number":5},{"id":"4cd6c8a78aeff9bdbfd2665f04d7382435ce3b89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 11:44:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0205a76c4e634cc7871e0792e28f5049 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"691ddeb614bdc85ed4f3594744c7bc7a732832f4","revisions":{"962b0ea0b96f5812b57500200c577b506f2de3df":{"kind":"REWORK","_number":1,"created":"2019-06-21 10:38:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/12/666812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/12/666812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/12/666812/1"}}},"commit":{"parents":[{"commit":"cb7476637866974cdd2c71af0838a31af85797cc","subject":"Update packages in virtualenvs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb7476637866974cdd2c71af0838a31af85797cc"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-21 10:37:45.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-21 10:38:24.000000000","tz":60},"subject":"Execute kayobe in verbose mode in CI","message":"Execute kayobe in verbose mode in CI\n\nChange-Id: I10c2b872887a47fab1a373fc0525dcd6fc78f7e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/962b0ea0b96f5812b57500200c577b506f2de3df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/962b0ea0b96f5812b57500200c577b506f2de3df"}]},"branch":"refs/heads/master"},"85f69cefefcb346bf60e76c1c0d965d46da9e662":{"kind":"REWORK","_number":2,"created":"2019-06-21 13:34:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/12/666812/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/12/666812/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/12/666812/2"}}},"commit":{"parents":[{"commit":"cb7476637866974cdd2c71af0838a31af85797cc","subject":"Update packages in virtualenvs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb7476637866974cdd2c71af0838a31af85797cc"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-21 10:37:45.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-21 13:33:45.000000000","tz":60},"subject":"Execute kayobe in verbose mode in CI","message":"Execute kayobe in verbose mode in CI\n\nAlso refactors the CI job playbooks to create the previous git repo\nclones for upgrade jobs in the kayobe-base job.\n\nChange-Id: I10c2b872887a47fab1a373fc0525dcd6fc78f7e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/85f69cefefcb346bf60e76c1c0d965d46da9e662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/85f69cefefcb346bf60e76c1c0d965d46da9e662"}]},"branch":"refs/heads/master"},"2ab72fb2b0946d2d0e8ddac735f463ae52112ab0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-21 14:10:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/12/666812/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/12/666812/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/12/666812/3"}}},"commit":{"parents":[{"commit":"d851e3c13a2c6316f12bac99bf960e01c8bc7cf5","subject":"Update packages in virtualenvs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d851e3c13a2c6316f12bac99bf960e01c8bc7cf5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-21 10:37:45.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-21 14:10:18.000000000","tz":60},"subject":"Execute kayobe in verbose mode in CI","message":"Execute kayobe in verbose mode in CI\n\nAlso refactors the CI job playbooks to create the previous git repo\nclones for upgrade jobs in the kayobe-base job.\n\nChange-Id: I10c2b872887a47fab1a373fc0525dcd6fc78f7e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ab72fb2b0946d2d0e8ddac735f463ae52112ab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ab72fb2b0946d2d0e8ddac735f463ae52112ab0"}]},"branch":"refs/heads/master"},"211dc1050b101b9fc910e39fc26f02c1f78a4d4a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-06-21 14:18:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/12/666812/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/12/666812/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/12/666812/4"}}},"commit":{"parents":[{"commit":"59c807d10968ebb2983d99e527d32dd3da7415d2","subject":"Update packages in virtualenvs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/59c807d10968ebb2983d99e527d32dd3da7415d2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-21 10:37:45.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-21 14:18:03.000000000","tz":60},"subject":"Execute kayobe in verbose mode in CI","message":"Execute kayobe in verbose mode in CI\n\nAlso refactors the CI job playbooks to create the previous git repo\nclones for upgrade jobs in the kayobe-base job.\n\nChange-Id: I10c2b872887a47fab1a373fc0525dcd6fc78f7e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211dc1050b101b9fc910e39fc26f02c1f78a4d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211dc1050b101b9fc910e39fc26f02c1f78a4d4a"}]},"branch":"refs/heads/master"},"691ddeb614bdc85ed4f3594744c7bc7a732832f4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-22 14:36:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/12/666812/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/12/666812/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/666812/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/12/666812/5"}}},"commit":{"parents":[{"commit":"45b4961de190224c91f66bcac874dc1dc5f183f4","subject":"Merge \"Update release documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45b4961de190224c91f66bcac874dc1dc5f183f4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-21 10:37:45.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-22 14:36:01.000000000","tz":0},"subject":"Execute kayobe in verbose mode in CI","message":"Execute kayobe in verbose mode in CI\n\nAlso refactors the CI job playbooks to create the previous git repo\nclones for upgrade jobs in the kayobe-base job.\n\nChange-Id: I10c2b872887a47fab1a373fc0525dcd6fc78f7e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/691ddeb614bdc85ed4f3594744c7bc7a732832f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/691ddeb614bdc85ed4f3594744c7bc7a732832f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
