)]}'
{"id":"openstack%2Fkayobe~667422","triplet_id":"openstack%2Fkayobe~master~Ice620746e3ab0e9aac05f085e8f43b432679f221","project":"openstack/kayobe","branch":"master","topic":"fix/ironic-agent","hashtags":[],"change_id":"Ice620746e3ab0e9aac05f085e8f43b432679f221","subject":"Fix custom ironic python agent environmental variables","status":"MERGED","created":"2019-06-25 17:48:19.000000000","updated":"2019-07-08 16:32:39.000000000","submitted":"2019-07-08 16:24:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667422-1562603094620-7de5f3e4","meta_rev_id":"5d4d62f83b6002871fc81651bb0606330f7cae5f","_number":667422,"virtual_id_number":667422,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-08 16:24:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-07-08 14:19:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-07-01 17:39:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-07-08 14:19:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-01 17:39:05.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-08 14:19:36.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-08 16:24:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23028b166941648894b386bfe61a73208002ee8f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-06-25 17:48:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d382606f297d042d9e06a1d158187dc535f5bb6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-25 19:14:40.000000000","message":"Patch Set 1: Code-Review+1\n\nNeeds a story.","accounts_in_message":[],"_revision_number":1},{"id":"fab894e565d8b1ad524e8f34a820d2a09dafd851","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 20:23:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/22/667422/1/check/openstack-tox-docs/6dc8baa/html/ : SUCCESS in 3m 21s\n- openstack-tox-cover http://logs.openstack.org/22/667422/1/check/openstack-tox-cover/c5cf158/cover/ : SUCCESS in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/22/667422/1/check/openstack-tox-py35/7668003/ : SUCCESS in 3m 46s\n- openstack-tox-py36 http://logs.openstack.org/22/667422/1/check/openstack-tox-py36/9183424/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/22/667422/1/check/openstack-tox-pep8/3835bbf/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/22/667422/1/check/openstack-tox-py27/9a0fd76/ : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/22/667422/1/check/kayobe-tox-ansible-syntax/99a3659/ : SUCCESS in 4m 00s\n- kayobe-tox-ansible http://logs.openstack.org/22/667422/1/check/kayobe-tox-ansible/42c4b69/ : SUCCESS in 8m 06s\n- kayobe-tox-molecule http://logs.openstack.org/22/667422/1/check/kayobe-tox-molecule/02749d7/ : SUCCESS in 51m 55s\n- kayobe-overcloud-centos http://logs.openstack.org/22/667422/1/check/kayobe-overcloud-centos/b654771/ : SUCCESS in 1h 14m 24s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/22/667422/1/check/kayobe-overcloud-upgrade-centos/0468b98/ : SUCCESS in 1h 42m 37s\n- kayobe-seed-centos http://logs.openstack.org/22/667422/1/check/kayobe-seed-centos/ec42782/ : SUCCESS in 20m 06s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/22/667422/1/check/kayobe-seed-upgrade-centos/df68b98/ : SUCCESS in 24m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6aa3007a6fdae916f7299d1a575f498b3f1677e1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-07-01 15:47:43.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"88e709040a91c884960a3c901f77ad0a743aef32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 17:36:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/22/667422/2/check/openstack-tox-docs/d93412c/html/ : SUCCESS in 3m 18s\n- openstack-tox-cover http://logs.openstack.org/22/667422/2/check/openstack-tox-cover/c0f9744/cover/ : SUCCESS in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/22/667422/2/check/openstack-tox-py35/8986d34/ : SUCCESS in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/22/667422/2/check/openstack-tox-py36/bf4b681/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/22/667422/2/check/openstack-tox-pep8/9d7ac36/ : SUCCESS in 3m 46s\n- openstack-tox-py27 http://logs.openstack.org/22/667422/2/check/openstack-tox-py27/6481c54/ : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/22/667422/2/check/kayobe-tox-ansible-syntax/53f2b6a/ : SUCCESS in 3m 47s\n- kayobe-tox-ansible http://logs.openstack.org/22/667422/2/check/kayobe-tox-ansible/0403eae/ : SUCCESS in 7m 29s\n- kayobe-tox-molecule http://logs.openstack.org/22/667422/2/check/kayobe-tox-molecule/3051900/ : SUCCESS in 54m 18s\n- kayobe-overcloud-centos http://logs.openstack.org/22/667422/2/check/kayobe-overcloud-centos/427996a/ : SUCCESS in 1h 10m 29s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/22/667422/2/check/kayobe-overcloud-upgrade-centos/fafdc6c/ : SUCCESS in 1h 40m 22s\n- kayobe-seed-centos http://logs.openstack.org/22/667422/2/check/kayobe-seed-centos/f489928/ : SUCCESS in 21m 15s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/22/667422/2/check/kayobe-seed-upgrade-centos/fb74161/ : SUCCESS in 26m 16s","accounts_in_message":[],"_revision_number":2},{"id":"dbdeab83a6d4fa85b424e69c6bde29462ec7b23d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-01 17:39:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f8cbad17df3730d54271d294f27237134d032a25","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-08 14:19:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"43d29497b8293b336b5cc6fcd005e166d997052e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 14:19:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"70e8c7b266a2be3dd5c9741c272115f67b5c1822","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:24:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6ddf330929ade54e48954ba5fdd0c1c4c6a3f0af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:24:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/22/667422/2/gate/openstack-tox-docs/de95758/html/ : SUCCESS in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/22/667422/2/gate/openstack-tox-py35/e8158d4/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/22/667422/2/gate/openstack-tox-py36/401189d/ : SUCCESS in 4m 08s\n- openstack-tox-pep8 http://logs.openstack.org/22/667422/2/gate/openstack-tox-pep8/4cb7f0c/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/22/667422/2/gate/openstack-tox-py27/6f2c5aa/ : SUCCESS in 4m 20s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/22/667422/2/gate/kayobe-tox-ansible-syntax/3df3e8f/ : SUCCESS in 4m 30s\n- kayobe-tox-ansible http://logs.openstack.org/22/667422/2/gate/kayobe-tox-ansible/a7683c2/ : SUCCESS in 8m 52s\n- kayobe-tox-molecule http://logs.openstack.org/22/667422/2/gate/kayobe-tox-molecule/986f2d3/ : SUCCESS in 52m 50s\n- kayobe-overcloud-centos http://logs.openstack.org/22/667422/2/gate/kayobe-overcloud-centos/122aec2/ : SUCCESS in 1h 16m 28s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/22/667422/2/gate/kayobe-overcloud-upgrade-centos/b371be1/ : SUCCESS in 1h 58m 58s\n- kayobe-seed-centos http://logs.openstack.org/22/667422/2/gate/kayobe-seed-centos/76cef50/ : SUCCESS in 21m 43s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/22/667422/2/gate/kayobe-seed-upgrade-centos/9517e3f/ : SUCCESS in 26m 59s","accounts_in_message":[],"_revision_number":2},{"id":"37354fb682e3d8f00d27fd50293d646882b01031","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 16:32:27.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 3db4c5c4acc6d9582c9376b1e837698029165a0a","accounts_in_message":[],"_revision_number":2},{"id":"867391763abde8657acebca0f13e9a0c20f43a43","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 16:32:39.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit dffb2dc94c745a6edee7d3d914899e20256a7ef3","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"96a946fee11b1b99fc191fd104948ba161774e97","revisions":{"c2e2e203761b1c33693f8bb3b337b9ba2d574ab3":{"kind":"REWORK","_number":1,"created":"2019-06-25 17:48:19.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/22/667422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/667422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/667422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/667422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/667422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/667422/1"}}},"commit":{"parents":[{"commit":"6baad19b0b96183c022006674f45f33f288eccad","subject":"Merge \"Restrict cliff to \u003c2.15.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6baad19b0b96183c022006674f45f33f288eccad"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-06-25 16:30:57.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-06-25 16:30:57.000000000","tz":60},"subject":"Fix custom ironic python agent environmental variables","message":"Fix custom ironic python agent environmental variables\n\nChange-Id: Ice620746e3ab0e9aac05f085e8f43b432679f221\nSee: https://github.com/openstack/diskimage-builder/blob/e796b3bc1884cbb0a7259be486d835ca114cca9e/diskimage_builder/elements/ironic-agent/environment.d/01-ironic-agent.sh#L5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2e2e203761b1c33693f8bb3b337b9ba2d574ab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2e2e203761b1c33693f8bb3b337b9ba2d574ab3"}]},"branch":"refs/heads/master"},"96a946fee11b1b99fc191fd104948ba161774e97":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-01 15:47:43.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/22/667422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/667422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/667422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/667422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/667422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/667422/2"}}},"commit":{"parents":[{"commit":"6baad19b0b96183c022006674f45f33f288eccad","subject":"Merge \"Restrict cliff to \u003c2.15.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6baad19b0b96183c022006674f45f33f288eccad"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-06-25 16:30:57.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-07-01 15:46:01.000000000","tz":0},"subject":"Fix custom ironic python agent environmental variables","message":"Fix custom ironic python agent environmental variables\n\nSee: https://github.com/openstack/diskimage-builder/blob/e796b3bc1884cbb0a7259be486d835ca114cca9e/diskimage_builder/elements/ironic-agent/environment.d/01-ironic-agent.sh#L5\n\nStory: 2006146\nTask: 35642\nChange-Id: Ice620746e3ab0e9aac05f085e8f43b432679f221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96a946fee11b1b99fc191fd104948ba161774e97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96a946fee11b1b99fc191fd104948ba161774e97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
