)]}'
{"id":"openstack%2Fkayobe~667632","triplet_id":"openstack%2Fkayobe~master~If0fe1115f01738254a61c0f5460548b0d6ae6b9a","project":"openstack/kayobe","branch":"master","topic":"story/2006050","hashtags":[],"change_id":"If0fe1115f01738254a61c0f5460548b0d6ae6b9a","subject":"Use new kolla-ansible upgrade-bifrost command","status":"MERGED","created":"2019-06-26 14:26:11.000000000","updated":"2019-08-07 17:48:19.000000000","submitted":"2019-08-07 17:48:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667632-1565200099573-3f7c88c0","meta_rev_id":"d3ed4a11325a1566d9fdb8664f801fbe4c9a3bcc","_number":667632,"virtual_id_number":667632,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-07 17:48:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-08-07 15:29:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-08-07 15:54:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-08-07 15:54:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-08-07 11:42:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 16:33:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-08-07 15:29:29.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-08-07 15:54:22.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-08-07 17:48:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38690eacf943a09f323db18d9553448e36c9de67","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-26 14:26:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea4d21f96ca97c7a8e41c743d333077364d6085b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 14:27:21.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"01d679fc44114ae887b36a75bbbd9487e95f4235","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-26 14:27:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"60ff362c33c5a587a1c46daf46dc113acd0d050b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 15:02:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/667632/1/check/openstack-tox-docs/fb12bbf/html/ : SUCCESS in 3m 22s\n- openstack-tox-cover http://logs.openstack.org/32/667632/1/check/openstack-tox-cover/91797b9/cover/ : SUCCESS in 3m 30s\n- openstack-tox-py35 http://logs.openstack.org/32/667632/1/check/openstack-tox-py35/251d7a2/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/32/667632/1/check/openstack-tox-py36/d6fce6a/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 http://logs.openstack.org/32/667632/1/check/openstack-tox-pep8/bdb0453/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/32/667632/1/check/openstack-tox-py27/8af09cd/ : SUCCESS in 3m 22s\n- kayobe-seed-upgrade-centos-1 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-1/273887b/ : SUCCESS in 24m 14s\n- kayobe-seed-upgrade-centos-2 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-2/cde2293/ : SUCCESS in 27m 07s\n- kayobe-seed-upgrade-centos-3 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-3/e478738/ : SUCCESS in 24m 03s\n- kayobe-seed-upgrade-centos-4 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-4/bd0f17a/ : SUCCESS in 24m 36s","accounts_in_message":[],"_revision_number":1},{"id":"26e643bf31c087f3d2ce110bff32b529c1815815","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-26 15:05:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"15a7ec2f1913a7cd350de4328c4ec4a9146ac39f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 15:59:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/667632/1/check/openstack-tox-docs/2d2f0b2/html/ : SUCCESS in 3m 20s\n- openstack-tox-cover http://logs.openstack.org/32/667632/1/check/openstack-tox-cover/155db87/cover/ : SUCCESS in 3m 45s\n- openstack-tox-py35 http://logs.openstack.org/32/667632/1/check/openstack-tox-py35/b941980/ : SUCCESS in 3m 53s\n- openstack-tox-py36 http://logs.openstack.org/32/667632/1/check/openstack-tox-py36/b0bbef8/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/32/667632/1/check/openstack-tox-pep8/b03df68/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/32/667632/1/check/openstack-tox-py27/afaff97/ : SUCCESS in 5m 05s\n- kayobe-seed-upgrade-centos-1 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-1/7bb587f/ : SUCCESS in 25m 50s\n- kayobe-seed-upgrade-centos-2 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-2/91ae79f/ : SUCCESS in 26m 58s\n- kayobe-seed-upgrade-centos-3 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-3/6a23575/ : SUCCESS in 23m 51s\n- kayobe-seed-upgrade-centos-4 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-4/128d71d/ : SUCCESS in 24m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f0b7ca638d14836436c0a212bfd46a3a4f500da1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-26 16:00:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1549e3a9f0781eb805dab34a476f6db3eadee334","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 16:37:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/667632/1/check/openstack-tox-docs/0987421/html/ : SUCCESS in 2m 59s\n- openstack-tox-cover http://logs.openstack.org/32/667632/1/check/openstack-tox-cover/1e12c8a/cover/ : SUCCESS in 3m 33s\n- openstack-tox-py35 http://logs.openstack.org/32/667632/1/check/openstack-tox-py35/d818329/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/32/667632/1/check/openstack-tox-py36/cc51546/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/32/667632/1/check/openstack-tox-pep8/5363359/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/32/667632/1/check/openstack-tox-py27/00974f5/ : SUCCESS in 3m 40s\n- kayobe-seed-upgrade-centos-1 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-1/17021fe/ : SUCCESS in 26m 52s\n- kayobe-seed-upgrade-centos-2 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-2/1283a2b/ : SUCCESS in 26m 25s\n- kayobe-seed-upgrade-centos-3 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-3/6ce84a4/ : SUCCESS in 25m 49s\n- kayobe-seed-upgrade-centos-4 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-4/b79ec7c/ : SUCCESS in 27m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5d9e2339093593b57a18ba89da9d1aa92049570b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-26 16:39:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f626899ab8b4a5face26089b61924707e57d7f89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 17:52:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/667632/1/check/openstack-tox-docs/be4ad84/html/ : SUCCESS in 3m 09s\n- openstack-tox-cover http://logs.openstack.org/32/667632/1/check/openstack-tox-cover/211418c/cover/ : SUCCESS in 4m 41s\n- openstack-tox-py35 http://logs.openstack.org/32/667632/1/check/openstack-tox-py35/a5b3327/ : SUCCESS in 3m 46s\n- openstack-tox-py36 http://logs.openstack.org/32/667632/1/check/openstack-tox-py36/38da09c/ : SUCCESS in 3m 14s\n- openstack-tox-pep8 http://logs.openstack.org/32/667632/1/check/openstack-tox-pep8/681d779/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/32/667632/1/check/openstack-tox-py27/980fe70/ : SUCCESS in 3m 33s\n- kayobe-seed-upgrade-centos-1 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-1/310810d/ : SUCCESS in 26m 50s\n- kayobe-seed-upgrade-centos-2 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-2/914f3d8/ : SUCCESS in 26m 50s\n- kayobe-seed-upgrade-centos-3 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-3/b8d6640/ : SUCCESS in 24m 38s\n- kayobe-seed-upgrade-centos-4 http://logs.openstack.org/32/667632/1/check/kayobe-seed-upgrade-centos-4/0d5001f/ : SUCCESS in 27m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c359273bb8fc3f98a8305f43d07cecb43bb0f973","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-26 18:12:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e6bc71fcc1f0415e0a6f527ae8bd9cf21a277b0e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-26 18:12:37.000000000","message":"Patch Set 2:\n\nRan the kayobe-seed-upgrade-centos job 16 times with this change and no failures occurred.","accounts_in_message":[],"_revision_number":2},{"id":"f0778d92c4794a5df86e294aa4bf4cdd6f9e0284","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 20:01:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/667632/2/check/openstack-tox-docs/c50ee90/html/ : SUCCESS in 2m 55s\n- openstack-tox-cover http://logs.openstack.org/32/667632/2/check/openstack-tox-cover/8ce4643/cover/ : SUCCESS in 3m 47s\n- openstack-tox-py35 http://logs.openstack.org/32/667632/2/check/openstack-tox-py35/a90bedf/ : SUCCESS in 3m 52s\n- openstack-tox-py36 http://logs.openstack.org/32/667632/2/check/openstack-tox-py36/54b1653/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/32/667632/2/check/openstack-tox-pep8/6270eb3/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/32/667632/2/check/openstack-tox-py27/8784555/ : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/32/667632/2/check/kayobe-tox-ansible-syntax/52bcdd7/ : SUCCESS in 4m 11s\n- kayobe-tox-ansible http://logs.openstack.org/32/667632/2/check/kayobe-tox-ansible/193df4c/ : SUCCESS in 7m 46s\n- kayobe-tox-molecule http://logs.openstack.org/32/667632/2/check/kayobe-tox-molecule/6527216/ : SUCCESS in 50m 26s\n- kayobe-overcloud-centos http://logs.openstack.org/32/667632/2/check/kayobe-overcloud-centos/26c8172/ : SUCCESS in 1h 02m 35s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/32/667632/2/check/kayobe-overcloud-upgrade-centos/cbfa11e/ : SUCCESS in 1h 40m 35s\n- kayobe-seed-centos http://logs.openstack.org/32/667632/2/check/kayobe-seed-centos/e6cf4f2/ : SUCCESS in 20m 28s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/32/667632/2/check/kayobe-seed-upgrade-centos/9f378e0/ : SUCCESS in 26m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f68477702e3eee639289637b2bbbf693794ae2e6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-03 18:25:25.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"7bd65f3bed6b2be4819c9131e3341ed5839b438b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 19:19:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/32/667632/3/check/openstack-tox-docs/474b1b4/html/ : SUCCESS in 3m 39s\n- openstack-tox-cover http://logs.openstack.org/32/667632/3/check/openstack-tox-cover/be6f685/cover/ : SUCCESS in 4m 38s\n- openstack-tox-py35 http://logs.openstack.org/32/667632/3/check/openstack-tox-py35/339d084/ : SUCCESS in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/32/667632/3/check/openstack-tox-py36/0d10c7a/ : SUCCESS in 5m 07s\n- openstack-tox-pep8 http://logs.openstack.org/32/667632/3/check/openstack-tox-pep8/aa544a7/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/32/667632/3/check/openstack-tox-py27/2a4f945/ : SUCCESS in 3m 36s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/32/667632/3/check/kayobe-tox-ansible-syntax/8165501/ : SUCCESS in 4m 00s\n- kayobe-tox-ansible http://logs.openstack.org/32/667632/3/check/kayobe-tox-ansible/cedab01/ : SUCCESS in 8m 11s\n- kayobe-tox-molecule http://logs.openstack.org/32/667632/3/check/kayobe-tox-molecule/3927ac3/ : SUCCESS in 50m 40s\n- kayobe-overcloud-centos http://logs.openstack.org/32/667632/3/check/kayobe-overcloud-centos/883d7d5/ : FAILURE in 8m 54s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/32/667632/3/check/kayobe-overcloud-upgrade-centos/29ab9a6/ : FAILURE in 8m 12s\n- kayobe-seed-centos http://logs.openstack.org/32/667632/3/check/kayobe-seed-centos/50fef22/ : FAILURE in 8m 06s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/32/667632/3/check/kayobe-seed-upgrade-centos/14c01ec/ : SUCCESS in 23m 54s","accounts_in_message":[],"_revision_number":3},{"id":"c434c56358ceb3c90405f53abf568b31dd83a44e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-04 08:52:38.000000000","message":"Patch Set 3:\n\nrecheck\n\nmirror issues","accounts_in_message":[],"_revision_number":3},{"id":"8e60cec182ecb38042b59b7e004cde2132375d2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 08:52:50.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"130a6242dfa8ca86ca70a4a0919d1ab170dfe29e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 15:03:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5043a3f98f715ea01e9de56bf97337e181de8c94","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 15:03:52.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"20b2dfdef388e9e02cd5f280b072787b7091f8c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 17:11:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/32/667632/4/check/openstack-tox-docs/ffb6082/html/ : SUCCESS in 2m 34s\n- openstack-tox-cover http://logs.openstack.org/32/667632/4/check/openstack-tox-cover/7baf000/cover/ : SUCCESS in 3m 46s\n- openstack-tox-py35 http://logs.openstack.org/32/667632/4/check/openstack-tox-py35/1df3b3a/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/32/667632/4/check/openstack-tox-py36/f91f30d/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/32/667632/4/check/openstack-tox-pep8/8783c84/ : SUCCESS in 3m 49s\n- openstack-tox-py27 http://logs.openstack.org/32/667632/4/check/openstack-tox-py27/201af6d/ : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/32/667632/4/check/kayobe-tox-ansible-syntax/64894fa/ : SUCCESS in 3m 46s\n- kayobe-tox-ansible http://logs.openstack.org/32/667632/4/check/kayobe-tox-ansible/f588c57/ : SUCCESS in 11m 06s\n- kayobe-tox-molecule http://logs.openstack.org/32/667632/4/check/kayobe-tox-molecule/b1cbd68/ : SUCCESS in 52m 26s\n- kayobe-overcloud-centos http://logs.openstack.org/32/667632/4/check/kayobe-overcloud-centos/94cb5a4/ : FAILURE in 40m 39s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/32/667632/4/check/kayobe-overcloud-upgrade-centos/bb88ae5/ : SUCCESS in 1h 58m 00s\n- kayobe-seed-centos http://logs.openstack.org/32/667632/4/check/kayobe-seed-centos/d964598/ : SUCCESS in 19m 30s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/32/667632/4/check/kayobe-seed-upgrade-centos/50f64b0/ : SUCCESS in 28m 13s","accounts_in_message":[],"_revision_number":4},{"id":"a14ec1403ba30979b1170f8668ae0781570de343","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-10 13:06:14.000000000","message":"Patch Set 4:\n\nrecheck\n\nkolla issue fixed","accounts_in_message":[],"_revision_number":4},{"id":"79e61865148c4b7ae9bb6c7015dd0674da815c38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 15:20:06.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/32/667632/4/check/openstack-tox-docs/a1596fe/html/ : SUCCESS in 3m 09s\n- openstack-tox-cover http://logs.openstack.org/32/667632/4/check/openstack-tox-cover/5a41817/cover/ : SUCCESS in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/32/667632/4/check/openstack-tox-py35/65f4d5d/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/32/667632/4/check/openstack-tox-py36/29a8463/ : SUCCESS in 4m 46s\n- openstack-tox-pep8 http://logs.openstack.org/32/667632/4/check/openstack-tox-pep8/81cb4f0/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/32/667632/4/check/openstack-tox-py27/9adab50/ : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/32/667632/4/check/kayobe-tox-ansible-syntax/8d4b768/ : SUCCESS in 4m 36s\n- kayobe-tox-ansible http://logs.openstack.org/32/667632/4/check/kayobe-tox-ansible/cfa5396/ : SUCCESS in 8m 24s\n- kayobe-tox-molecule http://logs.openstack.org/32/667632/4/check/kayobe-tox-molecule/94a3dc5/ : SUCCESS in 53m 03s\n- kayobe-overcloud-centos http://logs.openstack.org/32/667632/4/check/kayobe-overcloud-centos/722754d/ : FAILURE in 46m 26s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/32/667632/4/check/kayobe-overcloud-upgrade-centos/c8f4365/ : SUCCESS in 2h 12m 42s\n- kayobe-seed-centos http://logs.openstack.org/32/667632/4/check/kayobe-seed-centos/1e54168/ : SUCCESS in 22m 50s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/32/667632/4/check/kayobe-seed-upgrade-centos/9122598/ : SUCCESS in 25m 45s","accounts_in_message":[],"_revision_number":4},{"id":"4970af63c5ff9d6247b8afe37c6d3ef39e159c56","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-11 16:40:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c4fbcd080a0941bb8361ba6ea7fbafb655911f41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 18:28:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/667632/4/check/openstack-tox-docs/048428d/html/ : SUCCESS in 3m 01s\n- openstack-tox-cover http://logs.openstack.org/32/667632/4/check/openstack-tox-cover/29bc6d6/cover/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/32/667632/4/check/openstack-tox-py35/7d4397e/ : SUCCESS in 3m 53s\n- openstack-tox-py36 http://logs.openstack.org/32/667632/4/check/openstack-tox-py36/b7ab4fe/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/32/667632/4/check/openstack-tox-pep8/66e73d0/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/32/667632/4/check/openstack-tox-py27/bcc0a96/ : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/32/667632/4/check/kayobe-tox-ansible-syntax/11c9103/ : SUCCESS in 4m 41s\n- kayobe-tox-ansible http://logs.openstack.org/32/667632/4/check/kayobe-tox-ansible/4cd7594/ : SUCCESS in 8m 32s\n- kayobe-tox-molecule http://logs.openstack.org/32/667632/4/check/kayobe-tox-molecule/a015947/ : SUCCESS in 53m 08s\n- kayobe-overcloud-centos http://logs.openstack.org/32/667632/4/check/kayobe-overcloud-centos/ae3ae97/ : SUCCESS in 1h 01m 40s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/32/667632/4/check/kayobe-overcloud-upgrade-centos/8deeaaf/ : SUCCESS in 1h 38m 55s\n- kayobe-seed-centos http://logs.openstack.org/32/667632/4/check/kayobe-seed-centos/6c9af8b/ : SUCCESS in 20m 04s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/32/667632/4/check/kayobe-seed-upgrade-centos/36bfea3/ : SUCCESS in 24m 24s","accounts_in_message":[],"_revision_number":4},{"id":"78259da9f039803c242e6c10b2529bc37dd9de8c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-12 16:33:44.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7435ca57cd2faf11466df8478a1904e8e3aba707","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-02 11:10:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2ffb0a58200c9fec291c1cca864fb722acbaa28a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 12:51:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/32/667632/5/check/openstack-tox-docs/feef36b/html/ : SUCCESS in 6m 08s\n- openstack-tox-cover https://logs.opendev.org/32/667632/5/check/openstack-tox-cover/704a782/cover/ : SUCCESS in 4m 47s\n- openstack-tox-py35 https://logs.opendev.org/32/667632/5/check/openstack-tox-py35/719624c/ : SUCCESS in 5m 00s\n- openstack-tox-py36 https://logs.opendev.org/32/667632/5/check/openstack-tox-py36/8ab9f52/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://logs.opendev.org/32/667632/5/check/openstack-tox-pep8/0b100e0/ : SUCCESS in 6m 05s\n- openstack-tox-py27 https://logs.opendev.org/32/667632/5/check/openstack-tox-py27/70c86a5/ : SUCCESS in 4m 23s\n- kayobe-tox-ansible-syntax https://logs.opendev.org/32/667632/5/check/kayobe-tox-ansible-syntax/a723219/ : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://logs.opendev.org/32/667632/5/check/kayobe-tox-ansible/e733c08/ : SUCCESS in 8m 56s\n- kayobe-tox-molecule https://logs.opendev.org/32/667632/5/check/kayobe-tox-molecule/c1d5afc/ : SUCCESS in 44m 20s\n- kayobe-overcloud-centos https://logs.opendev.org/32/667632/5/check/kayobe-overcloud-centos/6e925bc/ : SUCCESS in 1h 10m 02s\n- kayobe-overcloud-upgrade-centos https://logs.opendev.org/32/667632/5/check/kayobe-overcloud-upgrade-centos/77083b3/ : SUCCESS in 1h 38m 55s\n- kayobe-seed-centos https://logs.opendev.org/32/667632/5/check/kayobe-seed-centos/bb43dfa/ : SUCCESS in 22m 47s\n- kayobe-seed-upgrade-centos https://logs.opendev.org/32/667632/5/check/kayobe-seed-upgrade-centos/d006f11/ : SUCCESS in 28m 52s","accounts_in_message":[],"_revision_number":5},{"id":"589cb30e95c168d941447e90dc923f406f22670e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-07 11:42:52.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"c612e699b378352b33b5d2332d79e810ddeab686","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-08-07 15:29:29.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"78f548381d3d764c100247bc5ba6edd0b3dcdbf9","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-07 15:54:22.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5c15d92bc5caa207a80b9fba3236e90ce9ab8f16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 15:54:34.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8c115caace64fb622c6b1e1d4604d50935b88662","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 17:48:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"47824b1e6eb7f660f4f8bf3f001b92fb5ba5d11a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 17:48:19.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/32/667632/5/gate/openstack-tox-docs/199d8e9/html/ : SUCCESS in 3m 45s\n- openstack-tox-py35 https://logs.opendev.org/32/667632/5/gate/openstack-tox-py35/8cc7363/ : SUCCESS in 5m 14s\n- openstack-tox-py36 https://logs.opendev.org/32/667632/5/gate/openstack-tox-py36/d75e5a9/ : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://logs.opendev.org/32/667632/5/gate/openstack-tox-pep8/4ef0db1/ : SUCCESS in 5m 38s\n- openstack-tox-py27 https://logs.opendev.org/32/667632/5/gate/openstack-tox-py27/825372d/ : SUCCESS in 5m 47s\n- kayobe-tox-ansible-syntax https://logs.opendev.org/32/667632/5/gate/kayobe-tox-ansible-syntax/4179916/ : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://logs.opendev.org/32/667632/5/gate/kayobe-tox-ansible/40d6c70/ : SUCCESS in 8m 42s\n- kayobe-tox-molecule https://logs.opendev.org/32/667632/5/gate/kayobe-tox-molecule/f8d4eca/ : SUCCESS in 46m 41s\n- kayobe-overcloud-centos https://logs.opendev.org/32/667632/5/gate/kayobe-overcloud-centos/a582e2b/ : SUCCESS in 1h 09m 52s\n- kayobe-overcloud-upgrade-centos https://logs.opendev.org/32/667632/5/gate/kayobe-overcloud-upgrade-centos/c4acaaa/ : SUCCESS in 1h 48m 21s\n- kayobe-seed-centos https://logs.opendev.org/32/667632/5/gate/kayobe-seed-centos/075031f/ : SUCCESS in 20m 36s\n- kayobe-seed-upgrade-centos https://logs.opendev.org/32/667632/5/gate/kayobe-seed-upgrade-centos/463a452/ : SUCCESS in 27m 36s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"cd97a9d721bc8d9536f7b24bce09fa07487b25a8","revisions":{"d30d0d5ec4b725324c7f63922f1aa799e8ed71f5":{"kind":"REWORK","_number":1,"created":"2019-06-26 14:26:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/667632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/667632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/667632/1"}}},"commit":{"parents":[{"commit":"ab1e5321a96fd119d46c3c70c013ce53fdca8de4","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab1e5321a96fd119d46c3c70c013ce53fdca8de4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-26 14:22:11.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-26 14:26:04.000000000","tz":60},"subject":"Use new kolla-ansible upgrade-bifrost command","message":"Use new kolla-ansible upgrade-bifrost command\n\nThis command slightly modifies the deploy operation to first stop the\nbifrost services and container if they are running.\n\nChange-Id: If0fe1115f01738254a61c0f5460548b0d6ae6b9a\nDepends-On: https://review.opendev.org/667630\nStory: 2006050\nTask: 34743\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d30d0d5ec4b725324c7f63922f1aa799e8ed71f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d30d0d5ec4b725324c7f63922f1aa799e8ed71f5"}]},"branch":"refs/heads/master"},"db90664b09d6a0e1f09f0a0670f53a4d2b1b272d":{"kind":"REWORK","_number":2,"created":"2019-06-26 18:12:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/667632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/667632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/667632/2"}}},"commit":{"parents":[{"commit":"ab1e5321a96fd119d46c3c70c013ce53fdca8de4","subject":"Merge \"Fix overcloud introspection data save\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab1e5321a96fd119d46c3c70c013ce53fdca8de4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-26 14:22:11.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-26 18:11:58.000000000","tz":0},"subject":"Use new kolla-ansible upgrade-bifrost command","message":"Use new kolla-ansible upgrade-bifrost command\n\nThis command slightly modifies the deploy operation to first stop the\nbifrost services and container if they are running.\n\nChange-Id: If0fe1115f01738254a61c0f5460548b0d6ae6b9a\nDepends-On: https://review.opendev.org/667630\nStory: 2006050\nTask: 34743\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db90664b09d6a0e1f09f0a0670f53a4d2b1b272d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db90664b09d6a0e1f09f0a0670f53a4d2b1b272d"}]},"branch":"refs/heads/master"},"bf80a68216d7c3218d0fbdedfaee30d03963897d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-03 18:25:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/667632/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/667632/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/667632/3"}}},"commit":{"parents":[{"commit":"7115972232b75b776f7ece2f66d48bb00e17f719","subject":"Merge \"Add config tags to kolla-openstack play\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7115972232b75b776f7ece2f66d48bb00e17f719"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-26 14:22:11.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-03 18:24:39.000000000","tz":60},"subject":"Use new kolla-ansible upgrade-bifrost command","message":"Use new kolla-ansible upgrade-bifrost command\n\nThis command slightly modifies the deploy operation to first stop the\nbifrost services and container if they are running.\n\nChange-Id: If0fe1115f01738254a61c0f5460548b0d6ae6b9a\nDepends-On: https://review.opendev.org/667630\nStory: 2006050\nTask: 34743\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf80a68216d7c3218d0fbdedfaee30d03963897d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf80a68216d7c3218d0fbdedfaee30d03963897d"}]},"branch":"refs/heads/master"},"fe03e8aac208cb38ab439b9d83dc9cd4ee109b69":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-08 15:03:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/667632/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/667632/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/667632/4"}}},"commit":{"parents":[{"commit":"b1f54eabe13988b7091f85ebcd98d304d314798e","subject":"Merge \"Add retries to ansible galaxy install for all envs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1f54eabe13988b7091f85ebcd98d304d314798e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-26 14:22:11.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-08 15:03:52.000000000","tz":0},"subject":"Use new kolla-ansible upgrade-bifrost command","message":"Use new kolla-ansible upgrade-bifrost command\n\nThis command slightly modifies the deploy operation to first stop the\nbifrost services and container if they are running.\n\nChange-Id: If0fe1115f01738254a61c0f5460548b0d6ae6b9a\nDepends-On: https://review.opendev.org/667630\nStory: 2006050\nTask: 34743\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fe03e8aac208cb38ab439b9d83dc9cd4ee109b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fe03e8aac208cb38ab439b9d83dc9cd4ee109b69"}]},"branch":"refs/heads/master"},"cd97a9d721bc8d9536f7b24bce09fa07487b25a8":{"kind":"REWORK","_number":5,"created":"2019-08-02 11:10:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/667632/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/667632/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/667632/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/667632/5"}}},"commit":{"parents":[{"commit":"d1b51adb3d9638a393277b7100406a52d421fbfe","subject":"Merge \"Ignore --limit for localhost plays\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d1b51adb3d9638a393277b7100406a52d421fbfe"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-26 14:22:11.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-02 11:10:16.000000000","tz":60},"subject":"Use new kolla-ansible upgrade-bifrost command","message":"Use new kolla-ansible upgrade-bifrost command\n\nThis command slightly modifies the deploy operation to first stop the\nbifrost services and container if they are running.\n\nChange-Id: If0fe1115f01738254a61c0f5460548b0d6ae6b9a\nDepends-On: https://review.opendev.org/667630\nStory: 2006050\nTask: 34743\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd97a9d721bc8d9536f7b24bce09fa07487b25a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd97a9d721bc8d9536f7b24bce09fa07487b25a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
