)]}'
{"id":"openstack%2Fkayobe~667669","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I2ad53aa0db3bab8db6d6d9de4b00c5dac756fb15","project":"openstack/kayobe","branch":"stable/rocky","topic":"fix-introspection-data-save-stable/rocky","hashtags":[],"change_id":"I2ad53aa0db3bab8db6d6d9de4b00c5dac756fb15","subject":"Fix overcloud introspection data save","status":"MERGED","created":"2019-06-26 16:09:53.000000000","updated":"2019-07-17 18:47:20.000000000","submitted":"2019-07-17 18:47:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667669-1563389240591-a91269ad","meta_rev_id":"a31761d48628d8e2d2188771101601f94a3d4f1d","_number":667669,"virtual_id_number":667669,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-07-17 18:47:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-07-17 16:51:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-12 16:38:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-07-17 16:51:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 16:38:19.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-07-17 16:51:12.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-17 18:47:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec0f182974ecafe8fa2d9aa1435362c5922f6078","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-26 16:09:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5290a66ab9de1ff0bc7e66452369d105320b8741","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 18:02:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/69/667669/1/check/build-openstack-sphinx-docs/bc0ff6e/html/ : SUCCESS in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/69/667669/1/check/openstack-tox-docs/61e071e/html/ : SUCCESS in 2m 55s\n- openstack-tox-cover http://logs.openstack.org/69/667669/1/check/openstack-tox-cover/315291e/cover/ : SUCCESS in 5m 01s\n- openstack-tox-py35 http://logs.openstack.org/69/667669/1/check/openstack-tox-py35/a379863/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/69/667669/1/check/openstack-tox-pep8/92a9e63/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/69/667669/1/check/openstack-tox-py27/2e5866f/ : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/69/667669/1/check/kayobe-tox-ansible-syntax/0395bee/ : SUCCESS in 4m 04s\n- kayobe-tox-ansible http://logs.openstack.org/69/667669/1/check/kayobe-tox-ansible/910af5b/ : SUCCESS in 7m 07s\n- kayobe-tox-molecule http://logs.openstack.org/69/667669/1/check/kayobe-tox-molecule/eb2c896/ : SUCCESS in 49m 05s\n- kayobe-overcloud-centos http://logs.openstack.org/69/667669/1/check/kayobe-overcloud-centos/4d5429b/ : SUCCESS in 57m 33s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/69/667669/1/check/kayobe-overcloud-upgrade-centos/5df8095/ : SUCCESS in 1h 41m 35s\n- kayobe-seed-centos http://logs.openstack.org/69/667669/1/check/kayobe-seed-centos/f1ca9f5/ : SUCCESS in 16m 21s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/69/667669/1/check/kayobe-seed-upgrade-centos/8a31467/ : SUCCESS in 28m 07s","accounts_in_message":[],"_revision_number":1},{"id":"26b89d1e472cc877fdea4bed13a9c19db3f59650","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-01 09:40:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e554cc0b3ce2f57ccf5b6ef6149ede53a26055bf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-01 18:32:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ece99b1a93948662a11ca2f226f4a7d8c8ca9d4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 20:34:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/69/667669/2/check/build-openstack-sphinx-docs/e9d20f2/html/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/69/667669/2/check/openstack-tox-docs/fc36bd5/html/ : SUCCESS in 4m 08s\n- openstack-tox-cover http://logs.openstack.org/69/667669/2/check/openstack-tox-cover/b2f7d09/cover/ : SUCCESS in 4m 45s\n- openstack-tox-py35 http://logs.openstack.org/69/667669/2/check/openstack-tox-py35/b0af8d1/ : SUCCESS in 5m 41s\n- openstack-tox-pep8 http://logs.openstack.org/69/667669/2/check/openstack-tox-pep8/c39d41e/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/69/667669/2/check/openstack-tox-py27/f665648/ : SUCCESS in 5m 06s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/69/667669/2/check/kayobe-tox-ansible-syntax/59c735a/ : SUCCESS in 3m 56s\n- kayobe-tox-ansible http://logs.openstack.org/69/667669/2/check/kayobe-tox-ansible/b9a13c4/ : SUCCESS in 11m 51s\n- kayobe-tox-molecule http://logs.openstack.org/69/667669/2/check/kayobe-tox-molecule/13dbf70/ : SUCCESS in 53m 27s\n- kayobe-overcloud-centos http://logs.openstack.org/69/667669/2/check/kayobe-overcloud-centos/1608ef7/ : SUCCESS in 1h 12m 41s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/69/667669/2/check/kayobe-overcloud-upgrade-centos/1ef5bd2/ : SUCCESS in 1h 51m 08s\n- kayobe-seed-centos http://logs.openstack.org/69/667669/2/check/kayobe-seed-centos/7e03b6d/ : SUCCESS in 17m 32s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/69/667669/2/check/kayobe-seed-upgrade-centos/05a4ca7/ : SUCCESS in 27m 08s","accounts_in_message":[],"_revision_number":2},{"id":"81230871d76819a2445b65a387954a74eaa07aaf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-12 16:38:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"03db7ffc16aed15096c340bf7c594e05bf0aa327","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-17 16:51:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c50441b26a27466d416b07a0f80cbb2d47b03144","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 16:51:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4a2675c1c547f4897b1e294862feae4c11cae9c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:47:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"866a62ce578fb38b4e28760cb0d896260797dfd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:47:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/69/667669/2/gate/build-openstack-sphinx-docs/270837a/html/ : SUCCESS in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/69/667669/2/gate/openstack-tox-docs/793db60/html/ : SUCCESS in 3m 58s\n- openstack-tox-py35 http://logs.openstack.org/69/667669/2/gate/openstack-tox-py35/f57a6f4/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/69/667669/2/gate/openstack-tox-pep8/f165582/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/69/667669/2/gate/openstack-tox-py27/5222a78/ : SUCCESS in 9m 12s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/69/667669/2/gate/kayobe-tox-ansible-syntax/1da4055/ : SUCCESS in 5m 41s\n- kayobe-tox-ansible http://logs.openstack.org/69/667669/2/gate/kayobe-tox-ansible/6b2ac5d/ : SUCCESS in 10m 07s\n- kayobe-tox-molecule http://logs.openstack.org/69/667669/2/gate/kayobe-tox-molecule/486a0b9/ : SUCCESS in 31m 24s\n- kayobe-overcloud-centos http://logs.openstack.org/69/667669/2/gate/kayobe-overcloud-centos/4fcd963/ : SUCCESS in 1h 39m 42s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/69/667669/2/gate/kayobe-overcloud-upgrade-centos/b14a176/ : SUCCESS in 1h 47m 33s\n- kayobe-seed-centos http://logs.openstack.org/69/667669/2/gate/kayobe-seed-centos/8cb62f7/ : SUCCESS in 16m 03s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/69/667669/2/gate/kayobe-seed-upgrade-centos/212ea71/ : SUCCESS in 25m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"191f35004bc162d0fdc16acf9727d8c797ce6e99","revisions":{"191c71ade540c3f62462b55c0661242d6d48e8cd":{"kind":"REWORK","_number":1,"created":"2019-06-26 16:09:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/69/667669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/667669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/667669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/667669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/667669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/667669/1"}}},"commit":{"parents":[{"commit":"f99aee461ff4055abbbd25bfee1305110a3c0812","subject":"Restrict cliff to \u003c2.15.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f99aee461ff4055abbbd25bfee1305110a3c0812"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-17 14:00:53.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-26 16:09:53.000000000","tz":0},"subject":"Fix overcloud introspection data save","message":"Fix overcloud introspection data save\n\nIf running \u0027kayobe overcloud introspection data save\u0027 in an environment\nwhere nodes have been registered in the seed\u0027s ironic service manually\n(rather than through auto-discovery), previously the bifrost ansible\ninventory would not have been populated.\n\nThis change ensures the bifrost ansible inventory exists prior to\ncollecting introspection data.\n\nChange-Id: I2ad53aa0db3bab8db6d6d9de4b00c5dac756fb15\n(cherry picked from commit c7ab69b38b8eebda2620f53366e4a9156d431f3a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/191c71ade540c3f62462b55c0661242d6d48e8cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/191c71ade540c3f62462b55c0661242d6d48e8cd"}]},"branch":"refs/heads/stable/rocky"},"191f35004bc162d0fdc16acf9727d8c797ce6e99":{"kind":"REWORK","_number":2,"created":"2019-07-01 18:32:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/69/667669/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/69/667669/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/667669/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/667669/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/69/667669/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/69/667669/2"}}},"commit":{"parents":[{"commit":"f99aee461ff4055abbbd25bfee1305110a3c0812","subject":"Restrict cliff to \u003c2.15.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f99aee461ff4055abbbd25bfee1305110a3c0812"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-17 14:00:53.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-01 18:25:26.000000000","tz":0},"subject":"Fix overcloud introspection data save","message":"Fix overcloud introspection data save\n\nIf running \u0027kayobe overcloud introspection data save\u0027 in an environment\nwhere nodes have been registered in the seed\u0027s ironic service manually\n(rather than through auto-discovery), previously the bifrost ansible\ninventory would not have been populated.\n\nThis change ensures the bifrost ansible inventory exists prior to\ncollecting introspection data.\n\nChange-Id: I2ad53aa0db3bab8db6d6d9de4b00c5dac756fb15\n(cherry picked from commit c7ab69b38b8eebda2620f53366e4a9156d431f3a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/191f35004bc162d0fdc16acf9727d8c797ce6e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/191f35004bc162d0fdc16acf9727d8c797ce6e99"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
