)]}'
{"id":"openstack%2Fkayobe~667672","triplet_id":"openstack%2Fkayobe~stable%2Frocky~Iaa304221b8093ac71f9cdbb23edc84d1517578da","project":"openstack/kayobe","branch":"stable/rocky","hashtags":[],"change_id":"Iaa304221b8093ac71f9cdbb23edc84d1517578da","subject":"Multiple updates for Kayobe","status":"MERGED","created":"2019-06-26 16:17:36.000000000","updated":"2019-07-31 18:17:40.000000000","submitted":"2019-07-31 18:17:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667672-1564597060113-0139e600","meta_rev_id":"fdacacc90fe2d4f18e5c1f3b20ffbef1f7458092","_number":667672,"virtual_id_number":667672,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-07-31 18:17:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":20301,"name":"Ross Martyn","email":"ross.martyn@gresearch.co.uk","username":"rossmartyn04"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-07-17 16:58:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-01 21:25:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2019-07-03 09:42:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20301,"name":"Ross Martyn","email":"ross.martyn@gresearch.co.uk","username":"rossmartyn04"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-07-31 16:33:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":20301,"name":"Ross Martyn","email":"ross.martyn@gresearch.co.uk","username":"rossmartyn04"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":20301,"name":"Ross Martyn","email":"ross.martyn@gresearch.co.uk","username":"rossmartyn04"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":20301,"name":"Ross Martyn","email":"ross.martyn@gresearch.co.uk","username":"rossmartyn04"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":20301,"name":"Ross Martyn","email":"ross.martyn@gresearch.co.uk","username":"rossmartyn04"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-01 21:25:54.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-07-03 09:42:20.000000000","updated_by":{"_account_id":20301,"name":"Ross Martyn","email":"ross.martyn@gresearch.co.uk","username":"rossmartyn04"},"reviewer":{"_account_id":20301,"name":"Ross Martyn","email":"ross.martyn@gresearch.co.uk","username":"rossmartyn04"},"state":"REVIEWER"},{"updated":"2019-07-31 16:33:40.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-31 18:17:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87dfc271a5afe592b0785d215e2fe62ccb8de064","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-06-26 16:17:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4cfd54a3ebc1b9f00537b2c8eae33c33980c82d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 19:04:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/667672/1/check/build-openstack-sphinx-docs/b2bf8cc/html/ : SUCCESS in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/72/667672/1/check/openstack-tox-docs/c3a95af/html/ : SUCCESS in 2m 48s\n- openstack-tox-cover http://logs.openstack.org/72/667672/1/check/openstack-tox-cover/79b9c6b/cover/ : SUCCESS in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/72/667672/1/check/openstack-tox-py35/f6f92e2/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/72/667672/1/check/openstack-tox-pep8/7cfd582/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/72/667672/1/check/openstack-tox-py27/fe296a4/ : SUCCESS in 3m 41s\n- build-openstack-releasenotes http://logs.openstack.org/72/667672/1/check/build-openstack-releasenotes/2274f7a/html/ : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/72/667672/1/check/kayobe-tox-ansible-syntax/8e1b2a0/ : SUCCESS in 5m 21s\n- kayobe-tox-ansible http://logs.openstack.org/72/667672/1/check/kayobe-tox-ansible/0229856/ : SUCCESS in 7m 35s\n- kayobe-tox-molecule http://logs.openstack.org/72/667672/1/check/kayobe-tox-molecule/08c3f62/ : SUCCESS in 37m 45s\n- kayobe-overcloud-centos http://logs.openstack.org/72/667672/1/check/kayobe-overcloud-centos/41eb7a6/ : SUCCESS in 1h 03m 19s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/72/667672/1/check/kayobe-overcloud-upgrade-centos/9e07d5e/ : SUCCESS in 1h 37m 32s\n- kayobe-seed-centos http://logs.openstack.org/72/667672/1/check/kayobe-seed-centos/164558e/ : SUCCESS in 32m 07s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/72/667672/1/check/kayobe-seed-upgrade-centos/6b2c837/ : SUCCESS in 26m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f332cd4e57acb4d330112bb3ceed3752510a88ed","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-01 21:25:54.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM. Let\u0027s remember to backport https://review.opendev.org/#/c/666831 once merged in master.","accounts_in_message":[],"_revision_number":1},{"id":"22cbac1d5cf3809ae40d8f4cffe00561b99ac512","author":{"_account_id":20301,"name":"Ross Martyn","email":"ross.martyn@gresearch.co.uk","username":"rossmartyn04"},"date":"2019-07-03 09:42:20.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"440511dbad484e0fdcbe8b03040df1dfdecf84cb","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-17 16:58:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c031965dc02046712a98e77ce851fc4544c8a4bf","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-17 16:59:44.000000000","message":"Patch Set 1:\n\nI\u0027ve backported Pierre\u0027s fix - will hold off W+1 until that it is good to go.","accounts_in_message":[],"_revision_number":1},{"id":"f989f42ae498970d3a487afdb7d2f6e528d33723","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-31 16:33:40.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad1a809cf6ff22d47206a21e2a1773ebc10c6e7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 16:33:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2910bd31c69c675fcd40bd432f88b58d09c9d8f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 18:17:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/667672/1/gate/build-openstack-sphinx-docs/b50540b/html/ : SUCCESS in 4m 58s\n- openstack-tox-docs http://logs.openstack.org/72/667672/1/gate/openstack-tox-docs/7fd7913/html/ : SUCCESS in 5m 21s\n- openstack-tox-py35 http://logs.openstack.org/72/667672/1/gate/openstack-tox-py35/e8e7827/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 http://logs.openstack.org/72/667672/1/gate/openstack-tox-pep8/d71b7fc/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/72/667672/1/gate/openstack-tox-py27/bec6caa/ : SUCCESS in 4m 29s\n- build-openstack-releasenotes http://logs.openstack.org/72/667672/1/gate/build-openstack-releasenotes/a9e8518/html/ : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/72/667672/1/gate/kayobe-tox-ansible-syntax/43ab6c4/ : SUCCESS in 5m 01s\n- kayobe-tox-ansible http://logs.openstack.org/72/667672/1/gate/kayobe-tox-ansible/b4c1da8/ : SUCCESS in 8m 16s\n- kayobe-tox-molecule http://logs.openstack.org/72/667672/1/gate/kayobe-tox-molecule/4a6a5f1/ : SUCCESS in 41m 59s\n- kayobe-overcloud-centos http://logs.openstack.org/72/667672/1/gate/kayobe-overcloud-centos/ec839ae/ : SUCCESS in 59m 04s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/72/667672/1/gate/kayobe-overcloud-upgrade-centos/caf373d/ : SUCCESS in 1h 40m 55s\n- kayobe-seed-centos http://logs.openstack.org/72/667672/1/gate/kayobe-seed-centos/b55ec6c/ : SUCCESS in 19m 25s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/72/667672/1/gate/kayobe-seed-upgrade-centos/09c9dde/ : SUCCESS in 29m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fdacacc90fe2d4f18e5c1f3b20ffbef1f7458092","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 18:17:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a133818016fbc11eed26554817b6e65f358a756","revisions":{"0a133818016fbc11eed26554817b6e65f358a756":{"kind":"REWORK","_number":1,"created":"2019-06-26 16:17:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/667672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/667672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/667672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/667672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/667672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/667672/1"}}},"commit":{"parents":[{"commit":"f99aee461ff4055abbbd25bfee1305110a3c0812","subject":"Restrict cliff to \u003c2.15.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f99aee461ff4055abbbd25bfee1305110a3c0812"}]}],"author":{"name":"Ross Martyn","email":"ross.martyn@gresearch.co.uk","date":"2019-04-04 10:46:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-26 16:17:27.000000000","tz":60},"subject":"Multiple updates for Kayobe","message":"Multiple updates for Kayobe\n\n---\n\n1. Gather facts for localhost in kolla-ansible.yml\n2. Don\u0027t include unconfigured networks in network_interfaces\n3. Added Keystone configuration extra config merge\n\n---\n\n1. Facts are necessary for the kolla-ansible role,\nwhich references the ansible_user_uid fact\n\n2. It is possible to skip configuring a network,\nby setting its name to None\n\nThis is done in networks.yml as follows:\n\nadmin_oc_net_name:\n\nCurrently, these networks may still be included in the \u0027network_interfaces\u0027\nlist for each host, despite the fact that they are not in use.\n\nA classic example is when ironic is not enabled, it is currently still\nnecessary to define provisioning and cleaning networks.\n\nThis change avoids including any networks that have their name set to None\nin network_interfaces.\n\n3. Added support for Keystone custom configuration\n\nAdded tests and documentation to add support for keystone extra configuration\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\n\nChange-Id: Iaa304221b8093ac71f9cdbb23edc84d1517578da\n(cherry picked from commit d8ee0fa027140fb8a7504bd6685b6dcdfe3ac8fe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a133818016fbc11eed26554817b6e65f358a756"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a133818016fbc11eed26554817b6e65f358a756"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
