)]}'
{"id":"openstack%2Fkayobe~668452","triplet_id":"openstack%2Fkayobe~master~If8071a9b5c85e5e69fbb333e91c84d10c20d80f3","project":"openstack/kayobe","branch":"master","topic":"story/2006143","hashtags":[],"change_id":"If8071a9b5c85e5e69fbb333e91c84d10c20d80f3","subject":"Adds support for Ansible 2.7, bump mimimum to 2.5","status":"MERGED","created":"2019-07-01 12:24:26.000000000","updated":"2019-07-12 21:28:37.000000000","submitted":"2019-07-12 21:28:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668452-1562966917402-a65eadfb","meta_rev_id":"997ebe63a2b7dc1a081cbe4d32a6e587e2fc4c17","_number":668452,"virtual_id_number":668452,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30322,"name":"Raimund Hook","email":"openstack@sting-ray.za.net","username":"stingrayza"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2019-07-04 10:48:40.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-12 21:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"recommended":{"_account_id":30322,"name":"Raimund Hook","email":"openstack@sting-ray.za.net","username":"stingrayza"},"all":[{"value":1,"date":"2019-07-04 12:48:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30322,"name":"Raimund Hook","email":"openstack@sting-ray.za.net","username":"stingrayza"},{"value":2,"date":"2019-07-12 16:34:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-04 13:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":30322,"name":"Raimund Hook","email":"openstack@sting-ray.za.net","username":"stingrayza"},{"value":1,"date":"2019-07-12 16:34:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30322,"name":"Raimund Hook","email":"openstack@sting-ray.za.net","username":"stingrayza"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30322,"name":"Raimund Hook","email":"openstack@sting-ray.za.net","username":"stingrayza"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30322,"name":"Raimund Hook","email":"openstack@sting-ray.za.net","username":"stingrayza"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 12:48:21.000000000","updated_by":{"_account_id":30322,"name":"Raimund Hook","email":"openstack@sting-ray.za.net","username":"stingrayza"},"reviewer":{"_account_id":30322,"name":"Raimund Hook","email":"openstack@sting-ray.za.net","username":"stingrayza"},"state":"REVIEWER"},{"updated":"2019-07-04 13:06:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-07-12 16:34:09.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-12 21:28:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"339249f31c79e1cdda8a2898ab8b60dca7160dfb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-01 12:24:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0a1e9840b730c6a3d983d6170d3c5c6de6e957f","author":{"_account_id":30322,"name":"Raimund Hook","email":"openstack@sting-ray.za.net","username":"stingrayza"},"date":"2019-07-01 12:44:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2476fd8390bc962a96ca8e021c69bc6aa91c0164","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 14:23:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/52/668452/1/check/openstack-tox-docs/1f9e566/html/ : SUCCESS in 3m 19s\n- openstack-tox-cover http://logs.openstack.org/52/668452/1/check/openstack-tox-cover/a99c54e/cover/ : SUCCESS in 4m 06s\n- openstack-tox-py35 http://logs.openstack.org/52/668452/1/check/openstack-tox-py35/4ad67f8/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/52/668452/1/check/openstack-tox-py36/0082854/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/52/668452/1/check/openstack-tox-pep8/e0e1641/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/52/668452/1/check/openstack-tox-py27/52767e6/ : SUCCESS in 4m 34s\n- build-openstack-releasenotes http://logs.openstack.org/52/668452/1/check/build-openstack-releasenotes/2187e42/html/ : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/52/668452/1/check/kayobe-tox-ansible-syntax/92554b8/ : SUCCESS in 4m 20s\n- kayobe-tox-ansible http://logs.openstack.org/52/668452/1/check/kayobe-tox-ansible/4a828f4/ : FAILURE in 4m 59s\n- kayobe-tox-molecule http://logs.openstack.org/52/668452/1/check/kayobe-tox-molecule/0cc6dae/ : FAILURE in 14m 10s\n- kayobe-overcloud-centos http://logs.openstack.org/52/668452/1/check/kayobe-overcloud-centos/bd5e01a/ : FAILURE in 5m 14s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/52/668452/1/check/kayobe-overcloud-upgrade-centos/7648985/ : SUCCESS in 1h 52m 03s\n- kayobe-seed-centos http://logs.openstack.org/52/668452/1/check/kayobe-seed-centos/65fff0c/ : FAILURE in 5m 44s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/52/668452/1/check/kayobe-seed-upgrade-centos/aaaeff8/ : SUCCESS in 24m 39s","accounts_in_message":[],"_revision_number":1},{"id":"380d951cdcbfe121aa0b29fc8b879f5a7c54b0e6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-01 14:46:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"221d1399c94424d1cdf499125094d93ab4a0de19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 16:55:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/52/668452/2/check/openstack-tox-docs/3629a9f/html/ : SUCCESS in 3m 43s\n- openstack-tox-cover http://logs.openstack.org/52/668452/2/check/openstack-tox-cover/2470e3e/cover/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/52/668452/2/check/openstack-tox-py35/1f8bf0b/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/52/668452/2/check/openstack-tox-py36/244fc6e/ : SUCCESS in 3m 55s\n- openstack-tox-pep8 http://logs.openstack.org/52/668452/2/check/openstack-tox-pep8/7f2d8fd/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/52/668452/2/check/openstack-tox-py27/57e9ad6/ : SUCCESS in 3m 49s\n- build-openstack-releasenotes http://logs.openstack.org/52/668452/2/check/build-openstack-releasenotes/b9335bb/html/ : SUCCESS in 2m 29s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/52/668452/2/check/kayobe-tox-ansible-syntax/493c109/ : SUCCESS in 4m 25s\n- kayobe-tox-ansible http://logs.openstack.org/52/668452/2/check/kayobe-tox-ansible/97dff39/ : SUCCESS in 8m 09s\n- kayobe-tox-molecule http://logs.openstack.org/52/668452/2/check/kayobe-tox-molecule/987795f/ : FAILURE in 14m 37s\n- kayobe-overcloud-centos http://logs.openstack.org/52/668452/2/check/kayobe-overcloud-centos/6322dd2/ : FAILURE in 5m 15s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/52/668452/2/check/kayobe-overcloud-upgrade-centos/d9d8149/ : SUCCESS in 1h 46m 00s\n- kayobe-seed-centos http://logs.openstack.org/52/668452/2/check/kayobe-seed-centos/ca55cd8/ : FAILURE in 4m 42s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/52/668452/2/check/kayobe-seed-upgrade-centos/7dd46d2/ : SUCCESS in 26m 34s","accounts_in_message":[],"_revision_number":2},{"id":"c9f0ebd000477039d36a3f32eddc575b6469ae1e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-01 17:38:02.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d6074d11b2d6acf9866667f21669d57f0a65944e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-01 17:38:14.000000000","message":"Patch Set 3:\n\nTrying 2.7 instead.","accounts_in_message":[],"_revision_number":3},{"id":"b8a749fffa571ce62aba861829883c657896e98d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-01 17:38:18.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"e1fb3fba8e636f34a4fc241db6eb76a757d02ab0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 19:38:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/52/668452/3/check/openstack-tox-docs/4c9787b/html/ : SUCCESS in 3m 13s\n- openstack-tox-cover http://logs.openstack.org/52/668452/3/check/openstack-tox-cover/b11576d/cover/ : SUCCESS in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/52/668452/3/check/openstack-tox-py35/d2d046d/ : SUCCESS in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/52/668452/3/check/openstack-tox-py36/ff60bd8/ : SUCCESS in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/52/668452/3/check/openstack-tox-pep8/3c05d1d/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/52/668452/3/check/openstack-tox-py27/a48c12f/ : SUCCESS in 5m 21s\n- build-openstack-releasenotes http://logs.openstack.org/52/668452/3/check/build-openstack-releasenotes/f29c950/html/ : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/52/668452/3/check/kayobe-tox-ansible-syntax/8815073/ : SUCCESS in 4m 39s\n- kayobe-tox-ansible http://logs.openstack.org/52/668452/3/check/kayobe-tox-ansible/64297f7/ : SUCCESS in 10m 06s\n- kayobe-tox-molecule http://logs.openstack.org/52/668452/3/check/kayobe-tox-molecule/0429c03/ : SUCCESS in 43m 53s\n- kayobe-overcloud-centos http://logs.openstack.org/52/668452/3/check/kayobe-overcloud-centos/dfa0774/ : SUCCESS in 1h 03m 00s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/52/668452/3/check/kayobe-overcloud-upgrade-centos/5406109/ : SUCCESS in 1h 46m 19s\n- kayobe-seed-centos http://logs.openstack.org/52/668452/3/check/kayobe-seed-centos/78607ec/ : SUCCESS in 20m 59s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/52/668452/3/check/kayobe-seed-upgrade-centos/5e7ac82/ : SUCCESS in 24m 32s","accounts_in_message":[],"_revision_number":3},{"id":"cc6819a310bdd68b9d147b835abd9d259cc50720","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-02 07:34:39.000000000","message":"Patch Set 3:\n\nInteresting, at least we know 2.7 works.","accounts_in_message":[],"_revision_number":3},{"id":"83457f65ed2efb2009ee1df1d8c24a64f5126d10","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-02 09:39:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f3cbbb9674c25c89f6cf1a5b4d74803c7d8e7c9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 11:26:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/52/668452/4/check/openstack-tox-docs/f538ad7/html/ : SUCCESS in 3m 06s\n- openstack-tox-cover http://logs.openstack.org/52/668452/4/check/openstack-tox-cover/bec7483/cover/ : SUCCESS in 3m 51s\n- openstack-tox-py35 http://logs.openstack.org/52/668452/4/check/openstack-tox-py35/4043574/ : SUCCESS in 4m 06s\n- openstack-tox-py36 http://logs.openstack.org/52/668452/4/check/openstack-tox-py36/78638b3/ : SUCCESS in 3m 30s\n- openstack-tox-pep8 http://logs.openstack.org/52/668452/4/check/openstack-tox-pep8/532cfee/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/52/668452/4/check/openstack-tox-py27/425f160/ : SUCCESS in 3m 16s\n- build-openstack-releasenotes http://logs.openstack.org/52/668452/4/check/build-openstack-releasenotes/ac92675/html/ : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/52/668452/4/check/kayobe-tox-ansible-syntax/5fde4fd/ : SUCCESS in 3m 57s\n- kayobe-tox-ansible http://logs.openstack.org/52/668452/4/check/kayobe-tox-ansible/138bbeb/ : SUCCESS in 8m 35s\n- kayobe-tox-molecule http://logs.openstack.org/52/668452/4/check/kayobe-tox-molecule/3790378/ : SUCCESS in 51m 20s\n- kayobe-overcloud-centos http://logs.openstack.org/52/668452/4/check/kayobe-overcloud-centos/a61630c/ : SUCCESS in 1h 06m 56s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/52/668452/4/check/kayobe-overcloud-upgrade-centos/68f8460/ : SUCCESS in 1h 37m 55s\n- kayobe-seed-centos http://logs.openstack.org/52/668452/4/check/kayobe-seed-centos/906f699/ : SUCCESS in 23m 26s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/52/668452/4/check/kayobe-seed-upgrade-centos/d746101/ : FAILURE in 1h 07m 46s","accounts_in_message":[],"_revision_number":4},{"id":"bd72735c4173441b1a58e707ab4801c9cc2fffa6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-02 13:16:38.000000000","message":"Patch Set 4:\n\nrecheck\n\nseed mariadb upgrade issue","accounts_in_message":[],"_revision_number":4},{"id":"6df286159c3e0a6ccbd35e946b45ef62d14cc45c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 14:07:02.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/52/668452/4/check/openstack-tox-docs/6157a1a/html/ : SUCCESS in 3m 03s\n- openstack-tox-cover http://logs.openstack.org/52/668452/4/check/openstack-tox-cover/c57170a/cover/ : SUCCESS in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/52/668452/4/check/openstack-tox-py35/da83a6b/ : SUCCESS in 3m 45s\n- openstack-tox-py36 http://logs.openstack.org/52/668452/4/check/openstack-tox-py36/8453cbc/ : SUCCESS in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/52/668452/4/check/openstack-tox-pep8/3165049/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/52/668452/4/check/openstack-tox-py27/b8f8225/ : SUCCESS in 3m 23s\n- build-openstack-releasenotes http://logs.openstack.org/52/668452/4/check/build-openstack-releasenotes/5338673/html/ : SUCCESS in 2m 30s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/52/668452/4/check/kayobe-tox-ansible-syntax/fc74a7e/ : SUCCESS in 3m 50s\n- kayobe-tox-ansible http://logs.openstack.org/52/668452/4/check/kayobe-tox-ansible/cf40803/ : SUCCESS in 6m 55s\n- kayobe-tox-molecule http://logs.openstack.org/52/668452/4/check/kayobe-tox-molecule/6aac010/ : SUCCESS in 43m 32s\n- kayobe-overcloud-centos http://logs.openstack.org/52/668452/4/check/kayobe-overcloud-centos/5569f4c/ : FAILURE in 40m 18s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/52/668452/4/check/kayobe-overcloud-upgrade-centos/fd318a6/ : FAILURE in 41m 19s\n- kayobe-seed-centos http://logs.openstack.org/52/668452/4/check/kayobe-seed-centos/1850d49/ : SUCCESS in 17m 14s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/52/668452/4/check/kayobe-seed-upgrade-centos/e1e420c/ : SUCCESS in 22m 04s","accounts_in_message":[],"_revision_number":4},{"id":"8544ea7b6a12e8a0ea690a6d4f5252e64c9ab17d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-04 08:58:12.000000000","message":"Patch Set 4:\n\nrecheck\n\ngalaxy issues during tenks install","accounts_in_message":[],"_revision_number":4},{"id":"9db3d4278b4d103c3788b59c0012cec3b9aa408c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 10:45:57.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/52/668452/4/check/openstack-tox-docs/957a031/html/ : SUCCESS in 3m 20s\n- openstack-tox-cover http://logs.openstack.org/52/668452/4/check/openstack-tox-cover/5303a52/cover/ : SUCCESS in 4m 03s\n- openstack-tox-py35 http://logs.openstack.org/52/668452/4/check/openstack-tox-py35/3d699c4/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/52/668452/4/check/openstack-tox-py36/58dbdb9/ : SUCCESS in 4m 02s\n- openstack-tox-pep8 http://logs.openstack.org/52/668452/4/check/openstack-tox-pep8/ffe4600/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/52/668452/4/check/openstack-tox-py27/b37bea1/ : SUCCESS in 3m 32s\n- build-openstack-releasenotes http://logs.openstack.org/52/668452/4/check/build-openstack-releasenotes/6157c35/html/ : SUCCESS in 2m 49s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/52/668452/4/check/kayobe-tox-ansible-syntax/9bcdf49/ : SUCCESS in 4m 55s\n- kayobe-tox-ansible http://logs.openstack.org/52/668452/4/check/kayobe-tox-ansible/a8094a3/ : SUCCESS in 8m 49s\n- kayobe-tox-molecule http://logs.openstack.org/52/668452/4/check/kayobe-tox-molecule/ece8f46/ : FAILURE in 4m 32s\n- kayobe-overcloud-centos http://logs.openstack.org/52/668452/4/check/kayobe-overcloud-centos/0c7f98e/ : SUCCESS in 1h 05m 18s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/52/668452/4/check/kayobe-overcloud-upgrade-centos/1bf46c8/ : SUCCESS in 1h 40m 26s\n- kayobe-seed-centos http://logs.openstack.org/52/668452/4/check/kayobe-seed-centos/7c30405/ : SUCCESS in 24m 14s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/52/668452/4/check/kayobe-seed-upgrade-centos/95368d9/ : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":4},{"id":"cc33428e521606dd5d4644e69e5297c9f4683b26","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-04 10:48:40.000000000","message":"Patch Set 4:\n\nrecheck\n\ngalaxy again :(","accounts_in_message":[],"_revision_number":4},{"id":"acdb5a3a1c8d6937bb2d39aef306044879181c79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 12:43:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/52/668452/4/check/openstack-tox-docs/e22b002/html/ : SUCCESS in 3m 36s\n- openstack-tox-cover http://logs.openstack.org/52/668452/4/check/openstack-tox-cover/c4fa9a8/cover/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/52/668452/4/check/openstack-tox-py35/98a220a/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/52/668452/4/check/openstack-tox-py36/409bfba/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/52/668452/4/check/openstack-tox-pep8/12b4ad5/ : SUCCESS in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/52/668452/4/check/openstack-tox-py27/88df2b0/ : SUCCESS in 3m 51s\n- build-openstack-releasenotes http://logs.openstack.org/52/668452/4/check/build-openstack-releasenotes/c213a7f/html/ : SUCCESS in 3m 02s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/52/668452/4/check/kayobe-tox-ansible-syntax/1d4b778/ : SUCCESS in 4m 16s\n- kayobe-tox-ansible http://logs.openstack.org/52/668452/4/check/kayobe-tox-ansible/3cb3238/ : SUCCESS in 7m 46s\n- kayobe-tox-molecule http://logs.openstack.org/52/668452/4/check/kayobe-tox-molecule/597ff14/ : SUCCESS in 39m 21s\n- kayobe-overcloud-centos http://logs.openstack.org/52/668452/4/check/kayobe-overcloud-centos/d3b8857/ : SUCCESS in 1h 04m 56s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/52/668452/4/check/kayobe-overcloud-upgrade-centos/b5a156c/ : SUCCESS in 1h 47m 51s\n- kayobe-seed-centos http://logs.openstack.org/52/668452/4/check/kayobe-seed-centos/ff2bc83/ : SUCCESS in 20m 08s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/52/668452/4/check/kayobe-seed-upgrade-centos/6d7adcb/ : SUCCESS in 27m 28s","accounts_in_message":[],"_revision_number":4},{"id":"76a1bef8ce6f8f39982b521d65ca76af3b71a82e","author":{"_account_id":30322,"name":"Raimund Hook","email":"openstack@sting-ray.za.net","username":"stingrayza"},"date":"2019-07-04 12:48:21.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"346a1c1ccdb68fc3c6630c5d8011bd744af525fb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-04 13:06:36.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a8c07143bc72da1e7f0f5da5286fdce3ebfd60cd","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-12 16:34:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2d223679abbce1ebdac9bd7cb4a1cb6bdbe58178","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 16:34:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"feef597b5f2a00e35c03839bb64fcc04aa21c8a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 21:28:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5ae96b629d2b159ccbac165a39711a52646738dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 21:28:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/52/668452/4/gate/openstack-tox-docs/c0bb2b8/html/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/52/668452/4/gate/openstack-tox-py35/4d25401/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/52/668452/4/gate/openstack-tox-py36/43e4007/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/52/668452/4/gate/openstack-tox-pep8/f1986dc/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/52/668452/4/gate/openstack-tox-py27/2097c97/ : SUCCESS in 4m 32s\n- build-openstack-releasenotes http://logs.openstack.org/52/668452/4/gate/build-openstack-releasenotes/0e53012/html/ : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/52/668452/4/gate/kayobe-tox-ansible-syntax/1f83573/ : SUCCESS in 4m 07s\n- kayobe-tox-ansible http://logs.openstack.org/52/668452/4/gate/kayobe-tox-ansible/5c27cfd/ : SUCCESS in 11m 01s\n- kayobe-tox-molecule http://logs.openstack.org/52/668452/4/gate/kayobe-tox-molecule/6d1ed4c/ : SUCCESS in 52m 39s\n- kayobe-overcloud-centos http://logs.openstack.org/52/668452/4/gate/kayobe-overcloud-centos/d93e185/ : SUCCESS in 1h 16m 22s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/52/668452/4/gate/kayobe-overcloud-upgrade-centos/cf8f3c5/ : SUCCESS in 1h 45m 48s\n- kayobe-seed-centos http://logs.openstack.org/52/668452/4/gate/kayobe-seed-centos/ba7d459/ : SUCCESS in 20m 35s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/52/668452/4/gate/kayobe-seed-upgrade-centos/25536b2/ : SUCCESS in 28m 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"405ccdada2d39da3dae93ea98da052b151e06689","revisions":{"0a481a166c5eae6908255b60dcfe8ea9628c48c3":{"kind":"REWORK","_number":1,"created":"2019-07-01 12:24:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/52/668452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/668452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/668452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/668452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/668452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/668452/1"}}},"commit":{"parents":[{"commit":"67a4d5028879ed5e842ae856f110216ef5378611","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67a4d5028879ed5e842ae856f110216ef5378611"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-01 12:22:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-01 12:22:36.000000000","tz":60},"subject":"Adds support for Ansible 2.8, bump mimimum to 2.5","message":"Adds support for Ansible 2.8, bump mimimum to 2.5\n\nUpdates the minimum version of Ansible from 2.4 to 2.5, and the maximum\nsupported version from 2.6 to 2.8.\n\nChange-Id: If8071a9b5c85e5e69fbb333e91c84d10c20d80f3\nStory: 2006143\nTask: 35639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a481a166c5eae6908255b60dcfe8ea9628c48c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a481a166c5eae6908255b60dcfe8ea9628c48c3"}]},"branch":"refs/heads/master"},"df4ccd07903b3217f3457b0c2339d37d666f396e":{"kind":"REWORK","_number":2,"created":"2019-07-01 14:46:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/52/668452/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/668452/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/668452/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/668452/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/668452/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/668452/2"}}},"commit":{"parents":[{"commit":"67a4d5028879ed5e842ae856f110216ef5378611","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67a4d5028879ed5e842ae856f110216ef5378611"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-01 12:22:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-01 14:45:48.000000000","tz":60},"subject":"Adds support for Ansible 2.8, bump mimimum to 2.5","message":"Adds support for Ansible 2.8, bump mimimum to 2.5\n\nUpdates the minimum version of Ansible from 2.4 to 2.5, and the maximum\nsupported version from 2.6 to 2.8.\n\nChange-Id: If8071a9b5c85e5e69fbb333e91c84d10c20d80f3\nStory: 2006143\nTask: 35639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df4ccd07903b3217f3457b0c2339d37d666f396e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df4ccd07903b3217f3457b0c2339d37d666f396e"}]},"branch":"refs/heads/master"},"ddbb9a9011fa83df4955c7e5bee16c5f939a8a3a":{"kind":"REWORK","_number":3,"created":"2019-07-01 17:38:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/52/668452/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/668452/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/668452/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/668452/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/668452/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/668452/3"}}},"commit":{"parents":[{"commit":"67a4d5028879ed5e842ae856f110216ef5378611","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67a4d5028879ed5e842ae856f110216ef5378611"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-01 12:22:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-01 17:37:49.000000000","tz":0},"subject":"Adds support for Ansible 2.8, bump mimimum to 2.5","message":"Adds support for Ansible 2.8, bump mimimum to 2.5\n\nUpdates the minimum version of Ansible from 2.4 to 2.5, and the maximum\nsupported version from 2.6 to 2.8.\n\nChange-Id: If8071a9b5c85e5e69fbb333e91c84d10c20d80f3\nStory: 2006143\nTask: 35639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ddbb9a9011fa83df4955c7e5bee16c5f939a8a3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ddbb9a9011fa83df4955c7e5bee16c5f939a8a3a"}]},"branch":"refs/heads/master"},"405ccdada2d39da3dae93ea98da052b151e06689":{"kind":"REWORK","_number":4,"created":"2019-07-02 09:39:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/52/668452/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/668452/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/668452/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/668452/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/668452/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/668452/4"}}},"commit":{"parents":[{"commit":"67a4d5028879ed5e842ae856f110216ef5378611","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67a4d5028879ed5e842ae856f110216ef5378611"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-01 12:22:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-02 09:38:54.000000000","tz":60},"subject":"Adds support for Ansible 2.7, bump mimimum to 2.5","message":"Adds support for Ansible 2.7, bump mimimum to 2.5\n\nUpdates the minimum version of Ansible from 2.4 to 2.5, and the maximum\nsupported version from 2.6 to 2.7.\n\nChange-Id: If8071a9b5c85e5e69fbb333e91c84d10c20d80f3\nStory: 2006143\nTask: 35639\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/405ccdada2d39da3dae93ea98da052b151e06689"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/405ccdada2d39da3dae93ea98da052b151e06689"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
