)]}'
{"id":"openstack%2Fkayobe~668632","triplet_id":"openstack%2Fkayobe~master~Ic3932d4af5c0a7639dfeb6cda7e1cd3f61d287c6","project":"openstack/kayobe","branch":"master","topic":"story/2006050","hashtags":[],"change_id":"Ic3932d4af5c0a7639dfeb6cda7e1cd3f61d287c6","subject":"Workaround: stop bifrost services before seed upgrade","status":"MERGED","created":"2019-07-02 08:23:02.000000000","updated":"2019-07-03 12:25:08.000000000","submitted":"2019-07-03 10:29:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668632-1562149751082-ba9f104c","meta_rev_id":"7e3f895027e276422b2bdfcf12746a67481c8a04","_number":668632,"virtual_id_number":668632,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"date":"2019-07-02 13:14:12.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-03 10:29:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-07-03 08:27:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-02 13:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-07-03 08:27:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 13:16:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-07-03 08:27:57.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-07-03 10:29:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"334810d18e9078290d379143eae0dc95b77ed3bc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-02 08:23:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc54fca71b2451bec1ca3ba70250e0e743c1470c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-02 08:39:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f90150aeae969d1b912b9df561b1c068ba15e07","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-02 08:41:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dea260261535451a492a17903a2deb7d43e41ce4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-02 08:51:05.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9771a08163d5a18bc17f69652725c73663887930","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-02 08:52:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f52290027ba5acba28134ca48ec280fb3338869a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 10:39:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/668632/2/check/openstack-tox-docs/1ab8009/html/ : SUCCESS in 3m 07s\n- openstack-tox-cover http://logs.openstack.org/32/668632/2/check/openstack-tox-cover/7e5ea12/cover/ : SUCCESS in 3m 47s\n- openstack-tox-py35 http://logs.openstack.org/32/668632/2/check/openstack-tox-py35/4dd1337/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/32/668632/2/check/openstack-tox-py36/ce2d954/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/32/668632/2/check/openstack-tox-pep8/2cfddf3/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/32/668632/2/check/openstack-tox-py27/6108874/ : SUCCESS in 3m 48s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/32/668632/2/check/kayobe-tox-ansible-syntax/892ebd7/ : SUCCESS in 4m 47s\n- kayobe-tox-ansible http://logs.openstack.org/32/668632/2/check/kayobe-tox-ansible/8486beb/ : SUCCESS in 7m 33s\n- kayobe-tox-molecule http://logs.openstack.org/32/668632/2/check/kayobe-tox-molecule/2ba1cc6/ : SUCCESS in 41m 36s\n- kayobe-overcloud-centos http://logs.openstack.org/32/668632/2/check/kayobe-overcloud-centos/e149c29/ : SUCCESS in 1h 30m 06s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/32/668632/2/check/kayobe-overcloud-upgrade-centos/323b0f9/ : SUCCESS in 1h 36m 47s\n- kayobe-seed-centos http://logs.openstack.org/32/668632/2/check/kayobe-seed-centos/bf35509/ : SUCCESS in 18m 01s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/32/668632/2/check/kayobe-seed-upgrade-centos/9554335/ : SUCCESS in 26m 26s","accounts_in_message":[],"_revision_number":2},{"id":"3c8f04316982e9f6659d72d9de76951bc0add107","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-02 11:07:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5c95f5da1d15486b54084912da2ff3da36ad5776","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-02 11:39:21.000000000","message":"Patch Set 3:\n\nI checked the logs and there are MySQL connection errors if ironic-inspector is not stopped: http://logs.openstack.org/32/668632/2/check/kayobe-seed-upgrade-centos/9554335/primary/kolla/ironic-inspector/ironic-inspector.txt.gz?level\u003dERROR\n\nI pushed a new patch adding it.","accounts_in_message":[],"_revision_number":3},{"id":"0356c9e7e420eb1609bd371af33750e35448a6b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 13:04:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/668632/3/check/openstack-tox-docs/7d052c9/html/ : SUCCESS in 3m 03s\n- openstack-tox-cover http://logs.openstack.org/32/668632/3/check/openstack-tox-cover/c141cf1/cover/ : SUCCESS in 4m 58s\n- openstack-tox-py35 http://logs.openstack.org/32/668632/3/check/openstack-tox-py35/f8b4697/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/32/668632/3/check/openstack-tox-py36/ddb42bc/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/32/668632/3/check/openstack-tox-pep8/62d10f2/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/32/668632/3/check/openstack-tox-py27/fec2354/ : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/32/668632/3/check/kayobe-tox-ansible-syntax/0b190f8/ : SUCCESS in 3m 27s\n- kayobe-tox-ansible http://logs.openstack.org/32/668632/3/check/kayobe-tox-ansible/b122434/ : SUCCESS in 8m 22s\n- kayobe-tox-molecule http://logs.openstack.org/32/668632/3/check/kayobe-tox-molecule/b756d97/ : SUCCESS in 52m 35s\n- kayobe-overcloud-centos http://logs.openstack.org/32/668632/3/check/kayobe-overcloud-centos/103d647/ : SUCCESS in 59m 53s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/32/668632/3/check/kayobe-overcloud-upgrade-centos/a0e2b82/ : SUCCESS in 1h 50m 16s\n- kayobe-seed-centos http://logs.openstack.org/32/668632/3/check/kayobe-seed-centos/caf7fdb/ : SUCCESS in 22m 28s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/32/668632/3/check/kayobe-seed-upgrade-centos/dcb1508/ : SUCCESS in 23m 38s","accounts_in_message":[],"_revision_number":3},{"id":"e4e4777d5a4313a538e1fee2de50f24819403172","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-02 13:14:12.000000000","message":"Patch Set 3:\n\nOops, thanks","accounts_in_message":[],"_revision_number":3},{"id":"4ec4cd297ac32b83c34bf22ceff5e32306918993","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-02 13:16:02.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"21da49df9cf728d77dc1445ad73905a4f03fc818","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-07-03 08:27:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"83b423b99dfb95d7f2b9fe6d848ac58632c4a175","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 08:28:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fb5773c6def3851239c17e06944d303ea76d8e6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 10:29:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/668632/3/gate/openstack-tox-docs/e15cc9c/html/ : SUCCESS in 3m 03s\n- openstack-tox-py35 http://logs.openstack.org/32/668632/3/gate/openstack-tox-py35/2abaf93/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/32/668632/3/gate/openstack-tox-py36/6e13371/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 http://logs.openstack.org/32/668632/3/gate/openstack-tox-pep8/b7cb70b/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/32/668632/3/gate/openstack-tox-py27/7ed0f77/ : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/32/668632/3/gate/kayobe-tox-ansible-syntax/8041f07/ : SUCCESS in 6m 25s\n- kayobe-tox-ansible http://logs.openstack.org/32/668632/3/gate/kayobe-tox-ansible/c823843/ : SUCCESS in 7m 46s\n- kayobe-tox-molecule http://logs.openstack.org/32/668632/3/gate/kayobe-tox-molecule/c410f49/ : SUCCESS in 51m 48s\n- kayobe-overcloud-centos http://logs.openstack.org/32/668632/3/gate/kayobe-overcloud-centos/eb00232/ : SUCCESS in 1h 00m 23s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/32/668632/3/gate/kayobe-overcloud-upgrade-centos/54a2e82/ : SUCCESS in 1h 52m 15s\n- kayobe-seed-centos http://logs.openstack.org/32/668632/3/gate/kayobe-seed-centos/3a8f802/ : SUCCESS in 25m 09s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/32/668632/3/gate/kayobe-seed-upgrade-centos/acec224/ : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":3},{"id":"989bfb034f4fd62fdd4669c636166fbaed9c4b97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 10:29:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ae54bf7ecd24002d08a51bed7cf7f52486a3423e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-03 12:25:08.000000000","message":"Created a revert of this change as I97827a9cf0056b5341811c37dc996fd6b69f2f1f","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"77e8044a406b0cfeef377e0155b2fbcb5f05a67e","revisions":{"3b4d98348adc43ba62b25535626d5964d8231e8d":{"kind":"REWORK","_number":1,"created":"2019-07-02 08:23:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/668632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/668632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/668632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/668632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/668632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/668632/1"}}},"commit":{"parents":[{"commit":"67a4d5028879ed5e842ae856f110216ef5378611","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67a4d5028879ed5e842ae856f110216ef5378611"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-02 08:19:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-02 08:19:09.000000000","tz":60},"subject":"Workaround: stop bifrost services before seed upgrade","message":"Workaround: stop bifrost services before seed upgrade\n\nThis avoids mariadb being ungracefully shut down, which prevents it from\nbeing started again.\n\nThis should be reverted once bug 1834332 has been fixed in kolla ansible\nstein.\n\nChange-Id: Ic3932d4af5c0a7639dfeb6cda7e1cd3f61d287c6\nStory: 2006050\nTask: 34743\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b4d98348adc43ba62b25535626d5964d8231e8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b4d98348adc43ba62b25535626d5964d8231e8d"}]},"branch":"refs/heads/master"},"89c5b4f79735ee0971f6acbf9a83735382cdbd46":{"kind":"REWORK","_number":2,"created":"2019-07-02 08:52:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/668632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/668632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/668632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/668632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/668632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/668632/2"}}},"commit":{"parents":[{"commit":"67a4d5028879ed5e842ae856f110216ef5378611","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67a4d5028879ed5e842ae856f110216ef5378611"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-02 08:19:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-02 08:51:59.000000000","tz":60},"subject":"Workaround: stop bifrost services before seed upgrade","message":"Workaround: stop bifrost services before seed upgrade\n\nThis avoids mariadb being ungracefully shut down, which prevents it from\nbeing started again.\n\nThis should be reverted once bug 1834332 has been fixed in kolla ansible\nstein.\n\nChange-Id: Ic3932d4af5c0a7639dfeb6cda7e1cd3f61d287c6\nStory: 2006050\nTask: 34743\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89c5b4f79735ee0971f6acbf9a83735382cdbd46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89c5b4f79735ee0971f6acbf9a83735382cdbd46"}]},"branch":"refs/heads/master"},"77e8044a406b0cfeef377e0155b2fbcb5f05a67e":{"kind":"REWORK","_number":3,"created":"2019-07-02 11:07:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/32/668632/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/668632/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/668632/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/668632/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/668632/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/668632/3"}}},"commit":{"parents":[{"commit":"67a4d5028879ed5e842ae856f110216ef5378611","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67a4d5028879ed5e842ae856f110216ef5378611"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-02 08:19:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-02 11:06:47.000000000","tz":60},"subject":"Workaround: stop bifrost services before seed upgrade","message":"Workaround: stop bifrost services before seed upgrade\n\nThis avoids mariadb being ungracefully shut down, which prevents it from\nbeing started again.\n\nThis should be reverted once bug 1834332 has been fixed in kolla ansible\nstein.\n\nChange-Id: Ic3932d4af5c0a7639dfeb6cda7e1cd3f61d287c6\nStory: 2006050\nTask: 34743\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77e8044a406b0cfeef377e0155b2fbcb5f05a67e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77e8044a406b0cfeef377e0155b2fbcb5f05a67e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
