)]}'
{"id":"openstack%2Fkayobe~669342","triplet_id":"openstack%2Fkayobe~master~I9cc0bf08aa69386a18278f240c751a4d9956f926","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I9cc0bf08aa69386a18278f240c751a4d9956f926","subject":"Publish docs and release notes on openstack.o.o","status":"MERGED","created":"2019-07-05 12:05:02.000000000","updated":"2019-08-12 15:52:25.000000000","submitted":"2019-08-12 15:52:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669342-1565625145847-f7d3f53d","meta_rev_id":"cd4296be82ec7aa2ef3bd12cd9d61951435eaf06","_number":669342,"virtual_id_number":669342,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-12 15:52:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-08-08 09:26:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-08-07 16:00:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-08-08 09:26:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-08-07 13:29:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 16:00:21.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-08-07 16:03:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-08-08 09:26:09.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-08-12 15:52:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d535f4efb83ddfc922a939ed4d88c9ce79defcfa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-05 12:05:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d27cdb431b669344f263b1640ffaa7c3647fdc66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 13:52:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/42/669342/1/check/openstack-tox-cover/9b9b139/cover/ : SUCCESS in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/42/669342/1/check/openstack-tox-py35/472cb66/ : SUCCESS in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/42/669342/1/check/openstack-tox-py36/e21934d/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/42/669342/1/check/openstack-tox-pep8/f23a182/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/42/669342/1/check/openstack-tox-py27/e7e02cc/ : SUCCESS in 3m 43s\n- openstack-tox-docs http://logs.openstack.org/42/669342/1/check/openstack-tox-docs/2ef33dc/html/ : SUCCESS in 2m 41s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/42/669342/1/check/kayobe-tox-ansible-syntax/da07aa2/ : SUCCESS in 5m 27s\n- kayobe-tox-ansible http://logs.openstack.org/42/669342/1/check/kayobe-tox-ansible/e24c345/ : SUCCESS in 9m 52s\n- kayobe-tox-molecule http://logs.openstack.org/42/669342/1/check/kayobe-tox-molecule/08fc5a3/ : SUCCESS in 53m 25s\n- kayobe-overcloud-centos http://logs.openstack.org/42/669342/1/check/kayobe-overcloud-centos/a136295/ : SUCCESS in 1h 09m 40s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/42/669342/1/check/kayobe-overcloud-upgrade-centos/e51de2e/ : SUCCESS in 1h 41m 04s\n- kayobe-seed-centos http://logs.openstack.org/42/669342/1/check/kayobe-seed-centos/a37ff54/ : SUCCESS in 20m 28s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/42/669342/1/check/kayobe-seed-upgrade-centos/f320e0f/ : SUCCESS in 24m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0962aa31985379ac4257004751e3cb501cd781d5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 08:57:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"086ff7c2061e0a45bf04fa55079f83d9446de69d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 14:15:18.000000000","message":"Patch Set 1: Workflow-1\n\nLet\u0027s wait for governance patch to land.","accounts_in_message":[],"_revision_number":1},{"id":"ca3798f8d59ae5d0d7234d4755e74d786e88bc7a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-03 18:02:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"9e6be50a1045641406ec01e5b4dd227e081752df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 19:42:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/42/669342/2/check/test-release-openstack/d6c2f7f/ : SUCCESS in 2m 20s\n- openstack-tox-cover https://logs.opendev.org/42/669342/2/check/openstack-tox-cover/99505ab/cover/ : SUCCESS in 4m 26s\n- openstack-tox-py35 https://logs.opendev.org/42/669342/2/check/openstack-tox-py35/557cdb4/ : SUCCESS in 4m 00s\n- openstack-tox-py36 https://logs.opendev.org/42/669342/2/check/openstack-tox-py36/7908948/ : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://logs.opendev.org/42/669342/2/check/openstack-tox-pep8/0e887e4/ : SUCCESS in 4m 02s\n- openstack-tox-py27 https://logs.opendev.org/42/669342/2/check/openstack-tox-py27/0ebf8f9/ : SUCCESS in 3m 47s\n- openstack-tox-docs https://logs.opendev.org/42/669342/2/check/openstack-tox-docs/5f19def/html/ : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://logs.opendev.org/42/669342/2/check/build-openstack-releasenotes/3e82f97/html/ : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://logs.opendev.org/42/669342/2/check/kayobe-tox-ansible-syntax/6f36a0f/ : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://logs.opendev.org/42/669342/2/check/kayobe-tox-ansible/cc029b3/ : SUCCESS in 10m 12s\n- kayobe-tox-molecule https://logs.opendev.org/42/669342/2/check/kayobe-tox-molecule/57365a1/ : SUCCESS in 59m 22s\n- kayobe-overcloud-centos https://logs.opendev.org/42/669342/2/check/kayobe-overcloud-centos/a25753c/ : SUCCESS in 1h 36m 54s\n- kayobe-overcloud-upgrade-centos https://logs.opendev.org/42/669342/2/check/kayobe-overcloud-upgrade-centos/bca70c7/ : FAILURE in 3m 26s\n- kayobe-seed-centos https://logs.opendev.org/42/669342/2/check/kayobe-seed-centos/26cc680/ : SUCCESS in 17m 57s\n- kayobe-seed-upgrade-centos https://logs.opendev.org/42/669342/2/check/kayobe-seed-upgrade-centos/763b6e8/ : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":2},{"id":"cdc956470cc757006edc4be550347e9986161a26","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-07 11:39:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"37aaa6fad186d93e17d502927c65e329538401db","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-07 11:41:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"04a75b14d6443405da57724255a00b8d7adb2d39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 13:24:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://logs.opendev.org/42/669342/2/check/test-release-openstack/dc7dd6b/ : SUCCESS in 3m 29s\n- openstack-tox-cover https://logs.opendev.org/42/669342/2/check/openstack-tox-cover/2fba5b2/cover/ : SUCCESS in 5m 16s\n- openstack-tox-py35 https://logs.opendev.org/42/669342/2/check/openstack-tox-py35/f3a4c91/ : SUCCESS in 4m 51s\n- openstack-tox-py36 https://logs.opendev.org/42/669342/2/check/openstack-tox-py36/d30c662/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://logs.opendev.org/42/669342/2/check/openstack-tox-pep8/5934317/ : SUCCESS in 4m 12s\n- openstack-tox-py27 https://logs.opendev.org/42/669342/2/check/openstack-tox-py27/852a49e/ : SUCCESS in 3m 28s\n- openstack-tox-docs https://logs.opendev.org/42/669342/2/check/openstack-tox-docs/d5bf439/html/ : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://logs.opendev.org/42/669342/2/check/build-openstack-releasenotes/1ae21bd/html/ : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax https://logs.opendev.org/42/669342/2/check/kayobe-tox-ansible-syntax/3a29597/ : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://logs.opendev.org/42/669342/2/check/kayobe-tox-ansible/930a232/ : SUCCESS in 11m 53s\n- kayobe-tox-molecule https://logs.opendev.org/42/669342/2/check/kayobe-tox-molecule/de003f2/ : SUCCESS in 43m 09s\n- kayobe-overcloud-centos https://logs.opendev.org/42/669342/2/check/kayobe-overcloud-centos/70bfb30/ : SUCCESS in 1h 03m 33s\n- kayobe-overcloud-upgrade-centos https://logs.opendev.org/42/669342/2/check/kayobe-overcloud-upgrade-centos/4a9f74c/ : SUCCESS in 1h 39m 38s\n- kayobe-seed-centos https://logs.opendev.org/42/669342/2/check/kayobe-seed-centos/29840bc/ : FAILURE in 23m 03s\n- kayobe-seed-upgrade-centos https://logs.opendev.org/42/669342/2/check/kayobe-seed-upgrade-centos/b391183/ : SUCCESS in 32m 52s","accounts_in_message":[],"_revision_number":2},{"id":"091a41f58273c282cc07bbf5f5b872f5323733f5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-07 13:29:32.000000000","message":"Patch Set 2:\n\nrecheck\n\nnetwork issue","accounts_in_message":[],"_revision_number":2},{"id":"97733a48389d74fc0a3c8e8a62d57bfa17d2a595","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-07 13:29:40.000000000","message":"Patch Set 2: -Code-Review Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"a6f084fcfab56da9adcfe571297288c0d438511c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 15:30:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://logs.opendev.org/42/669342/2/check/test-release-openstack/00177f3/ : SUCCESS in 2m 17s\n- openstack-tox-cover https://logs.opendev.org/42/669342/2/check/openstack-tox-cover/e690f8d/cover/ : SUCCESS in 5m 27s\n- openstack-tox-py35 https://logs.opendev.org/42/669342/2/check/openstack-tox-py35/1b04e39/ : SUCCESS in 4m 21s\n- openstack-tox-py36 https://logs.opendev.org/42/669342/2/check/openstack-tox-py36/5065e21/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://logs.opendev.org/42/669342/2/check/openstack-tox-pep8/a7ac425/ : SUCCESS in 4m 45s\n- openstack-tox-py27 https://logs.opendev.org/42/669342/2/check/openstack-tox-py27/32a4c37/ : SUCCESS in 5m 07s\n- openstack-tox-docs https://logs.opendev.org/42/669342/2/check/openstack-tox-docs/80fd420/html/ : SUCCESS in 3m 03s\n- build-openstack-releasenotes https://logs.opendev.org/42/669342/2/check/build-openstack-releasenotes/3ae4a8d/html/ : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://logs.opendev.org/42/669342/2/check/kayobe-tox-ansible-syntax/928156f/ : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://logs.opendev.org/42/669342/2/check/kayobe-tox-ansible/90b81b7/ : SUCCESS in 9m 26s\n- kayobe-tox-molecule https://logs.opendev.org/42/669342/2/check/kayobe-tox-molecule/96dfabf/ : SUCCESS in 44m 08s\n- kayobe-overcloud-centos https://logs.opendev.org/42/669342/2/check/kayobe-overcloud-centos/e41660e/ : SUCCESS in 1h 20m 02s\n- kayobe-overcloud-upgrade-centos https://logs.opendev.org/42/669342/2/check/kayobe-overcloud-upgrade-centos/b0b6cfc/ : SUCCESS in 1h 50m 42s\n- kayobe-seed-centos https://logs.opendev.org/42/669342/2/check/kayobe-seed-centos/4ace060/ : SUCCESS in 21m 44s\n- kayobe-seed-upgrade-centos https://logs.opendev.org/42/669342/2/check/kayobe-seed-upgrade-centos/701725e/ : SUCCESS in 30m 05s","accounts_in_message":[],"_revision_number":2},{"id":"1294f4ef02faa0dd6ac02650e87c854443e1beab","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-07 16:00:21.000000000","message":"Patch Set 2: Code-Review+2\n\nRef: https://github.com/openstack/kolla-ansible/blob/master/zuul.d/project.yaml","accounts_in_message":[],"_revision_number":2},{"id":"587285d7cf3a42e3beeaf897fa4cf93c123ee8b3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-07 16:03:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca08f4690fa59230f61102584e0b4bed33fe5e58","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-07 16:40:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf1d08292fe305f123a80c5fbd281eb31986a0c9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-08-08 09:26:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a749b8e121eb88b1d93c3bdf3eb773a7d5fdb200","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 14:03:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"10eb98b48deb51a62880d19bb6d2cd220c035bee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:52:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1e2fd1b5b59d43f0932e1a8d5b6f125f : SUCCESS in 2m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/044b0b7acdb24235b65d17691c04a305 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9187f4a87889462284b170b7c6b7ac83 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7928dc909506440cb9556a56a6933602 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da13159ceb2041aa8f81304fb84d770e : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f18baa126c46e3b838f86f6934d991 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c4366ccd922417588dfa5128d1fc7f8 : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0380e16f75004c4d9c818ad4a963cea3 : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0373589f40f8425388d1069fae03ed0b : SUCCESS in 7m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/809c9ebf9fcc4bd487def95b42e860e5 : SUCCESS in 32m 43s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/54cb4dd5e81e4c3da7b6e3f22422756b : SUCCESS in 1h 21m 47s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/0dbd8263dc2f4c1bba8968a9fc1e5223 : SUCCESS in 1h 40m 03s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b7f619071f0442ef9985b02e06b8600f : SUCCESS in 19m 31s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e7869e2ff1364cc5b0411a4348b67e0c : SUCCESS in 29m 56s","accounts_in_message":[],"_revision_number":2},{"id":"cd4296be82ec7aa2ef3bd12cd9d61951435eaf06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:52:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4286c322990c3bc6689779e3fb7216c83ab81517","revisions":{"17beb150378338079ddecef06e8a88c8f771ce7b":{"kind":"REWORK","_number":1,"created":"2019-07-05 12:05:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/669342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/669342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/669342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/669342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/669342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/669342/1"}}},"commit":{"parents":[{"commit":"b1f54eabe13988b7091f85ebcd98d304d314798e","subject":"Merge \"Add retries to ansible galaxy install for all envs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b1f54eabe13988b7091f85ebcd98d304d314798e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-05 12:02:00.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-05 12:02:00.000000000","tz":60},"subject":"Publish docs and release notes on openstack.o.o","message":"Publish docs and release notes on openstack.o.o\n\nKayobe is now a kolla deliverable [1], so should use the official\nsite for publishing release notes and docs.\n\nLinks will be updated once these have been published.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007250.html\n\nChange-Id: I9cc0bf08aa69386a18278f240c751a4d9956f926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17beb150378338079ddecef06e8a88c8f771ce7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17beb150378338079ddecef06e8a88c8f771ce7b"}]},"branch":"refs/heads/master"},"4286c322990c3bc6689779e3fb7216c83ab81517":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-03 18:02:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/669342/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/669342/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/669342/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/669342/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/669342/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/669342/2"}}},"commit":{"parents":[{"commit":"c44f27b4ae83449dbbb44827e79ddd749c21bb21","subject":"Move to openstackdocstheme for docs \u0026 release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c44f27b4ae83449dbbb44827e79ddd749c21bb21"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-05 12:02:00.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-03 18:02:20.000000000","tz":60},"subject":"Publish docs and release notes on openstack.o.o","message":"Publish docs and release notes on openstack.o.o\n\nKayobe is now a kolla deliverable [1], so should use the official\nsite for publishing release notes and docs.\n\nLinks will be updated once these have been published.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-June/007250.html\n\nChange-Id: I9cc0bf08aa69386a18278f240c751a4d9956f926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4286c322990c3bc6689779e3fb7216c83ab81517"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4286c322990c3bc6689779e3fb7216c83ab81517"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
