)]}'
{"id":"openstack%2Fkayobe~669715","triplet_id":"openstack%2Fkayobe~stable%2Fqueens~Ia099772c41b4b6227d5b1da26b53781c69a1edd5","project":"openstack/kayobe","branch":"stable/queens","topic":"add-config-tags-stable/queens","hashtags":[],"change_id":"Ia099772c41b4b6227d5b1da26b53781c69a1edd5","subject":"Add config tags to kolla-openstack play","status":"MERGED","created":"2019-07-08 16:33:42.000000000","updated":"2019-07-17 18:47:21.000000000","submitted":"2019-07-17 18:47:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669715-1563389241973-5870c8eb","meta_rev_id":"3b4e910f740143935595102e63f87e1f3bdf53f1","_number":669715,"virtual_id_number":669715,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-07-17 18:47:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-07-17 16:51:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-08 16:54:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-07-17 16:51:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 16:54:58.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-07-17 16:51:50.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-17 18:47:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01e67b96295fabd1d7b8004fc78f7ceab62674d8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 16:33:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d4d380bc397398dbbac29d96053ef5a2ecf85b40","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-08 16:54:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1ebfcc1f9ade28c8afcfc2438b3cf3191134b2ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 18:50:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/15/669715/1/check/build-openstack-sphinx-docs/1a8da20/html/ : SUCCESS in 3m 47s\n- openstack-tox-cover http://logs.openstack.org/15/669715/1/check/openstack-tox-cover/ab5f143/cover/ : SUCCESS in 4m 02s\n- openstack-tox-py35 http://logs.openstack.org/15/669715/1/check/openstack-tox-py35/c8fdf95/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/15/669715/1/check/openstack-tox-pep8/41dd631/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/15/669715/1/check/openstack-tox-py27/a1210a8/ : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/15/669715/1/check/kayobe-tox-ansible-syntax/c2cbce6/ : SUCCESS in 4m 20s\n- kayobe-tox-ansible http://logs.openstack.org/15/669715/1/check/kayobe-tox-ansible/71d00d4/ : SUCCESS in 7m 29s\n- kayobe-tox-molecule http://logs.openstack.org/15/669715/1/check/kayobe-tox-molecule/f6b2e89/ : SUCCESS in 40m 20s\n- kayobe-overcloud-centos http://logs.openstack.org/15/669715/1/check/kayobe-overcloud-centos/a59d784/ : SUCCESS in 48m 54s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/15/669715/1/check/kayobe-overcloud-upgrade-centos/57eb3b1/ : SUCCESS in 1h 48m 05s\n- kayobe-seed-centos http://logs.openstack.org/15/669715/1/check/kayobe-seed-centos/ed5dbb5/ : SUCCESS in 14m 57s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/15/669715/1/check/kayobe-seed-upgrade-centos/d5a5f30/ : SUCCESS in 25m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b67e5d650db44b911977eb2503649755aee18129","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-17 16:51:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2acc8a67c8e17b33153d7bebe4a5319276168f06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 16:52:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ae5181cc112dc89ec7640fb5d9f4f26e9848601","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:47:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/15/669715/1/gate/build-openstack-sphinx-docs/1556b80/html/ : SUCCESS in 4m 42s\n- openstack-tox-py35 http://logs.openstack.org/15/669715/1/gate/openstack-tox-py35/25cb37d/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/15/669715/1/gate/openstack-tox-pep8/5dd7c68/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/15/669715/1/gate/openstack-tox-py27/89c2cb7/ : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/15/669715/1/gate/kayobe-tox-ansible-syntax/1d13e8a/ : SUCCESS in 4m 38s\n- kayobe-tox-ansible http://logs.openstack.org/15/669715/1/gate/kayobe-tox-ansible/7b24f67/ : SUCCESS in 8m 59s\n- kayobe-tox-molecule http://logs.openstack.org/15/669715/1/gate/kayobe-tox-molecule/0ea25f5/ : SUCCESS in 29m 33s\n- kayobe-overcloud-centos http://logs.openstack.org/15/669715/1/gate/kayobe-overcloud-centos/52f5094/ : SUCCESS in 45m 00s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/15/669715/1/gate/kayobe-overcloud-upgrade-centos/ddeb068/ : SUCCESS in 1h 18m 10s\n- kayobe-seed-centos http://logs.openstack.org/15/669715/1/gate/kayobe-seed-centos/f022bdb/ : SUCCESS in 16m 47s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/15/669715/1/gate/kayobe-seed-upgrade-centos/7c8a6c6/ : SUCCESS in 32m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5ae20528cb4cef53ec51bb9e4d21d93e96f41a39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:47:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b88bc521fdb9e412d379b1673864daf967da4d10","revisions":{"b88bc521fdb9e412d379b1673864daf967da4d10":{"kind":"REWORK","_number":1,"created":"2019-07-08 16:33:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/669715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/669715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/669715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/669715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/669715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/669715/1"}}},"commit":{"parents":[{"commit":"c5e6de8eb1787b722fb66c11a7dbab622748a967","subject":"Restrict cliff to \u003c2.15.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c5e6de8eb1787b722fb66c11a7dbab622748a967"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-20 14:11:31.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-08 16:33:42.000000000","tz":0},"subject":"Add config tags to kolla-openstack play","message":"Add config tags to kolla-openstack play\n\nEnsures that the group_by task runs to check if ironic is enabled.\n\nChange-Id: Ia099772c41b4b6227d5b1da26b53781c69a1edd5\nTrivialFix\n(cherry picked from commit 9de8efd0bf9858a1c312e53782746482216d6569)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b88bc521fdb9e412d379b1673864daf967da4d10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b88bc521fdb9e412d379b1673864daf967da4d10"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
