)]}'
{"id":"openstack%2Fkayobe~670131","triplet_id":"openstack%2Fkayobe~master~Id17d4ea00ceb684b34bf12e911758d504c520de0","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Id17d4ea00ceb684b34bf12e911758d504c520de0","subject":"CI: Don\u0027t create an external network in init-runonce","status":"MERGED","created":"2019-07-10 16:54:48.000000000","updated":"2019-07-11 15:50:02.000000000","submitted":"2019-07-11 15:50:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670131-1562860202619-0ae53db0","meta_rev_id":"84e549be25401d005dc70d6f0deb7d64d75440ac","_number":670131,"virtual_id_number":670131,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-11 15:50:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-07-11 14:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-11 14:02:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-07-11 14:05:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 16:57:50.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-11 14:02:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-07-11 15:50:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40b3915bc52df0624e4c1957dbfa3a91fa483df5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-10 16:54:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0628050bc071ee62d4574f0bf7aca8652a4e423f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-10 16:57:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a67740d82403d479e558813724a656bc0763d30a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-10 17:46:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e91662d5a52098d12c4a14d92fb324671873f08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:55:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/31/670131/1/check/openstack-tox-docs/3ba0d0d/html/ : SUCCESS in 3m 15s\n- openstack-tox-cover http://logs.openstack.org/31/670131/1/check/openstack-tox-cover/fddea7c/cover/ : SUCCESS in 4m 05s\n- openstack-tox-py35 http://logs.openstack.org/31/670131/1/check/openstack-tox-py35/d5e6d0b/ : SUCCESS in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/31/670131/1/check/openstack-tox-py36/bd88347/ : SUCCESS in 4m 01s\n- openstack-tox-pep8 http://logs.openstack.org/31/670131/1/check/openstack-tox-pep8/d947506/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/31/670131/1/check/openstack-tox-py27/e7d7d87/ : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/31/670131/1/check/kayobe-tox-ansible-syntax/0abf48e/ : SUCCESS in 5m 16s\n- kayobe-tox-ansible http://logs.openstack.org/31/670131/1/check/kayobe-tox-ansible/4babb0a/ : SUCCESS in 10m 25s\n- kayobe-tox-molecule http://logs.openstack.org/31/670131/1/check/kayobe-tox-molecule/82f3bc7/ : SUCCESS in 45m 42s\n- kayobe-overcloud-centos http://logs.openstack.org/31/670131/1/check/kayobe-overcloud-centos/14151d5/ : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/31/670131/1/check/kayobe-overcloud-upgrade-centos/35866a3/ : SUCCESS in 1h 46m 04s\n- kayobe-seed-centos http://logs.openstack.org/31/670131/1/check/kayobe-seed-centos/d5ce2e9/ : SUCCESS in 48m 31s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/31/670131/1/check/kayobe-seed-upgrade-centos/65dee30/ : SUCCESS in 28m 27s","accounts_in_message":[],"_revision_number":1},{"id":"70622de44170b623dd3e43a58960594b45613a9f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-11 06:58:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3b0d65ded27d3b0b615c379af02a8d77df30ccbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 08:43:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/31/670131/2/check/openstack-tox-docs/e5fd80c/html/ : SUCCESS in 3m 26s\n- openstack-tox-cover http://logs.openstack.org/31/670131/2/check/openstack-tox-cover/f5e3dfc/cover/ : SUCCESS in 4m 41s\n- openstack-tox-py35 http://logs.openstack.org/31/670131/2/check/openstack-tox-py35/b2e0fa5/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/31/670131/2/check/openstack-tox-py36/8a829ca/ : SUCCESS in 5m 42s\n- openstack-tox-pep8 http://logs.openstack.org/31/670131/2/check/openstack-tox-pep8/aef1830/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/31/670131/2/check/openstack-tox-py27/7ddc7da/ : SUCCESS in 3m 58s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/31/670131/2/check/kayobe-tox-ansible-syntax/1c343ec/ : SUCCESS in 6m 11s\n- kayobe-tox-ansible http://logs.openstack.org/31/670131/2/check/kayobe-tox-ansible/f7454ff/ : SUCCESS in 8m 22s\n- kayobe-tox-molecule http://logs.openstack.org/31/670131/2/check/kayobe-tox-molecule/7ccb7d6/ : SUCCESS in 53m 00s\n- kayobe-overcloud-centos http://logs.openstack.org/31/670131/2/check/kayobe-overcloud-centos/0a697c7/ : SUCCESS in 58m 45s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/31/670131/2/check/kayobe-overcloud-upgrade-centos/d859865/ : SUCCESS in 1h 38m 20s\n- kayobe-seed-centos http://logs.openstack.org/31/670131/2/check/kayobe-seed-centos/fba78b7/ : SUCCESS in 25m 37s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/31/670131/2/check/kayobe-seed-upgrade-centos/12f333d/ : SUCCESS in 28m 21s","accounts_in_message":[],"_revision_number":2},{"id":"18cd12a73da28bbb090bcea151a8c65dc5c94a1a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-11 14:02:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"16351fabb0406e039f84f4f7088f1fa17b77a4b6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-11 14:05:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"59f82fc854156c4faba9335ffba4e6cec5209560","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 14:05:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"85acc50ce333cc635b2400e92e30304c8b960931","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 15:50:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/31/670131/2/gate/openstack-tox-docs/adbb999/html/ : SUCCESS in 3m 00s\n- openstack-tox-py35 http://logs.openstack.org/31/670131/2/gate/openstack-tox-py35/52c50d6/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/31/670131/2/gate/openstack-tox-py36/94c7a83/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/31/670131/2/gate/openstack-tox-pep8/6635440/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/31/670131/2/gate/openstack-tox-py27/62d44cd/ : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/31/670131/2/gate/kayobe-tox-ansible-syntax/00aaec1/ : SUCCESS in 4m 02s\n- kayobe-tox-ansible http://logs.openstack.org/31/670131/2/gate/kayobe-tox-ansible/e5fdd2c/ : SUCCESS in 8m 25s\n- kayobe-tox-molecule http://logs.openstack.org/31/670131/2/gate/kayobe-tox-molecule/08cf398/ : SUCCESS in 51m 43s\n- kayobe-overcloud-centos http://logs.openstack.org/31/670131/2/gate/kayobe-overcloud-centos/ee4d822/ : SUCCESS in 55m 23s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/31/670131/2/gate/kayobe-overcloud-upgrade-centos/9df71ba/ : SUCCESS in 1h 41m 42s\n- kayobe-seed-centos http://logs.openstack.org/31/670131/2/gate/kayobe-seed-centos/a7ddc21/ : SUCCESS in 22m 52s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/31/670131/2/gate/kayobe-seed-upgrade-centos/6b63ab6/ : SUCCESS in 27m 52s","accounts_in_message":[],"_revision_number":2},{"id":"84e549be25401d005dc70d6f0deb7d64d75440ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 15:50:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0707e2c53d46801c12181d1bcd772f74f14f35b6","revisions":{"b87792ea1a0d1266b9f2e39ddac3193dd7ecff95":{"kind":"REWORK","_number":1,"created":"2019-07-10 16:54:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/670131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/670131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/670131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/670131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/670131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/670131/1"}}},"commit":{"parents":[{"commit":"7e83b264e02918afd51edd3c1a867098b2597947","subject":"Merge \"Fix custom ironic python agent environmental variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e83b264e02918afd51edd3c1a867098b2597947"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-10 16:52:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-10 16:54:04.000000000","tz":60},"subject":"CI: Don\u0027t create an external network in init-runonce","message":"CI: Don\u0027t create an external network in init-runonce\n\nThis network conflicts with the ironic provision-net since both are flat\nnetworks on physnet1. We don\u0027t need the external network in our CI, so\ndon\u0027t create it.\n\nChange-Id: Id17d4ea00ceb684b34bf12e911758d504c520de0\nDepends-On: https://review.opendev.org/#/c/670129/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b87792ea1a0d1266b9f2e39ddac3193dd7ecff95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b87792ea1a0d1266b9f2e39ddac3193dd7ecff95"}]},"branch":"refs/heads/master"},"0707e2c53d46801c12181d1bcd772f74f14f35b6":{"kind":"REWORK","_number":2,"created":"2019-07-11 06:58:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/670131/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/670131/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/670131/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/670131/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/670131/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/670131/2"}}},"commit":{"parents":[{"commit":"7e83b264e02918afd51edd3c1a867098b2597947","subject":"Merge \"Fix custom ironic python agent environmental variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e83b264e02918afd51edd3c1a867098b2597947"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-10 16:52:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-11 06:58:35.000000000","tz":0},"subject":"CI: Don\u0027t create an external network in init-runonce","message":"CI: Don\u0027t create an external network in init-runonce\n\nThis network conflicts with the ironic provision-net since both are flat\nnetworks on physnet1. We don\u0027t need the external network in our CI, so\ndon\u0027t create it.\n\nChange-Id: Id17d4ea00ceb684b34bf12e911758d504c520de0\nDepends-On: https://review.opendev.org/#/c/670129/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0707e2c53d46801c12181d1bcd772f74f14f35b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0707e2c53d46801c12181d1bcd772f74f14f35b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
