)]}'
{"id":"openstack%2Fkayobe~670502","triplet_id":"openstack%2Fkayobe~master~I8a103b7c6d3dd1f0433f922ae90deae9b92c0c78","project":"openstack/kayobe","branch":"master","topic":"story/2006214","hashtags":[],"change_id":"I8a103b7c6d3dd1f0433f922ae90deae9b92c0c78","subject":"Fix inspector iPXE boot with UEFI compute nodes","status":"MERGED","created":"2019-07-12 11:49:03.000000000","updated":"2019-08-22 08:17:24.000000000","submitted":"2019-08-21 22:00:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670502-1566424856804-f2959db3","meta_rev_id":"8a85685fc9018b6e73e9009b7a71f69851d3543f","_number":670502,"virtual_id_number":670502,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-21 22:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-08-21 15:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-15 08:39:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-08-21 15:24:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-08-07 11:44:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 08:39:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-08-21 15:24:34.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-08-21 22:00:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71b5631b181f831ee33cd238d4b6e7b2a89b7cea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-12 11:49:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9370cd4ea10430ce14c8100728a078d54f867317","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-12 12:05:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d8461bd5e9851a12228aa6b39aa98522a187a12","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-12 12:06:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7cef41dff02331d7f2bc0273b29dbe46d3724de","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-12 13:42:52.000000000","message":"Patch Set 2: Workflow-1\n\nTesting...","accounts_in_message":[],"_revision_number":2},{"id":"0ddb9103e065ffa1338bbeeb2fc981f9c51142e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 13:55:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/02/670502/2/check/openstack-tox-docs/c0102b3/html/ : SUCCESS in 3m 35s\n- openstack-tox-cover http://logs.openstack.org/02/670502/2/check/openstack-tox-cover/ba1de62/cover/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/02/670502/2/check/openstack-tox-py35/6a82b1b/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/02/670502/2/check/openstack-tox-py36/b4888e3/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/02/670502/2/check/openstack-tox-pep8/f8720d4/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/02/670502/2/check/openstack-tox-py27/778d23f/ : SUCCESS in 4m 59s\n- build-openstack-releasenotes http://logs.openstack.org/02/670502/2/check/build-openstack-releasenotes/ed2939a/html/ : SUCCESS in 2m 49s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/02/670502/2/check/kayobe-tox-ansible-syntax/d6f6bb4/ : SUCCESS in 4m 03s\n- kayobe-tox-ansible http://logs.openstack.org/02/670502/2/check/kayobe-tox-ansible/849b5a0/ : SUCCESS in 9m 50s\n- kayobe-tox-molecule http://logs.openstack.org/02/670502/2/check/kayobe-tox-molecule/f5919a1/ : SUCCESS in 56m 55s\n- kayobe-overcloud-centos http://logs.openstack.org/02/670502/2/check/kayobe-overcloud-centos/ad60a47/ : SUCCESS in 1h 10m 04s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/02/670502/2/check/kayobe-overcloud-upgrade-centos/724ebeb/ : SUCCESS in 1h 41m 52s\n- kayobe-seed-centos http://logs.openstack.org/02/670502/2/check/kayobe-seed-centos/9a3be84/ : SUCCESS in 24m 40s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/02/670502/2/check/kayobe-seed-upgrade-centos/a458d8a/ : SUCCESS in 26m 07s","accounts_in_message":[],"_revision_number":2},{"id":"dac343eb062e5189c28aa2f7322116e925973601","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-12 14:33:38.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e3d0f84d29d83b8eefa325fbebf113e30f060c65","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-12 14:34:56.000000000","message":"Patch Set 3: Workflow-1\n\nTurns out kolla-ansible also had the wrong value for initrd: https://bugs.launchpad.net/kolla-ansible/+bug/1836375.","accounts_in_message":[],"_revision_number":3},{"id":"d8752d5086a2d3e27114088ca8030914cceaa4ed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-12 15:05:56.000000000","message":"Patch Set 3: -Workflow\n\nTested and works when used with  https://review.opendev.org/670566","accounts_in_message":[],"_revision_number":3},{"id":"fef12007e2162078bd890b2b058337ba506509a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 16:22:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/02/670502/3/check/openstack-tox-docs/10c65f5/html/ : SUCCESS in 3m 46s\n- openstack-tox-cover http://logs.openstack.org/02/670502/3/check/openstack-tox-cover/92af343/cover/ : SUCCESS in 5m 41s\n- openstack-tox-py35 http://logs.openstack.org/02/670502/3/check/openstack-tox-py35/02efe81/ : SUCCESS in 7m 07s\n- openstack-tox-py36 http://logs.openstack.org/02/670502/3/check/openstack-tox-py36/4105cad/ : SUCCESS in 3m 39s\n- openstack-tox-pep8 http://logs.openstack.org/02/670502/3/check/openstack-tox-pep8/1cca3e8/ : SUCCESS in 6m 09s\n- openstack-tox-py27 http://logs.openstack.org/02/670502/3/check/openstack-tox-py27/bd71ce3/ : SUCCESS in 6m 05s\n- build-openstack-releasenotes http://logs.openstack.org/02/670502/3/check/build-openstack-releasenotes/db148dd/html/ : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/02/670502/3/check/kayobe-tox-ansible-syntax/5732a37/ : SUCCESS in 4m 11s\n- kayobe-tox-ansible http://logs.openstack.org/02/670502/3/check/kayobe-tox-ansible/20a9885/ : SUCCESS in 8m 14s\n- kayobe-tox-molecule http://logs.openstack.org/02/670502/3/check/kayobe-tox-molecule/a652905/ : TIMED_OUT in 1h 00m 49s\n- kayobe-overcloud-centos http://logs.openstack.org/02/670502/3/check/kayobe-overcloud-centos/561acb9/ : SUCCESS in 1h 05m 45s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/02/670502/3/check/kayobe-overcloud-upgrade-centos/267792f/ : SUCCESS in 1h 43m 01s\n- kayobe-seed-centos http://logs.openstack.org/02/670502/3/check/kayobe-seed-centos/730be30/ : SUCCESS in 27m 06s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/02/670502/3/check/kayobe-seed-upgrade-centos/4495c4f/ : SUCCESS in 28m 37s","accounts_in_message":[],"_revision_number":3},{"id":"6b26c5297ec8fb8aa2edaa79089b2e5979122a75","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-12 16:33:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"68ebdf6057ba84a19884dbd1932a65eab76ba7f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 19:13:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/02/670502/3/check/openstack-tox-docs/ef13abe/html/ : SUCCESS in 5m 11s\n- openstack-tox-cover http://logs.openstack.org/02/670502/3/check/openstack-tox-cover/8469ef8/cover/ : SUCCESS in 5m 17s\n- openstack-tox-py35 http://logs.openstack.org/02/670502/3/check/openstack-tox-py35/454b671/ : SUCCESS in 3m 53s\n- openstack-tox-py36 http://logs.openstack.org/02/670502/3/check/openstack-tox-py36/6bee67e/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/02/670502/3/check/openstack-tox-pep8/10fad69/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/02/670502/3/check/openstack-tox-py27/b51b7d9/ : SUCCESS in 5m 13s\n- build-openstack-releasenotes http://logs.openstack.org/02/670502/3/check/build-openstack-releasenotes/6e2e5ca/html/ : SUCCESS in 4m 53s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/02/670502/3/check/kayobe-tox-ansible-syntax/8aae807/ : SUCCESS in 7m 14s\n- kayobe-tox-ansible http://logs.openstack.org/02/670502/3/check/kayobe-tox-ansible/0428e89/ : SUCCESS in 8m 14s\n- kayobe-tox-molecule http://logs.openstack.org/02/670502/3/check/kayobe-tox-molecule/5cc9727/ : SUCCESS in 56m 14s\n- kayobe-overcloud-centos http://logs.openstack.org/02/670502/3/check/kayobe-overcloud-centos/3393ea5/ : SUCCESS in 58m 29s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/02/670502/3/check/kayobe-overcloud-upgrade-centos/f1905fb/ : TIMED_OUT in 2h 31m 45s\n- kayobe-seed-centos http://logs.openstack.org/02/670502/3/check/kayobe-seed-centos/9ea5ab9/ : SUCCESS in 24m 29s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/02/670502/3/check/kayobe-seed-upgrade-centos/0ea52aa/ : SUCCESS in 27m 00s","accounts_in_message":[],"_revision_number":3},{"id":"c4b974db0f9068682f2ac98f751651648310063f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-12 21:09:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b4b2c079090103bbea6fa6cb003af7fc80e76fbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 22:56:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/02/670502/3/check/openstack-tox-docs/5929f52/html/ : SUCCESS in 5m 23s\n- openstack-tox-cover http://logs.openstack.org/02/670502/3/check/openstack-tox-cover/028324f/cover/ : SUCCESS in 4m 27s\n- openstack-tox-py35 http://logs.openstack.org/02/670502/3/check/openstack-tox-py35/493319f/ : SUCCESS in 6m 15s\n- openstack-tox-py36 http://logs.openstack.org/02/670502/3/check/openstack-tox-py36/d9cc829/ : SUCCESS in 6m 26s\n- openstack-tox-pep8 http://logs.openstack.org/02/670502/3/check/openstack-tox-pep8/e458673/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/02/670502/3/check/openstack-tox-py27/036390d/ : SUCCESS in 3m 11s\n- build-openstack-releasenotes http://logs.openstack.org/02/670502/3/check/build-openstack-releasenotes/1f37ea7/html/ : SUCCESS in 2m 28s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/02/670502/3/check/kayobe-tox-ansible-syntax/c7bc73c/ : SUCCESS in 3m 43s\n- kayobe-tox-ansible http://logs.openstack.org/02/670502/3/check/kayobe-tox-ansible/7674fa5/ : SUCCESS in 14m 04s\n- kayobe-tox-molecule http://logs.openstack.org/02/670502/3/check/kayobe-tox-molecule/24bb260/ : SUCCESS in 1h 00m 53s\n- kayobe-overcloud-centos http://logs.openstack.org/02/670502/3/check/kayobe-overcloud-centos/3ddd55c/ : SUCCESS in 1h 45m 25s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/02/670502/3/check/kayobe-overcloud-upgrade-centos/0c90981/ : FAILURE in 3m 18s\n- kayobe-seed-centos http://logs.openstack.org/02/670502/3/check/kayobe-seed-centos/2c3153f/ : SUCCESS in 24m 25s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/02/670502/3/check/kayobe-seed-upgrade-centos/097a535/ : SUCCESS in 28m 22s","accounts_in_message":[],"_revision_number":3},{"id":"961ae637091f774110ef337858b13e45a3000a10","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-15 08:22:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"71d4f13b40a4f1443116e7741723df526cde827b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-15 08:31:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bb9dd89f5cf0bf1f81c13c552068f9349d5399d6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-15 08:31:50.000000000","message":"Patch Set 5:\n\nRebased onto https://review.opendev.org/#/c/645849 since they were conflicting.","accounts_in_message":[],"_revision_number":5},{"id":"7b5d095193151bd06f6903fe984bfdd894393203","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 10:34:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/02/670502/5/check/openstack-tox-docs/a165455/html/ : SUCCESS in 2m 46s\n- openstack-tox-cover http://logs.openstack.org/02/670502/5/check/openstack-tox-cover/82f96c4/cover/ : SUCCESS in 3m 12s\n- openstack-tox-py35 http://logs.openstack.org/02/670502/5/check/openstack-tox-py35/1f0426a/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/02/670502/5/check/openstack-tox-py36/e4714ce/ : SUCCESS in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/02/670502/5/check/openstack-tox-pep8/9bc179b/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/02/670502/5/check/openstack-tox-py27/eab4076/ : SUCCESS in 3m 53s\n- build-openstack-releasenotes http://logs.openstack.org/02/670502/5/check/build-openstack-releasenotes/1441ce4/html/ : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/02/670502/5/check/kayobe-tox-ansible-syntax/0a0c4ba/ : SUCCESS in 6m 32s\n- kayobe-tox-ansible http://logs.openstack.org/02/670502/5/check/kayobe-tox-ansible/5736c2a/ : SUCCESS in 9m 26s\n- kayobe-tox-molecule http://logs.openstack.org/02/670502/5/check/kayobe-tox-molecule/8de7b78/ : SUCCESS in 53m 13s\n- kayobe-overcloud-centos http://logs.openstack.org/02/670502/5/check/kayobe-overcloud-centos/43928c4/ : SUCCESS in 1h 07m 00s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/02/670502/5/check/kayobe-overcloud-upgrade-centos/4fb2701/ : SUCCESS in 2h 00m 45s\n- kayobe-seed-centos http://logs.openstack.org/02/670502/5/check/kayobe-seed-centos/6f8b27e/ : SUCCESS in 19m 38s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/02/670502/5/check/kayobe-seed-upgrade-centos/fb82e00/ : SUCCESS in 24m 38s","accounts_in_message":[],"_revision_number":5},{"id":"ec99a2d273fbcf204a77323b4672d8de2c5b2d55","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-07 11:44:39.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"1f1aeb5a97e5070b9037dc54d73bbd66d0f95104","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-15 08:39:11.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"10bed3b0d72abdf03492238b456b2df49592a524","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-08-21 15:24:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"246a706c88545694e14ba12dc4d6c061532affbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 15:24:47.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3d97ffddbde502001b47dd15d7471097a598cbe0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 17:35:02.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/87226765e57e4aa7a6f2fbcd08ccefc4 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd32e7f90a674aaebe4991b9a548daf8 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91989f15e4e5458287d3c1cac27c1822 : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef0e2d8c6d5b406d80fc3afdb3e4db8a : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ae5a6f1c54f42c09eae232c7b5616d4 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5be1618dca0d41ab9c6ac9d2a3b46061 : SUCCESS in 4m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3d739f4b30544b99a518116c05fe7071 : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9d4394b9107e4e7a933ac18748da557e : SUCCESS in 8m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7eab13c944804d6f8ab355a9001647f8 : SUCCESS in 44m 35s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8ab7f2a1bbca442ea33eb8d869cff8af : TIMED_OUT in 2h 02m 08s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/29661acc40fd4b6eac87f7a45a0424b3 : SUCCESS in 1h 53m 57s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b9fb55a257cb41afaffa9e8b03675b6d : SUCCESS in 19m 47s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/301a6d16c6b049ebabdb900d5161d650 : SUCCESS in 27m 11s","accounts_in_message":[],"_revision_number":5},{"id":"ba6a7838e87ec1dfdc40d59e16abf280765c6975","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-21 18:08:41.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"34fa435ed6af7965aec7eddc3e6b403e0427dac7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 20:09:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20cdf02f98f748d1a6569dfe078d0f96 : SUCCESS in 5m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6a478f5bef7648bfa0ee607d66d4be19 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be168da47ee74e66935e6ce4607b1f42 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/263393bf6e2542edb71cd0744c7295ff : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7468713aee844df39fa025fff55f5ff4 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd89048c918a414fa1acfc098a722b6e : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d965fded8f09477baa6ce60132ea3665 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/68994a48a40d46a2843b458b717e2ea4 : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/18f8b24fe68e469884942d04ff6d2e19 : SUCCESS in 9m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa13cc1852b447338ee667aa7736a54c : SUCCESS in 41m 03s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d12cbccfe31846cfbcd002263d9ee19e : SUCCESS in 1h 00m 43s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b0975c4f5b7d457fa3cbbda6fdd206a9 : SUCCESS in 1h 49m 32s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/7f044e349ef44daca482a6730fd0bdea : SUCCESS in 18m 52s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b97cab1aa34a46eab38bd853050b33d3 : SUCCESS in 29m 15s","accounts_in_message":[],"_revision_number":5},{"id":"859dde322afb924c157f6940f311dfdc3ba366b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 20:09:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0868736bc99695562dc5148062a88f6503aefb21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 22:00:56.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9fa40b90695747f08c21343bdafcc73a : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f95ad711c8b4d398e60de7428ddb549 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0f51229103b47608423aa01a4a23fb6 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9726379d6414449d83b8034d2aa7ba9e : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0168e82adc5948ec996ad6575b722446 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1cef14335c94456b909fdf9dc0fca4a : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f33f071a3afd4dc497fa53d5a4d9e3c6 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cdc028d264fc40e9bdd7a3c17e8ba33e : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/60f08badfe514e58a1db774c12f3de55 : SUCCESS in 43m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/0e7da43d46244e569d82256f1cf6f617 : SUCCESS in 1h 07m 25s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3d67139d769c4313a94cfa59e491607d : SUCCESS in 1h 47m 23s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b7851c4e2f1d4695acd26eff207fdd7a : SUCCESS in 21m 14s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f7660a18d8b14132bf26e6d872b888e4 : SUCCESS in 29m 28s","accounts_in_message":[],"_revision_number":5},{"id":"2802994bb2a772bcc000d0fbe9ec6aa7ac6125f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 22:00:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"671ed03ef318d6445fd1572c054c728d21593b15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 22:02:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37e364e399414a9fa0b534cdf4201c61 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":5},{"id":"a0a0d8d6bed76238c5addb88d2bff9468557e3ae","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-22 08:17:24.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 1d7b72dbac101998774b4b880f7bc8f1d7b7bbfc","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0ee55e19fa9889b807e72f7f018800a1761d92a0","revisions":{"3c01c087c2c28eb0a468d4f64d71bf30bb1160cb":{"kind":"REWORK","_number":1,"created":"2019-07-12 11:49:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/02/670502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/670502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/670502/1"}}},"commit":{"parents":[{"commit":"31cb4ff37e0db1e2836d749aa4356fcf707dcbd2","subject":"Merge \"Improve diagnostics for seed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31cb4ff37e0db1e2836d749aa4356fcf707dcbd2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-12 11:44:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-12 11:48:36.000000000","tz":60},"subject":"Fix inspector iPXE boot with UEFI compute nodes","message":"Fix inspector iPXE boot with UEFI compute nodes\n\nThe IPA ramdisk fails to boot on UEFI compute nodes with iPXE. We see\nthe following error on the console, followed by a kernel panic:\n\nFix kernel panic-not syncing: VFS: unable to mount root fs on unknown\nblock(0,0)\n\nThe cause is the lack of an initrd\u003dironic-agent.initramfs kernel\nargument in inspector.ipxe.\n\nThe initrd argument is present in the kolla-ansible version of\ninspector.ipxe, and now contains all parameters that we need (since\nRocky), so this change simply removes our custom template, and passes\nthrough the extra kernel argument option configuration value.\n\nStory: 2006214\nTask: 35804\n\nChange-Id: I8a103b7c6d3dd1f0433f922ae90deae9b92c0c78\nStory: 2006213\nTask: 35803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c01c087c2c28eb0a468d4f64d71bf30bb1160cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c01c087c2c28eb0a468d4f64d71bf30bb1160cb"}]},"branch":"refs/heads/master"},"6dc70e705ea3626429488027c4351f8dc382cea3":{"kind":"REWORK","_number":2,"created":"2019-07-12 12:06:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/02/670502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/670502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/670502/2"}}},"commit":{"parents":[{"commit":"31cb4ff37e0db1e2836d749aa4356fcf707dcbd2","subject":"Merge \"Improve diagnostics for seed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31cb4ff37e0db1e2836d749aa4356fcf707dcbd2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-12 11:44:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-12 12:06:00.000000000","tz":60},"subject":"Fix inspector iPXE boot with UEFI compute nodes","message":"Fix inspector iPXE boot with UEFI compute nodes\n\nThe IPA ramdisk fails to boot on UEFI compute nodes with iPXE. We see\nthe following error on the console, followed by a kernel panic:\n\nFix kernel panic-not syncing: VFS: unable to mount root fs on unknown\nblock(0,0)\n\nThe cause is the lack of an initrd\u003dironic-agent.initramfs kernel\nargument in inspector.ipxe.\n\nThe initrd argument is present in the kolla-ansible version of\ninspector.ipxe, and now contains all parameters that we need (since\nRocky), so this change simply removes our custom template, and passes\nthrough the extra kernel argument option configuration value.\n\nStory: 2006214\nTask: 35804\n\nChange-Id: I8a103b7c6d3dd1f0433f922ae90deae9b92c0c78\nStory: 2006213\nTask: 35803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6dc70e705ea3626429488027c4351f8dc382cea3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6dc70e705ea3626429488027c4351f8dc382cea3"}]},"branch":"refs/heads/master"},"257d54b26cfac10cc61bb15652a5ffca9a8cea4b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-12 14:33:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/02/670502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/670502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/670502/3"}}},"commit":{"parents":[{"commit":"31cb4ff37e0db1e2836d749aa4356fcf707dcbd2","subject":"Merge \"Improve diagnostics for seed jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31cb4ff37e0db1e2836d749aa4356fcf707dcbd2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-12 11:44:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-12 14:33:35.000000000","tz":0},"subject":"Fix inspector iPXE boot with UEFI compute nodes","message":"Fix inspector iPXE boot with UEFI compute nodes\n\nThe IPA ramdisk fails to boot on UEFI compute nodes with iPXE. We see\nthe following error on the console, followed by a kernel panic:\n\nFix kernel panic-not syncing: VFS: unable to mount root fs on unknown\nblock(0,0)\n\nThe cause is the lack of an initrd\u003dironic-agent.initramfs kernel\nargument in inspector.ipxe.\n\nThe initrd argument is present in the kolla-ansible version of\ninspector.ipxe, and now contains all parameters that we need (since\nRocky), so this change simply removes our custom template, and passes\nthrough the extra kernel argument option configuration value.\n\nStory: 2006214\nTask: 35804\n\nStory: 2006213\nTask: 35803\n\nDepends-On: https://review.opendev.org/670566\nChange-Id: I8a103b7c6d3dd1f0433f922ae90deae9b92c0c78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/257d54b26cfac10cc61bb15652a5ffca9a8cea4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/257d54b26cfac10cc61bb15652a5ffca9a8cea4b"}]},"branch":"refs/heads/master"},"8b7e8e449cd9880b5b0151ebfa0dd9eb2c2a951f":{"kind":"REWORK","_number":4,"created":"2019-07-15 08:22:34.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/02/670502/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/670502/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/670502/4"}}},"commit":{"parents":[{"commit":"abc179162340dab49c54b81dc48986d0740b8660","subject":"Merge \"Change examples to apply all security-related package updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abc179162340dab49c54b81dc48986d0740b8660"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-12 11:44:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-15 08:22:26.000000000","tz":60},"subject":"Fix inspector iPXE boot with UEFI compute nodes","message":"Fix inspector iPXE boot with UEFI compute nodes\n\nThe IPA ramdisk fails to boot on UEFI compute nodes with iPXE. We see\nthe following error on the console, followed by a kernel panic:\n\nFix kernel panic-not syncing: VFS: unable to mount root fs on unknown\nblock(0,0)\n\nThe cause is the lack of an initrd\u003dironic-agent.initramfs kernel\nargument in inspector.ipxe.\n\nThe initrd argument is present in the kolla-ansible version of\ninspector.ipxe, and now contains all parameters that we need (since\nRocky), so this change simply removes our custom template, and passes\nthrough the extra kernel argument option configuration value.\n\nStory: 2006214\nTask: 35804\n\nStory: 2006213\nTask: 35803\n\nDepends-On: https://review.opendev.org/670566\nChange-Id: I8a103b7c6d3dd1f0433f922ae90deae9b92c0c78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b7e8e449cd9880b5b0151ebfa0dd9eb2c2a951f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b7e8e449cd9880b5b0151ebfa0dd9eb2c2a951f"}]},"branch":"refs/heads/master"},"0ee55e19fa9889b807e72f7f018800a1761d92a0":{"kind":"REWORK","_number":5,"created":"2019-07-15 08:31:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/02/670502/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/670502/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/670502/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/670502/5"}}},"commit":{"parents":[{"commit":"650919f52736b297de50bc48bb7fc71e19cb64a9","subject":"Disable chrony container by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/650919f52736b297de50bc48bb7fc71e19cb64a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-12 11:44:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-15 08:31:03.000000000","tz":60},"subject":"Fix inspector iPXE boot with UEFI compute nodes","message":"Fix inspector iPXE boot with UEFI compute nodes\n\nThe IPA ramdisk fails to boot on UEFI compute nodes with iPXE. We see\nthe following error on the console, followed by a kernel panic:\n\nFix kernel panic-not syncing: VFS: unable to mount root fs on unknown\nblock(0,0)\n\nThe cause is the lack of an initrd\u003dironic-agent.initramfs kernel\nargument in inspector.ipxe.\n\nThe initrd argument is present in the kolla-ansible version of\ninspector.ipxe, and now contains all parameters that we need (since\nRocky), so this change simply removes our custom template, and passes\nthrough the extra kernel argument option configuration value.\n\nStory: 2006214\nTask: 35804\n\nStory: 2006213\nTask: 35803\n\nDepends-On: https://review.opendev.org/670566\nChange-Id: I8a103b7c6d3dd1f0433f922ae90deae9b92c0c78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ee55e19fa9889b807e72f7f018800a1761d92a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ee55e19fa9889b807e72f7f018800a1761d92a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
