)]}'
{"id":"openstack%2Fkayobe~670884","triplet_id":"openstack%2Fkayobe~master~I3a570adf43e1addce5eeab88e29ae4ded44669f0","project":"openstack/kayobe","branch":"master","topic":"add-arista","hashtags":[],"change_id":"I3a570adf43e1addce5eeab88e29ae4ded44669f0","subject":"Add support for Arista switches","status":"MERGED","created":"2019-07-15 16:53:12.000000000","updated":"2019-08-28 08:23:21.000000000","submitted":"2019-08-27 20:07:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670884-1566936443038-80306961","meta_rev_id":"9b20567e61534946392c969c12c4f1027cedd5ba","_number":670884,"virtual_id_number":670884,"owner":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-23 11:07:33.000000000","_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-27 20:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-08-22 14:29:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-22 14:27:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-08-22 14:29:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 14:27:54.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-08-22 14:29:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-23 11:07:33.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-08-27 20:07:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5d0ca1e3078cf8985f26b079c49b35930eb86c4","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2019-07-15 16:53:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f11505050782ce52c1701264f2762a587b732008","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 18:30:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/84/670884/1/check/openstack-tox-docs/b3dea19/html/ : SUCCESS in 3m 25s\n- openstack-tox-cover http://logs.openstack.org/84/670884/1/check/openstack-tox-cover/cda0456/cover/ : SUCCESS in 4m 43s\n- openstack-tox-py35 http://logs.openstack.org/84/670884/1/check/openstack-tox-py35/0ab9461/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/84/670884/1/check/openstack-tox-py36/775a398/ : SUCCESS in 3m 30s\n- openstack-tox-pep8 http://logs.openstack.org/84/670884/1/check/openstack-tox-pep8/6ed1bb6/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/84/670884/1/check/openstack-tox-py27/d3b438a/ : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/84/670884/1/check/kayobe-tox-ansible-syntax/c3d83d7/ : SUCCESS in 4m 59s\n- kayobe-tox-ansible http://logs.openstack.org/84/670884/1/check/kayobe-tox-ansible/0993aad/ : SUCCESS in 10m 31s\n- kayobe-tox-molecule http://logs.openstack.org/84/670884/1/check/kayobe-tox-molecule/1802270/ : SUCCESS in 43m 49s\n- kayobe-overcloud-centos http://logs.openstack.org/84/670884/1/check/kayobe-overcloud-centos/c6fd8dd/ : SUCCESS in 1h 06m 52s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/84/670884/1/check/kayobe-overcloud-upgrade-centos/bdd3759/ : SUCCESS in 1h 30m 00s\n- kayobe-seed-centos http://logs.openstack.org/84/670884/1/check/kayobe-seed-centos/9169615/ : SUCCESS in 17m 46s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/84/670884/1/check/kayobe-seed-upgrade-centos/be5c869/ : SUCCESS in 24m 06s","accounts_in_message":[],"_revision_number":1},{"id":"878e0e186993624468030d723fa6fb5efb504910","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-16 14:12:41.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"aa4d06b671902f351aae4cb53bdf4e25f1a79dfa","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2019-07-25 11:16:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d882549ad2372b208d3d79d8d3f567c0551dbe73","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2019-07-25 11:18:00.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThanks Mark.  Added a configurable timeout with a longer default than the module default (apparently 10s to complete the entire transaction)","accounts_in_message":[],"_revision_number":1},{"id":"215fae6673679a4564565dabad1e8437587f6818","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 13:17:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/84/670884/2/check/openstack-tox-docs/b87f3fd/html/ : SUCCESS in 3m 21s\n- openstack-tox-cover http://logs.openstack.org/84/670884/2/check/openstack-tox-cover/f9b6e43/cover/ : SUCCESS in 6m 30s\n- openstack-tox-py35 http://logs.openstack.org/84/670884/2/check/openstack-tox-py35/fdbf11c/ : SUCCESS in 6m 00s\n- openstack-tox-py36 http://logs.openstack.org/84/670884/2/check/openstack-tox-py36/3609534/ : SUCCESS in 5m 53s\n- openstack-tox-pep8 http://logs.openstack.org/84/670884/2/check/openstack-tox-pep8/3f1d206/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/84/670884/2/check/openstack-tox-py27/4ecd88c/ : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/84/670884/2/check/kayobe-tox-ansible-syntax/96641f5/ : SUCCESS in 4m 56s\n- kayobe-tox-ansible http://logs.openstack.org/84/670884/2/check/kayobe-tox-ansible/b3c3678/ : SUCCESS in 11m 37s\n- kayobe-tox-molecule http://logs.openstack.org/84/670884/2/check/kayobe-tox-molecule/541b555/ : SUCCESS in 45m 21s\n- kayobe-overcloud-centos http://logs.openstack.org/84/670884/2/check/kayobe-overcloud-centos/9fc2bd4/ : SUCCESS in 1h 03m 32s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/84/670884/2/check/kayobe-overcloud-upgrade-centos/a2747a2/ : SUCCESS in 1h 43m 25s\n- kayobe-seed-centos http://logs.openstack.org/84/670884/2/check/kayobe-seed-centos/2bafb3c/ : SUCCESS in 21m 04s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/84/670884/2/check/kayobe-seed-upgrade-centos/c1d3fed/ : SUCCESS in 27m 46s","accounts_in_message":[],"_revision_number":2},{"id":"087e2f21fc2d5882a79dffa96a3a7c39d19e501a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-02 10:58:27.000000000","message":"Patch Set 2: Code-Review+1\n\nCould use a release note, otherwise looks good to go.","accounts_in_message":[],"_revision_number":2},{"id":"fdb9be8bb02f23852f165d6429625509e8d84c32","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-22 14:27:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"66a8c0405ccc8d822a4ae6df13ffc4d527f38061","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-22 14:27:54.000000000","message":"Patch Set 3: Code-Review+2\n\nI added a release note.","accounts_in_message":[],"_revision_number":3},{"id":"94a90fe634ad3c6bd24d2fea971f24aeb542df11","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-22 14:29:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9bc15bfc6c3dc8561d0c907248976c592204bbdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 16:32:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af0845f1a9ca480e90c689e0db1c6c97 : SUCCESS in 4m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/39f080624f0b4c749188beeb93ea4470 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c8ac06fbc1741cf9ca40c1dc5eebc07 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/305d3f3263504506863bee8d950a676b : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee5fb88ff3604717abb2d1a745b6fa28 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f46174b3e5e34bccac9b29621c88ff1a : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccd00073e97545db89b1f03b11945f11 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e58fa9e99fb845538468e597fe7bd3f0 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d190cf3aadc5484988573cf2eef1101e : SUCCESS in 9m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/86921b78b32c48bd806718d33a229fc8 : SUCCESS in 45m 40s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d7e598b0bb264d1f8c3487c87a657a65 : SUCCESS in 1h 28m 11s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/bfa63bcbecc54e8189b346cbc37fd38e : SUCCESS in 1h 36m 57s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/da321c2eee884574bbdb9001b2e506d7 : FAILURE in 10m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3ac58202c5c44422bef24069f765c826 : SUCCESS in 25m 24s","accounts_in_message":[],"_revision_number":3},{"id":"db1ba8b7e5a4e0dcaa5e5d5bd813a94c0046ee9a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-08-23 11:07:33.000000000","message":"Patch Set 3:\n\nrecheck: https://review.opendev.org/#/c/678082 now landed","accounts_in_message":[],"_revision_number":3},{"id":"9476081fe90f7eb91ed6dab7caf0ad53101fd023","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 13:06:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab8b4207d0bc48bdbf0d51cec31280fd : SUCCESS in 7m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0b6ccf447ebc44c39e37eb385e5165de : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e39d7b1072d460cbfcdf4935bd21bc4 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c261467827545aca36b46e88e35027c : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c113f0b9a68f4f8c981607b6a1ebc51a : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0b4f61de9164524aa68010037f92565 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e921006c81a04e2e928dcef0a39247a1 : SUCCESS in 6m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d1c742f5fe70427881f594e791d45c60 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a4589b4298754e13ab3cc0eaf370362d : SUCCESS in 9m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afa98a8816b44fdaa2cbefddb8a4beec : SUCCESS in 44m 35s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/27e3f2d467ec4e32b2a6bd313c6935c5 : SUCCESS in 1h 08m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/7a88b14c32884f52a78ba4fc503b611f : SUCCESS in 1h 53m 25s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8bdcfeb36fa1485fbd377b70a6fb9d72 : SUCCESS in 25m 13s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/11b4d0e18dfb461e9085a61da25b0d9c : SUCCESS in 29m 05s","accounts_in_message":[],"_revision_number":3},{"id":"d67b181f5a4ba419c531d2cdee2e294eb1c798c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 13:06:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2ccaeb644d53002a81e01fa9fec51f7ddd38305e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 15:11:54.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5ea24766e72740e6871a4f4c49483a9e : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0437618039e4472593a27f23a3e3d32c : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6d591376b6641be8bb36047ef03883d : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f278be3d2ad64c55975976fb3d087696 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b70293a26cf46d7a5b00618f69be85b : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc653f7ff35f437fbe7d1d030c52543f : SUCCESS in 4m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/415c0cbcd8c74e84863b07a3aace54b8 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a04ade0289324fbc94d0d67a8600c6c5 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad87c46648664962a49029814d5336f6 : SUCCESS in 46m 14s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1c3bd9bb68e04bc09d698674cdd345e2 : FAILURE in 1h 59m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/e674d4f42ee24690841b0c15782f16be : SUCCESS in 2h 04m 21s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/dc00db2c3a684d09b8e551860aa7593c : SUCCESS in 19m 37s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8162696942b5449380ffc42e08b25be3 : SUCCESS in 36m 38s","accounts_in_message":[],"_revision_number":3},{"id":"e2374db9190d165a7df1860626949552f1a25857","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-27 10:39:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"119bb1574ee7515d61b1e888517ccec73048c9f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 12:30:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abd81242f2d14f42b0c33306ee2bad76 : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e902081cdb1d4e1cab90d23fce58b2a4 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1987ef18e8ff458d912710ed1f1969b4 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e43fdcedf1f1448bbbe73492e348eba6 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d8d6884759a4ff28d01877bfa5c70a2 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ca497667134bedb3808601b9e33057 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba1dc8a8e1334ca0ada40e4771a98266 : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb763f6a76d44150924589507b273352 : SUCCESS in 7m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/17ec6aeb757b46d5ac993683ddf8678c : SUCCESS in 9m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1f8f857a1db49e5a92f692f6e64bd1a : SUCCESS in 45m 31s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/47fd4f47cd7241f485c0edaa4ffa7983 : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1bca8e6c99224a999a61666d1d25f9c3 : SUCCESS in 1h 42m 10s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9c99a8fc78834e1a9281c496f43caa57 : SUCCESS in 24m 57s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/335fa606025d41878b3bc25f888169b5 : SUCCESS in 29m 03s","accounts_in_message":[],"_revision_number":3},{"id":"ad96e7597f460e32e07ca40421c6635ba6163183","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 12:30:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4ac1ed6ee4190348331391ff8c6f568a558d0bac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 15:22:17.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4faf25ec9a92457f8e914958fe4215b3 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71e051e8a9c94089b7abd614961639a7 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee7979ac69594ed8a28017c47c20fea4 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fdf4c97b0da14841ac2d072993f28f5b : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e63b52552a204171934f7ea951339edc : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecd3b178fe724d87bc528efd6b8b9014 : SUCCESS in 5m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/852db9ec89a049979d676811342c9b72 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aaa21205f6af461d9c65196502f56215 : SUCCESS in 9m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9a592e1272c44721abec17379ce77e9d : SUCCESS in 50m 13s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/66e5fbe6ffa44913a7d4db2b93454d55 : TIMED_OUT in 2h 02m 38s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/150714c20e144493a841976f3221953c : SUCCESS in 2h 30m 12s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/24335db216d54f779d7f92f811f317ae : SUCCESS in 23m 40s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e208030e12134d9bab14778171c62184 : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":3},{"id":"99043e20cf9233123700d7914f60ef8d0b63554e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-27 16:07:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0fc74d4fb17c3e18b1b7b9b4332f426fc21af023","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 18:09:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c853a7b286d14b2580d811cb5e0ac7e7 : SUCCESS in 8m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1b78f85e410c409f93b4cd4c15fd1ce3 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a80bfd21cd264c5691319bbaceb6bd3f : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2738b9ebef34480fba5fc3a98f015b8d : SUCCESS in 6m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8fff21d3d984cbdaa8c8e0901df5751 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be33b56019a4354bdc16b769624a89a : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75fe927770974b44b1c9146d2c560235 : SUCCESS in 6m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0405782c95a34e708fe76fef7ae263d2 : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/acf16dcbd5ba4406ab2b12afd7deb276 : SUCCESS in 12m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3664c18300684e85b0067f4b46bf5034 : SUCCESS in 44m 34s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1c7ae318def145b0aea552477e84439d : SUCCESS in 1h 06m 21s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4394148874d440c0b3bf7d0cb6ad1bdd : SUCCESS in 1h 55m 10s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/35cdca17c5224ab29decc09a68139a1a : SUCCESS in 22m 08s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/6f46fcc54ba647c491d71e9cefac0a67 : SUCCESS in 28m 17s","accounts_in_message":[],"_revision_number":3},{"id":"f9d0a846adf2b6b6d3f24dcb648c4700f62dfcee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 18:09:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8e0800889f6822f6520cea673a7031fbbbf4017d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 20:07:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/77a8a2bb2d824334bce2fa53d0f4e396 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5972c01748a14a97b2be6324db925e5a : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/167e20fff628410dae20245ea8d3f47b : SUCCESS in 6m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e21e09eb6cdc46339a1b9f79fb18aae2 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0dc3c8de0d4c61b73e68dcb1c7ba50 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ced6ffe3ff0f47a3b5fd318730ca33b6 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/326761eda1d74a49b5578466e2db3dd2 : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/25aaa33e8a624253917e077b48f6bd0f : SUCCESS in 10m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f84c2fdcc8b54b9fa0a12746a7d90afc : SUCCESS in 48m 39s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e3d7d439937342498c6c1b6bd3658232 : SUCCESS in 1h 03m 20s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/69dcfbd846404991bb21dd893415826c : SUCCESS in 1h 46m 19s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e1f9bddefedb4693b12be9ca834b78ca : SUCCESS in 29m 16s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/342c0dd6cf274792a43931cc093bffc5 : SUCCESS in 26m 11s","accounts_in_message":[],"_revision_number":3},{"id":"c302cc178075db3ef1ce8a0e99393716530f69ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 20:07:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0603d9cfd3d304fc7e41a9d81779829ea3d00e9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 20:09:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4d9344b99740dab63f6c8fd25704e0 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":3},{"id":"eb41c0a6dad7ea59be908e95344dddd75d2bde9b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-28 08:23:21.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 0da672abfd591af17f6e8f4d9f59952ae6e635f9","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c6263dbfd4dfbac5f53a4fdd94b154811fbd0af6","revisions":{"11c378fd278ae84eca454a5e8f0cd1b1c93656df":{"kind":"REWORK","_number":1,"created":"2019-07-15 16:53:12.000000000","uploader":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"ref":"refs/changes/84/670884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/670884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/670884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/670884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/670884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/670884/1"}}},"commit":{"parents":[{"commit":"abc179162340dab49c54b81dc48986d0740b8660","subject":"Merge \"Change examples to apply all security-related package updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abc179162340dab49c54b81dc48986d0740b8660"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2019-07-15 16:48:31.000000000","tz":60},"committer":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2019-07-15 16:48:31.000000000","tz":60},"subject":"Add support for Arista switches","message":"Add support for Arista switches\n\nSupport for invoking Ansible EOS network modules for Arista switches.\n\nChange-Id: I3a570adf43e1addce5eeab88e29ae4ded44669f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11c378fd278ae84eca454a5e8f0cd1b1c93656df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11c378fd278ae84eca454a5e8f0cd1b1c93656df"}]},"branch":"refs/heads/master"},"0509e99b73f9742cced5b932e666407f129e5481":{"kind":"REWORK","_number":2,"created":"2019-07-25 11:16:46.000000000","uploader":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"ref":"refs/changes/84/670884/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/670884/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/670884/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/670884/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/670884/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/670884/2"}}},"commit":{"parents":[{"commit":"abc179162340dab49c54b81dc48986d0740b8660","subject":"Merge \"Change examples to apply all security-related package updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abc179162340dab49c54b81dc48986d0740b8660"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2019-07-15 16:48:31.000000000","tz":60},"committer":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2019-07-25 11:16:23.000000000","tz":60},"subject":"Add support for Arista switches","message":"Add support for Arista switches\n\nSupport for invoking Ansible EOS network modules for Arista switches.\n\nChange-Id: I3a570adf43e1addce5eeab88e29ae4ded44669f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0509e99b73f9742cced5b932e666407f129e5481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0509e99b73f9742cced5b932e666407f129e5481"}]},"branch":"refs/heads/master"},"c6263dbfd4dfbac5f53a4fdd94b154811fbd0af6":{"kind":"REWORK","_number":3,"created":"2019-08-22 14:27:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/670884/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/670884/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/670884/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/670884/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/670884/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/670884/3"}}},"commit":{"parents":[{"commit":"45b4961de190224c91f66bcac874dc1dc5f183f4","subject":"Merge \"Update release documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45b4961de190224c91f66bcac874dc1dc5f183f4"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2019-07-15 16:48:31.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-22 14:20:40.000000000","tz":120},"subject":"Add support for Arista switches","message":"Add support for Arista switches\n\nSupport for invoking Ansible EOS network modules for Arista switches.\n\nChange-Id: I3a570adf43e1addce5eeab88e29ae4ded44669f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6263dbfd4dfbac5f53a4fdd94b154811fbd0af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6263dbfd4dfbac5f53a4fdd94b154811fbd0af6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
