)]}'
{"id":"openstack%2Fkayobe~671073","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I5a2c71e7e65f99687437f6a26a0c4a942022f4e9","project":"openstack/kayobe","branch":"stable/rocky","topic":"uuid-token-provider-stable/rocky","hashtags":[],"change_id":"I5a2c71e7e65f99687437f6a26a0c4a942022f4e9","subject":"Remove mentions of UUID Keystone token provider","status":"MERGED","created":"2019-07-16 15:02:10.000000000","updated":"2019-07-17 18:47:19.000000000","submitted":"2019-07-17 18:47:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671073-1563389239962-9977374c","meta_rev_id":"48be76e146a70ed944dc315ff3ad54fc34842b59","_number":671073,"virtual_id_number":671073,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-17 18:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-07-17 16:50:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-07-16 15:15:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-07-17 16:50:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 15:15:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-17 16:50:36.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-17 18:47:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c87797d7c87f1e4e9f0c23e85ce58e71a96ce32b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-16 15:02:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebd0d34a3a48a57244dc172fe12c8a4efce3fcd7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-16 15:15:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2189880fc05c84dab3b00f021c25820f9f18fe2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 16:49:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/73/671073/1/check/build-openstack-sphinx-docs/4ce8c93/html/ : SUCCESS in 3m 25s\n- openstack-tox-docs http://logs.openstack.org/73/671073/1/check/openstack-tox-docs/d757cc5/html/ : SUCCESS in 2m 57s\n- openstack-tox-cover http://logs.openstack.org/73/671073/1/check/openstack-tox-cover/aba35ac/cover/ : SUCCESS in 3m 32s\n- openstack-tox-py35 http://logs.openstack.org/73/671073/1/check/openstack-tox-py35/252c505/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/73/671073/1/check/openstack-tox-pep8/8b30ccb/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/73/671073/1/check/openstack-tox-py27/ddc19fb/ : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/73/671073/1/check/kayobe-tox-ansible-syntax/775e6ea/ : SUCCESS in 6m 29s\n- kayobe-tox-ansible http://logs.openstack.org/73/671073/1/check/kayobe-tox-ansible/3297c34/ : SUCCESS in 7m 12s\n- kayobe-tox-molecule http://logs.openstack.org/73/671073/1/check/kayobe-tox-molecule/74992c3/ : SUCCESS in 53m 09s\n- kayobe-overcloud-centos http://logs.openstack.org/73/671073/1/check/kayobe-overcloud-centos/f2f85b4/ : SUCCESS in 57m 39s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/73/671073/1/check/kayobe-overcloud-upgrade-centos/347b1f5/ : SUCCESS in 1h 40m 51s\n- kayobe-seed-centos http://logs.openstack.org/73/671073/1/check/kayobe-seed-centos/45f57f1/ : SUCCESS in 17m 10s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/73/671073/1/check/kayobe-seed-upgrade-centos/6f137b3/ : SUCCESS in 29m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c839a827a4f1126e48327398b8812c90de2f39cd","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-17 16:50:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac1f484a07c740e0d16e7ff0100714edd9711dfc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 16:50:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5eaf424e066645b9cbb6492076da9995b7f1d8f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:47:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/73/671073/1/gate/build-openstack-sphinx-docs/e686a9a/html/ : SUCCESS in 5m 45s\n- openstack-tox-docs http://logs.openstack.org/73/671073/1/gate/openstack-tox-docs/a1620eb/html/ : SUCCESS in 6m 03s\n- openstack-tox-py35 http://logs.openstack.org/73/671073/1/gate/openstack-tox-py35/f29b974/ : SUCCESS in 5m 46s\n- openstack-tox-pep8 http://logs.openstack.org/73/671073/1/gate/openstack-tox-pep8/facc6d6/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/73/671073/1/gate/openstack-tox-py27/25215b0/ : SUCCESS in 5m 31s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/73/671073/1/gate/kayobe-tox-ansible-syntax/b72e870/ : SUCCESS in 6m 35s\n- kayobe-tox-ansible http://logs.openstack.org/73/671073/1/gate/kayobe-tox-ansible/faf9ec7/ : SUCCESS in 7m 32s\n- kayobe-tox-molecule http://logs.openstack.org/73/671073/1/gate/kayobe-tox-molecule/7c311ed/ : SUCCESS in 32m 06s\n- kayobe-overcloud-centos http://logs.openstack.org/73/671073/1/gate/kayobe-overcloud-centos/c35485d/ : SUCCESS in 57m 07s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/73/671073/1/gate/kayobe-overcloud-upgrade-centos/38463f0/ : SUCCESS in 1h 28m 02s\n- kayobe-seed-centos http://logs.openstack.org/73/671073/1/gate/kayobe-seed-centos/c9267ed/ : SUCCESS in 28m 13s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/73/671073/1/gate/kayobe-seed-upgrade-centos/78f05ac/ : SUCCESS in 26m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d48e8ca478f9f5e9f5e449f03714e294b6b652e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:47:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48c0ea8cac400098ade25d2d981e3ee83e5c29bd","revisions":{"48c0ea8cac400098ade25d2d981e3ee83e5c29bd":{"kind":"REWORK","_number":1,"created":"2019-07-16 15:02:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/671073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/671073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/671073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/671073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/671073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/671073/1"}}},"commit":{"parents":[{"commit":"58b0001dad25618686eb9e4a47a39d2bc13b67d4","subject":"Merge \"Use BatchMode to check whether host is accessible via SSH\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/58b0001dad25618686eb9e4a47a39d2bc13b67d4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-12 12:22:44.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-16 14:59:43.000000000","tz":120},"subject":"Remove mentions of UUID Keystone token provider","message":"Remove mentions of UUID Keystone token provider\n\nKeystone removed the UUID token provider in Rocky and Kolla Ansible\nremoved support for configuring it. This commit reflects these changes\nin Kayobe.\n\nChange-Id: I5a2c71e7e65f99687437f6a26a0c4a942022f4e9\n(cherry picked from commit 5d82c788a8a3af943abf4001ab393bc490d9ad3a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48c0ea8cac400098ade25d2d981e3ee83e5c29bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48c0ea8cac400098ade25d2d981e3ee83e5c29bd"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
