)]}'
{"id":"openstack%2Fkayobe~671332","triplet_id":"openstack%2Fkayobe~stable%2Frocky~Id090348a6f22f251555fa286a7d24923087122a9","project":"openstack/kayobe","branch":"stable/rocky","hashtags":[],"change_id":"Id090348a6f22f251555fa286a7d24923087122a9","subject":"Provide keystone.conf to kolla-ansible by default","status":"MERGED","created":"2019-07-17 16:57:22.000000000","updated":"2019-07-31 18:17:40.000000000","submitted":"2019-07-31 18:17:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671332-1564597060712-b98f6d83","meta_rev_id":"a79a153f78d0d2d6e15b45eae9bb0dfb649d54d0","_number":671332,"virtual_id_number":671332,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-07-31 18:17:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-22 09:39:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-07-31 16:32:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 09:39:54.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-07-31 18:17:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b55916091f5d52733c9ff53cbdf9355d632997a8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-17 16:57:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab271072a71120f2a5fd62b317b43529caec8316","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 18:38:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/32/671332/1/check/build-openstack-sphinx-docs/0e59794/html/ : SUCCESS in 4m 58s\n- openstack-tox-docs http://logs.openstack.org/32/671332/1/check/openstack-tox-docs/046f48b/html/ : SUCCESS in 2m 53s\n- openstack-tox-cover http://logs.openstack.org/32/671332/1/check/openstack-tox-cover/ff8bc20/cover/ : SUCCESS in 4m 00s\n- openstack-tox-py35 http://logs.openstack.org/32/671332/1/check/openstack-tox-py35/a4f1a3c/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/32/671332/1/check/openstack-tox-pep8/43e2ccc/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/32/671332/1/check/openstack-tox-py27/60186c9/ : SUCCESS in 5m 29s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/32/671332/1/check/kayobe-tox-ansible-syntax/892e2b4/ : SUCCESS in 5m 11s\n- kayobe-tox-ansible http://logs.openstack.org/32/671332/1/check/kayobe-tox-ansible/baa65b3/ : SUCCESS in 8m 29s\n- kayobe-tox-molecule http://logs.openstack.org/32/671332/1/check/kayobe-tox-molecule/66b90f0/ : SUCCESS in 33m 29s\n- kayobe-overcloud-centos http://logs.openstack.org/32/671332/1/check/kayobe-overcloud-centos/add9f69/ : SUCCESS in 58m 38s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/32/671332/1/check/kayobe-overcloud-upgrade-centos/cda9689/ : SUCCESS in 1h 36m 17s\n- kayobe-seed-centos http://logs.openstack.org/32/671332/1/check/kayobe-seed-centos/bf084e5/ : SUCCESS in 15m 50s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/32/671332/1/check/kayobe-seed-upgrade-centos/c5d6a1f/ : SUCCESS in 27m 37s","accounts_in_message":[],"_revision_number":1},{"id":"89b6fe2261ad50430d855a5ffc1a5dbfaea01c02","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-22 09:39:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd01096959b92ce643db16a0f7756a7da2b7a717","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-31 16:32:45.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"025108309f8fd9ea47a7bfb763809ec7e54923b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 16:33:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ca957c5508354660c7014d119b544639cc80242","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 18:17:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a79a153f78d0d2d6e15b45eae9bb0dfb649d54d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 18:17:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/32/671332/1/gate/build-openstack-sphinx-docs/8087e7a/html/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/32/671332/1/gate/openstack-tox-docs/5ac1058/html/ : SUCCESS in 3m 58s\n- openstack-tox-py35 http://logs.openstack.org/32/671332/1/gate/openstack-tox-py35/b9f61d4/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 http://logs.openstack.org/32/671332/1/gate/openstack-tox-pep8/d9a219c/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/32/671332/1/gate/openstack-tox-py27/04ac160/ : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/32/671332/1/gate/kayobe-tox-ansible-syntax/3d19afd/ : SUCCESS in 5m 21s\n- kayobe-tox-ansible http://logs.openstack.org/32/671332/1/gate/kayobe-tox-ansible/7a0d5cc/ : SUCCESS in 7m 40s\n- kayobe-tox-molecule http://logs.openstack.org/32/671332/1/gate/kayobe-tox-molecule/9584a8e/ : SUCCESS in 37m 53s\n- kayobe-overcloud-centos http://logs.openstack.org/32/671332/1/gate/kayobe-overcloud-centos/0332f5c/ : SUCCESS in 1h 01m 04s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/32/671332/1/gate/kayobe-overcloud-upgrade-centos/bf7b6c7/ : SUCCESS in 1h 35m 37s\n- kayobe-seed-centos http://logs.openstack.org/32/671332/1/gate/kayobe-seed-centos/c6ea6dd/ : SUCCESS in 16m 10s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/32/671332/1/gate/kayobe-seed-upgrade-centos/24fda6c/ : SUCCESS in 26m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2bf18f1873125a0ea8ec95417a13862ac495939","revisions":{"d2bf18f1873125a0ea8ec95417a13862ac495939":{"kind":"REWORK","_number":1,"created":"2019-07-17 16:57:22.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/32/671332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/671332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/671332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/671332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/671332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/671332/1"}}},"commit":{"parents":[{"commit":"0a133818016fbc11eed26554817b6e65f358a756","subject":"Multiple updates for Kayobe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a133818016fbc11eed26554817b6e65f358a756"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-06-21 14:55:46.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-07-17 16:56:34.000000000","tz":60},"subject":"Provide keystone.conf to kolla-ansible by default","message":"Provide keystone.conf to kolla-ansible by default\n\nCommit d8ee0fa027140fb8a7504bd6685b6dcdfe3ac8fe introduced the variable\nkolla_enable_keystone without setting a default value and used it to\ncheck whether to template and provide keystone.conf to kolla-ansible,\nunlike the extended keystone configuration which was always copied. This\nresulted in non-intuitive behaviour if kolla_enable_keystone wasn\u0027t\nexplicitly set by the operator.\n\nThis commit sets kolla_enable_keystone to true by default, which matches\nthe default in kolla-ansible, and uses it instead of hardcoded true\nvalues.\n\nConflicts:\n\tansible/roles/kolla-openstack/vars/main.yml\n\nChange-Id: Id090348a6f22f251555fa286a7d24923087122a9\nStory: 2005930\nTask: 34273\n(cherry picked from commit ec9b4826515f28dab43846cb80dcb58b9a5b0a6e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2bf18f1873125a0ea8ec95417a13862ac495939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2bf18f1873125a0ea8ec95417a13862ac495939"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
