)]}'
{"id":"openstack%2Fkayobe~672317","triplet_id":"openstack%2Fkayobe~master~Id6e14286b5eb2b767a515e7edfc56741fb8d2c78","project":"openstack/kayobe","branch":"master","topic":"ip-allocation","hashtags":[],"change_id":"Id6e14286b5eb2b767a515e7edfc56741fb8d2c78","subject":"Stop allocating network and broadcast addresses","status":"MERGED","created":"2019-07-23 15:22:53.000000000","updated":"2019-08-15 08:40:44.000000000","submitted":"2019-08-14 19:12:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672317-1565809966044-ff059232","meta_rev_id":"41282ada31dbf59e6664261e4d6e32a661a67777","_number":672317,"virtual_id_number":672317,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-14 19:12:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-08-14 17:13:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-08-03 16:25:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-08-14 17:13:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-08-07 11:42:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 11:42:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-14 17:13:43.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-08-14 19:12:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eccb3c6919937f2e5bb6e3fd291cc19c97fbbb7b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-23 15:22:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"748f61755b48eaff018a5c1544671e854ac99436","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-23 15:53:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fa34121c4f100e5bd20d44ada4e691c9fbe81892","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:44:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/17/672317/2/check/openstack-tox-docs/193b4ad/html/ : SUCCESS in 4m 13s\n- openstack-tox-cover http://logs.openstack.org/17/672317/2/check/openstack-tox-cover/2937ee5/cover/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/17/672317/2/check/openstack-tox-py35/13eb7cb/ : SUCCESS in 7m 29s\n- openstack-tox-py36 http://logs.openstack.org/17/672317/2/check/openstack-tox-py36/cc68cfe/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/17/672317/2/check/openstack-tox-pep8/574ca1b/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/17/672317/2/check/openstack-tox-py27/e9ace5a/ : SUCCESS in 4m 56s\n- build-openstack-releasenotes http://logs.openstack.org/17/672317/2/check/build-openstack-releasenotes/8a8c713/html/ : SUCCESS in 4m 24s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/17/672317/2/check/kayobe-tox-ansible-syntax/d102b28/ : SUCCESS in 5m 47s\n- kayobe-tox-ansible http://logs.openstack.org/17/672317/2/check/kayobe-tox-ansible/964efa3/ : SUCCESS in 9m 32s\n- kayobe-tox-molecule http://logs.openstack.org/17/672317/2/check/kayobe-tox-molecule/e35245f/ : SUCCESS in 44m 28s\n- kayobe-overcloud-centos http://logs.openstack.org/17/672317/2/check/kayobe-overcloud-centos/e0076f8/ : SUCCESS in 1h 05m 08s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/17/672317/2/check/kayobe-overcloud-upgrade-centos/f2b7ecf/ : SUCCESS in 1h 46m 59s\n- kayobe-seed-centos http://logs.openstack.org/17/672317/2/check/kayobe-seed-centos/aaf63eb/ : SUCCESS in 25m 45s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/17/672317/2/check/kayobe-seed-upgrade-centos/143a3c7/ : SUCCESS in 29m 10s","accounts_in_message":[],"_revision_number":2},{"id":"5469e339f42e285c91e756fa5216e4703fc1a2e3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-02 10:35:13.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nLGTM, just a doc nit.","accounts_in_message":[],"_revision_number":2},{"id":"50a6dfc193fcfc464725357d7107308c369183a3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-02 12:07:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"85257395c7399b0006406196a927557c02c00ba2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-02 12:08:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ef82428f9d9c105b49f4412c145650d39f10365e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-02 12:08:41.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"d2fcf3c9e5747d613e44109c086b2825ffec3ffc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 14:21:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/17/672317/4/check/openstack-tox-docs/b9e48ca/html/ : SUCCESS in 4m 06s\n- openstack-tox-cover https://logs.opendev.org/17/672317/4/check/openstack-tox-cover/e80615e/cover/ : SUCCESS in 5m 07s\n- openstack-tox-py35 https://logs.opendev.org/17/672317/4/check/openstack-tox-py35/de98fcf/ : SUCCESS in 5m 26s\n- openstack-tox-py36 https://logs.opendev.org/17/672317/4/check/openstack-tox-py36/b554e8b/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://logs.opendev.org/17/672317/4/check/openstack-tox-pep8/e2a027a/ : SUCCESS in 4m 51s\n- openstack-tox-py27 https://logs.opendev.org/17/672317/4/check/openstack-tox-py27/dacced0/ : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://logs.opendev.org/17/672317/4/check/build-openstack-releasenotes/5450184/html/ : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://logs.opendev.org/17/672317/4/check/kayobe-tox-ansible-syntax/594954c/ : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://logs.opendev.org/17/672317/4/check/kayobe-tox-ansible/9c007c9/ : SUCCESS in 9m 28s\n- kayobe-tox-molecule https://logs.opendev.org/17/672317/4/check/kayobe-tox-molecule/68812d8/ : SUCCESS in 35m 42s\n- kayobe-overcloud-centos https://logs.opendev.org/17/672317/4/check/kayobe-overcloud-centos/129cdfb/ : SUCCESS in 58m 28s\n- kayobe-overcloud-upgrade-centos https://logs.opendev.org/17/672317/4/check/kayobe-overcloud-upgrade-centos/1048cfb/ : SUCCESS in 1h 53m 20s\n- kayobe-seed-centos https://logs.opendev.org/17/672317/4/check/kayobe-seed-centos/5eb919d/ : SUCCESS in 25m 12s\n- kayobe-seed-upgrade-centos https://logs.opendev.org/17/672317/4/check/kayobe-seed-upgrade-centos/ccb3aad/ : SUCCESS in 27m 05s","accounts_in_message":[],"_revision_number":4},{"id":"e220c4fb8d90acadf0d281ef929ef2ea00aae1af","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-03 16:25:51.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":4},{"id":"2d9a02c9799651130f4b08ade59c4bfddbbe0bb9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-07 11:42:37.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"8b3a39c7c14a57cc6457501238c5615f78cfe71b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-14 17:13:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"90b2252ef9706cdd9641a1602afd53caa6734dea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 17:13:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b67239e04496ef1d061cd46818d736d3699a98eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 19:12:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/aa35a614fa9a4ae2b359ddf58c0be02f : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cf7365ad68c43bf87186176141fbc74 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5d89668802b42e29a4e2fe2215e3560 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbf9374923674be3b973c3bcf5cc9126 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a572816f51c64ec9be57af2401ae7562 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3c031bfb223404a9fd2f0272c765914 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/54254b975110480998eb885599f630bf : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3615af615b6e4387b357c6e491a7b88d : SUCCESS in 10m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/17b7666f1fb84d92808b0fa6e5723de4 : SUCCESS in 45m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7e1d6c4063854b559211d9ad2ba6b1cd : SUCCESS in 1h 39m 34s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/34e2acb5e8824c8fa0a5244a9b666baa : SUCCESS in 1h 45m 21s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4aac79c4359344f6a613770839903be6 : SUCCESS in 21m 49s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8779de92f34a4bf2af121d1d42dabf59 : SUCCESS in 29m 31s","accounts_in_message":[],"_revision_number":4},{"id":"f7f25a78338b0f40467bc5ff8e90c62279d6abdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 19:12:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a228a9d7f784323d3e2a6ec4a3a6fda87e2c72c0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 08:40:32.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 77914efe1d8cd58637a4c37b68ff2ae311f85d2f","accounts_in_message":[],"_revision_number":4},{"id":"0cbe04b8ae9cc5f044cb229b7dbc55b3d1943312","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 08:40:44.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 559408e85c8bafd3a3b1f9bb600ae3837fe840db","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"697e74e41acc7129bb6b0937012209136cd4e01a","revisions":{"1377c4178e732adeb2821d186359e59264bc9acb":{"kind":"REWORK","_number":1,"created":"2019-07-23 15:22:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/17/672317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/17/672317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/672317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/672317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/672317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/17/672317/1"}}},"commit":{"parents":[{"commit":"473df9b81187a6174eb4b6bee2e5ecbd66d68908","subject":"Merge \"Updating Jinja filters to conform to Ansible 2.5+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/473df9b81187a6174eb4b6bee2e5ecbd66d68908"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-23 15:22:22.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-23 15:22:22.000000000","tz":120},"subject":"Stop allocating network and broadcast addresses","message":"Stop allocating network and broadcast addresses\n\nWhen an allocation pool range is not defined by the operator, we should\nnot include network and broadcast addresses in the list of IP addresses\nto give to hosts.\n\nChange-Id: Id6e14286b5eb2b767a515e7edfc56741fb8d2c78\nStory: 2006267\nTask: 35958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1377c4178e732adeb2821d186359e59264bc9acb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1377c4178e732adeb2821d186359e59264bc9acb"}]},"branch":"refs/heads/master"},"a5a6a7cbb8bc61636cb69d7653b8f4a41b5f1b93":{"kind":"REWORK","_number":2,"created":"2019-07-23 15:53:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/17/672317/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/17/672317/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/672317/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/672317/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/672317/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/17/672317/2"}}},"commit":{"parents":[{"commit":"57d7d276a3add887ce5004475cf4dff0c4dfc828","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57d7d276a3add887ce5004475cf4dff0c4dfc828"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-23 15:53:05.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-23 15:53:16.000000000","tz":120},"subject":"Stop allocating network and broadcast addresses","message":"Stop allocating network and broadcast addresses\n\nWhen an allocation pool range is not defined by the operator, we should\nnot include network and broadcast addresses in the list of IP addresses\nto give to hosts.\n\nChange-Id: Id6e14286b5eb2b767a515e7edfc56741fb8d2c78\nStory: 2006267\nTask: 35958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5a6a7cbb8bc61636cb69d7653b8f4a41b5f1b93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5a6a7cbb8bc61636cb69d7653b8f4a41b5f1b93"}]},"branch":"refs/heads/master"},"4e069bde9abc997a6583e9c4a8f43f19cba21cec":{"kind":"REWORK","_number":3,"created":"2019-08-02 12:07:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/17/672317/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/17/672317/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/672317/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/672317/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/672317/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/17/672317/3"}}},"commit":{"parents":[{"commit":"d1b51adb3d9638a393277b7100406a52d421fbfe","subject":"Merge \"Ignore --limit for localhost plays\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d1b51adb3d9638a393277b7100406a52d421fbfe"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-02 12:07:15.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-02 12:07:15.000000000","tz":120},"subject":"Stop allocating network and broadcast addresses","message":"Stop allocating network and broadcast addresses\n\nWhen an allocation pool range is not defined by the operator, we should\nnot include network and broadcast addresses in the list of IP addresses\nto give to hosts.\n\nChange-Id: Id6e14286b5eb2b767a515e7edfc56741fb8d2c78\nStory: 2006267\nTask: 35958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e069bde9abc997a6583e9c4a8f43f19cba21cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e069bde9abc997a6583e9c4a8f43f19cba21cec"}]},"branch":"refs/heads/master"},"697e74e41acc7129bb6b0937012209136cd4e01a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-02 12:08:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/17/672317/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/17/672317/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/672317/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/672317/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/17/672317/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/17/672317/4"}}},"commit":{"parents":[{"commit":"2bce627040f6ccb659abf93c58a0cbee42c04477","subject":"Merge \"Bump version of stackhpc.dell-powerconnect-switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2bce627040f6ccb659abf93c58a0cbee42c04477"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-02 12:08:34.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-02 12:08:34.000000000","tz":120},"subject":"Stop allocating network and broadcast addresses","message":"Stop allocating network and broadcast addresses\n\nWhen an allocation pool range is not defined by the operator, we should\nnot include network and broadcast addresses in the list of IP addresses\nto give to hosts.\n\nChange-Id: Id6e14286b5eb2b767a515e7edfc56741fb8d2c78\nStory: 2006267\nTask: 35958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/697e74e41acc7129bb6b0937012209136cd4e01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/697e74e41acc7129bb6b0937012209136cd4e01a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
