)]}'
{"id":"openstack%2Fkayobe~672482","triplet_id":"openstack%2Fkayobe~master~I6fa7640658fd112aaddbe08c4af01711fb1f7916","project":"openstack/kayobe","branch":"master","topic":"storage-mgmt-net-controllers","hashtags":[],"change_id":"I6fa7640658fd112aaddbe08c4af01711fb1f7916","subject":"Remove storage management network from controllers","status":"MERGED","created":"2019-07-24 09:37:02.000000000","updated":"2019-08-14 19:15:34.000000000","submitted":"2019-08-14 19:15:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672482-1565810134691-070cffcc","meta_rev_id":"7cc42103f68ed486310146109666a6f0fcfe40fc","_number":672482,"virtual_id_number":672482,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-14 19:15:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-06 13:14:27.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-08-14 17:14:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-08-06 15:35:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-08-14 17:14:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-08-07 11:41:59.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 11:41:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-14 17:14:35.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-08-14 19:15:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0fefd0d2ff027a2814763fe8deb944d4acf09f9b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-24 09:37:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cccbb3329f8b2d330cfdb88fb474c75785509ab","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-24 09:37:44.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"73cf84aefb037c2ef2cc0ee3eb1a0aea1155aae7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-24 09:38:10.000000000","message":"Patch Set 2:\n\nAdding DNM tag since I would like to have Mark\u0027s opinion before merging.","accounts_in_message":[],"_revision_number":2},{"id":"3c08d2825f371aa3d73baff5f8a33fabb5617206","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 10:40:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/82/672482/2/check/openstack-tox-docs/6b98c49/html/ : SUCCESS in 3m 14s\n- openstack-tox-cover http://logs.openstack.org/82/672482/2/check/openstack-tox-cover/f4d1c75/cover/ : SUCCESS in 4m 15s\n- openstack-tox-py35 http://logs.openstack.org/82/672482/2/check/openstack-tox-py35/4ab6208/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/82/672482/2/check/openstack-tox-py36/a172d12/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/82/672482/2/check/openstack-tox-pep8/4c89d17/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/82/672482/2/check/openstack-tox-py27/e80146f/ : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/82/672482/2/check/kayobe-tox-ansible-syntax/4f6d911/ : SUCCESS in 5m 20s\n- kayobe-tox-ansible http://logs.openstack.org/82/672482/2/check/kayobe-tox-ansible/4691d48/ : SUCCESS in 7m 40s\n- kayobe-tox-molecule http://logs.openstack.org/82/672482/2/check/kayobe-tox-molecule/2a390b4/ : SUCCESS in 44m 32s\n- kayobe-overcloud-centos http://logs.openstack.org/82/672482/2/check/kayobe-overcloud-centos/db74e55/ : FAILURE in 48m 53s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/82/672482/2/check/kayobe-overcloud-upgrade-centos/b7adf9f/ : FAILURE in 49m 35s\n- kayobe-seed-centos http://logs.openstack.org/82/672482/2/check/kayobe-seed-centos/379b3b1/ : SUCCESS in 25m 14s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/82/672482/2/check/kayobe-seed-upgrade-centos/4a54ff2/ : SUCCESS in 26m 37s","accounts_in_message":[],"_revision_number":2},{"id":"57741aff2e524a23f9a379ee46bfa31d21691e14","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-24 12:16:39.000000000","message":"Patch Set 2:\n\npyparsing issue fixed by https://review.opendev.org/#/c/672395/\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0f7728b5609bcf2020b94a1f51e8b8598ed7ddaa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-24 14:42:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cf599b0aa3c1b275008e827ceb6199a97170830d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 16:20:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/82/672482/2/check/openstack-tox-docs/fd4e692/html/ : SUCCESS in 4m 16s\n- openstack-tox-cover http://logs.openstack.org/82/672482/2/check/openstack-tox-cover/996fda8/cover/ : SUCCESS in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/82/672482/2/check/openstack-tox-py35/e9350dd/ : SUCCESS in 7m 13s\n- openstack-tox-py36 http://logs.openstack.org/82/672482/2/check/openstack-tox-py36/1bf4634/ : SUCCESS in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/82/672482/2/check/openstack-tox-pep8/9b55daf/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/82/672482/2/check/openstack-tox-py27/22fdbd8/ : SUCCESS in 4m 42s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/82/672482/2/check/kayobe-tox-ansible-syntax/977ec86/ : SUCCESS in 4m 19s\n- kayobe-tox-ansible http://logs.openstack.org/82/672482/2/check/kayobe-tox-ansible/768053b/ : SUCCESS in 10m 13s\n- kayobe-tox-molecule http://logs.openstack.org/82/672482/2/check/kayobe-tox-molecule/3b084ae/ : SUCCESS in 51m 19s\n- kayobe-overcloud-centos http://logs.openstack.org/82/672482/2/check/kayobe-overcloud-centos/4a39c55/ : SUCCESS in 1h 00m 55s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/82/672482/2/check/kayobe-overcloud-upgrade-centos/c847d65/ : FAILURE in 1h 33m 46s\n- kayobe-seed-centos http://logs.openstack.org/82/672482/2/check/kayobe-seed-centos/b9a7d0d/ : SUCCESS in 20m 00s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/82/672482/2/check/kayobe-seed-upgrade-centos/4e45b75/ : SUCCESS in 28m 57s","accounts_in_message":[],"_revision_number":2},{"id":"fcd684976e3ccb302d43a32fd429594b6d68517e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-24 16:44:04.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8f197179c7a708dcfab66ef484d6dca9a770cfda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 18:26:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/82/672482/2/check/openstack-tox-docs/e03cde0/html/ : SUCCESS in 3m 06s\n- openstack-tox-cover http://logs.openstack.org/82/672482/2/check/openstack-tox-cover/7c4d291/cover/ : SUCCESS in 7m 04s\n- openstack-tox-py35 http://logs.openstack.org/82/672482/2/check/openstack-tox-py35/3d8c425/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/82/672482/2/check/openstack-tox-py36/1305ce3/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/82/672482/2/check/openstack-tox-pep8/527c64b/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/82/672482/2/check/openstack-tox-py27/ca51c1d/ : SUCCESS in 4m 45s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/82/672482/2/check/kayobe-tox-ansible-syntax/1b9c471/ : SUCCESS in 4m 13s\n- kayobe-tox-ansible http://logs.openstack.org/82/672482/2/check/kayobe-tox-ansible/e953035/ : SUCCESS in 9m 24s\n- kayobe-tox-molecule http://logs.openstack.org/82/672482/2/check/kayobe-tox-molecule/df0d80d/ : SUCCESS in 45m 57s\n- kayobe-overcloud-centos http://logs.openstack.org/82/672482/2/check/kayobe-overcloud-centos/be04aa7/ : SUCCESS in 1h 06m 51s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/82/672482/2/check/kayobe-overcloud-upgrade-centos/a9583ea/ : SUCCESS in 1h 39m 20s\n- kayobe-seed-centos http://logs.openstack.org/82/672482/2/check/kayobe-seed-centos/385d630/ : SUCCESS in 23m 38s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/82/672482/2/check/kayobe-seed-upgrade-centos/ee82b7a/ : SUCCESS in 24m 08s","accounts_in_message":[],"_revision_number":2},{"id":"64e59862385ce2e9627d8db5e4cf8d7f7fe3bfd7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-02 10:32:34.000000000","message":"Patch Set 2: Code-Review+1\n\nMakes sense, but please add an upgrade release note.","accounts_in_message":[],"_revision_number":2},{"id":"83bf1a43474164ad2cd489ce6943f388ea03ff22","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-02 12:17:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"799d24c9bfbae090897758da632bc37bdfc2a608","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-02 12:17:26.000000000","message":"Patch Set 3:\n\nThanks Mark.","accounts_in_message":[],"_revision_number":3},{"id":"3786c2ed9e24c6b33ff7fb6de0766bd7880d49ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 14:36:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/82/672482/3/check/openstack-tox-docs/fc15a77/html/ : SUCCESS in 3m 48s\n- openstack-tox-cover https://logs.opendev.org/82/672482/3/check/openstack-tox-cover/82aa57f/cover/ : SUCCESS in 6m 03s\n- openstack-tox-py35 https://logs.opendev.org/82/672482/3/check/openstack-tox-py35/22b0b34/ : SUCCESS in 6m 45s\n- openstack-tox-py36 https://logs.opendev.org/82/672482/3/check/openstack-tox-py36/16af06c/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://logs.opendev.org/82/672482/3/check/openstack-tox-pep8/40aeda1/ : SUCCESS in 4m 04s\n- openstack-tox-py27 https://logs.opendev.org/82/672482/3/check/openstack-tox-py27/785a805/ : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://logs.opendev.org/82/672482/3/check/build-openstack-releasenotes/3d6b7b1/html/ : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax https://logs.opendev.org/82/672482/3/check/kayobe-tox-ansible-syntax/768644a/ : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://logs.opendev.org/82/672482/3/check/kayobe-tox-ansible/959af9b/ : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://logs.opendev.org/82/672482/3/check/kayobe-tox-molecule/5323993/ : SUCCESS in 44m 19s\n- kayobe-overcloud-centos https://logs.opendev.org/82/672482/3/check/kayobe-overcloud-centos/4570c91/ : SUCCESS in 1h 14m 52s\n- kayobe-overcloud-upgrade-centos https://logs.opendev.org/82/672482/3/check/kayobe-overcloud-upgrade-centos/0fa002b/ : SUCCESS in 1h 57m 15s\n- kayobe-seed-centos https://logs.opendev.org/82/672482/3/check/kayobe-seed-centos/5bbac81/ : SUCCESS in 19m 45s\n- kayobe-seed-upgrade-centos https://logs.opendev.org/82/672482/3/check/kayobe-seed-upgrade-centos/f6acfe2/ : SUCCESS in 37m 08s","accounts_in_message":[],"_revision_number":3},{"id":"5a0ea1f8c1a7e86ec0ad54bd03a69d6c6698b345","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-03 16:25:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e1f41a294bc6fe5af3c669e998b9470a0b34201f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-06 13:13:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"69a9ce07f64f4d8e5f6d4e4d6fbb47ba8c772877","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-06 13:14:27.000000000","message":"Patch Set 4:\n\nAlso removed related variables from controllers and seed group_vars","accounts_in_message":[],"_revision_number":4},{"id":"eb8ea6144ddba8c6df050ee2839fbb1a39276101","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 15:17:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/82/672482/4/check/openstack-tox-docs/d6ef055/html/ : SUCCESS in 3m 55s\n- openstack-tox-cover https://logs.opendev.org/82/672482/4/check/openstack-tox-cover/5828584/cover/ : SUCCESS in 4m 51s\n- openstack-tox-py35 https://logs.opendev.org/82/672482/4/check/openstack-tox-py35/076de96/ : SUCCESS in 6m 31s\n- openstack-tox-py36 https://logs.opendev.org/82/672482/4/check/openstack-tox-py36/40b42e6/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://logs.opendev.org/82/672482/4/check/openstack-tox-pep8/473c7c5/ : SUCCESS in 5m 15s\n- openstack-tox-py27 https://logs.opendev.org/82/672482/4/check/openstack-tox-py27/7ab7eac/ : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://logs.opendev.org/82/672482/4/check/build-openstack-releasenotes/9695e9e/html/ : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://logs.opendev.org/82/672482/4/check/kayobe-tox-ansible-syntax/ed7fd33/ : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://logs.opendev.org/82/672482/4/check/kayobe-tox-ansible/83e5d32/ : SUCCESS in 9m 51s\n- kayobe-tox-molecule https://logs.opendev.org/82/672482/4/check/kayobe-tox-molecule/96da523/ : SUCCESS in 45m 51s\n- kayobe-overcloud-centos https://logs.opendev.org/82/672482/4/check/kayobe-overcloud-centos/12a7186/ : SUCCESS in 1h 05m 56s\n- kayobe-overcloud-upgrade-centos https://logs.opendev.org/82/672482/4/check/kayobe-overcloud-upgrade-centos/e7405e2/ : SUCCESS in 1h 55m 49s\n- kayobe-seed-centos https://logs.opendev.org/82/672482/4/check/kayobe-seed-centos/564564d/ : SUCCESS in 23m 08s\n- kayobe-seed-upgrade-centos https://logs.opendev.org/82/672482/4/check/kayobe-seed-upgrade-centos/8fa43b2/ : SUCCESS in 30m 43s","accounts_in_message":[],"_revision_number":4},{"id":"c8443bcbde09c103481889a698c24443da5784f9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-06 15:35:47.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"38c2aba957a552c0baa43a37dedde7b057d8d017","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-07 11:41:59.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"c09bc8be030167293ab24de3df19f9275e1a613f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-14 17:14:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1aca80a3fc43a864b9c7961efde70b69221331c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 17:15:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"226df4ae3e7019a3c245bf656e3cc74a1e682143","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 19:15:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4c0dd50a9ae1414b8ec2eedd66cf7de2 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf40416d9b4d4bc8972b945f41a43eaf : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a6515fac3043e090295720702a02ba : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13b872639d174a53a2874bc0871bba34 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7897b277b08b4bc5aa4b1f779416d21b : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6793bbfc6014ca897671e0aa1639034 : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6a7d50d5b6b648dea65c19338fa30899 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/66397eabd3db438c984d8159029d9739 : SUCCESS in 9m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8175e2c98b29443b8bfa9301e1320667 : SUCCESS in 44m 48s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f6857c67217d4ca8b82afe61ef4ef10c : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/cfb24083ef564f8f919c6353cd7237a0 : SUCCESS in 1h 53m 52s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9c5f023e4cba4b39838c2ece19a4d482 : SUCCESS in 27m 31s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3a51b71587164233ab0504d379e2be0c : SUCCESS in 25m 27s","accounts_in_message":[],"_revision_number":4},{"id":"86d28b5cd289c44678e957e4e4c59916e9d55589","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 19:15:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f5490793fbe6ec971c511f5c3aa57dc3a9dec555","revisions":{"e8c86c8fb16b0e6fbc8fa502c4957746d4b1d23d":{"kind":"REWORK","_number":1,"created":"2019-07-24 09:37:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/672482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/672482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/672482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/672482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/672482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/672482/1"}}},"commit":{"parents":[{"commit":"57d7d276a3add887ce5004475cf4dff0c4dfc828","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57d7d276a3add887ce5004475cf4dff0c4dfc828"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-24 09:36:35.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-24 09:36:35.000000000","tz":120},"subject":"Remove storage management network from controllers","message":"Remove storage management network from controllers\n\nThere is no need to connect controllers nodes to the storage management\nnetwork by default. It should only be used by storage nodes.\n\nChange-Id: I6fa7640658fd112aaddbe08c4af01711fb1f7916\nStory: 2006273\nTask: 35973\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8c86c8fb16b0e6fbc8fa502c4957746d4b1d23d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8c86c8fb16b0e6fbc8fa502c4957746d4b1d23d"}]},"branch":"refs/heads/master"},"385a52822fc0456777f3b515010ce388cb77fd3e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-24 09:37:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/672482/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/672482/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/672482/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/672482/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/672482/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/672482/2"}}},"commit":{"parents":[{"commit":"57d7d276a3add887ce5004475cf4dff0c4dfc828","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57d7d276a3add887ce5004475cf4dff0c4dfc828"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-24 09:37:31.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-24 09:37:31.000000000","tz":120},"subject":"[DNM] Remove storage management network from controllers","message":"[DNM] Remove storage management network from controllers\n\nThere is no need to connect controllers nodes to the storage management\nnetwork by default. It should only be used by storage nodes.\n\nChange-Id: I6fa7640658fd112aaddbe08c4af01711fb1f7916\nStory: 2006273\nTask: 35973\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/385a52822fc0456777f3b515010ce388cb77fd3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/385a52822fc0456777f3b515010ce388cb77fd3e"}]},"branch":"refs/heads/master"},"6bc3435e7f9c001f005cb9929d955294f044b51a":{"kind":"REWORK","_number":3,"created":"2019-08-02 12:17:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/672482/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/672482/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/672482/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/672482/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/672482/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/672482/3"}}},"commit":{"parents":[{"commit":"2bce627040f6ccb659abf93c58a0cbee42c04477","subject":"Merge \"Bump version of stackhpc.dell-powerconnect-switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2bce627040f6ccb659abf93c58a0cbee42c04477"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-02 12:16:32.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-02 12:16:32.000000000","tz":120},"subject":"Remove storage management network from controllers","message":"Remove storage management network from controllers\n\nThere is no need to connect controllers nodes to the storage management\nnetwork by default. It should only be used by storage nodes.\n\nChange-Id: I6fa7640658fd112aaddbe08c4af01711fb1f7916\nStory: 2006273\nTask: 35973\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6bc3435e7f9c001f005cb9929d955294f044b51a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6bc3435e7f9c001f005cb9929d955294f044b51a"}]},"branch":"refs/heads/master"},"f5490793fbe6ec971c511f5c3aa57dc3a9dec555":{"kind":"REWORK","_number":4,"created":"2019-08-06 13:13:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/672482/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/672482/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/672482/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/672482/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/672482/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/672482/4"}}},"commit":{"parents":[{"commit":"2bce627040f6ccb659abf93c58a0cbee42c04477","subject":"Merge \"Bump version of stackhpc.dell-powerconnect-switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2bce627040f6ccb659abf93c58a0cbee42c04477"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-06 13:13:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-06 13:13:41.000000000","tz":60},"subject":"Remove storage management network from controllers","message":"Remove storage management network from controllers\n\nThere is no need to connect controllers nodes to the storage management\nnetwork by default. It should only be used by storage nodes.\n\nChange-Id: I6fa7640658fd112aaddbe08c4af01711fb1f7916\nStory: 2006273\nTask: 35973\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5490793fbe6ec971c511f5c3aa57dc3a9dec555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5490793fbe6ec971c511f5c3aa57dc3a9dec555"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
