)]}'
{"id":"openstack%2Fkayobe~673139","triplet_id":"openstack%2Fkayobe~master~I0c0589387da53458f41df280dff0179715448aaa","project":"openstack/kayobe","branch":"master","topic":"cloudkitty-conf","hashtags":[],"change_id":"I0c0589387da53458f41df280dff0179715448aaa","subject":"Add support for custom CloudKitty configuration","status":"MERGED","created":"2019-07-27 14:00:05.000000000","updated":"2019-08-14 19:12:41.000000000","submitted":"2019-08-14 19:12:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673139-1565809961928-0fc9d827","meta_rev_id":"feaba44bbe58888e8cbaeb343bb7d265d19dcbc1","_number":673139,"virtual_id_number":673139,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-14 19:12:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-08-14 17:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-08-02 10:22:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-08-14 17:12:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-08-07 11:43:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 11:43:32.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-14 17:12:04.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-08-14 19:12:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c3ff59c542253426a1c905810d9135dd9857d3f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-27 14:00:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d84ea25c08737a97cac9381df028b20fcca8d56","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-27 15:21:14.000000000","message":"Topic set to cloudkitty-conf","accounts_in_message":[],"_revision_number":1},{"id":"ad6c5ed211f404cb95546a52209b5119ff65cb94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 16:06:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/39/673139/1/check/openstack-tox-docs/93e2ce2/html/ : SUCCESS in 3m 26s\n- openstack-tox-cover http://logs.openstack.org/39/673139/1/check/openstack-tox-cover/a2fd2c7/cover/ : SUCCESS in 4m 29s\n- openstack-tox-py35 http://logs.openstack.org/39/673139/1/check/openstack-tox-py35/9268e94/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/39/673139/1/check/openstack-tox-py36/acdc2d2/ : SUCCESS in 3m 25s\n- openstack-tox-pep8 http://logs.openstack.org/39/673139/1/check/openstack-tox-pep8/735fe63/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/39/673139/1/check/openstack-tox-py27/e287d85/ : SUCCESS in 3m 24s\n- build-openstack-releasenotes http://logs.openstack.org/39/673139/1/check/build-openstack-releasenotes/d31056c/html/ : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax http://logs.openstack.org/39/673139/1/check/kayobe-tox-ansible-syntax/cae0fa4/ : SUCCESS in 4m 35s\n- kayobe-tox-ansible http://logs.openstack.org/39/673139/1/check/kayobe-tox-ansible/a8701af/ : SUCCESS in 7m 47s\n- kayobe-tox-molecule http://logs.openstack.org/39/673139/1/check/kayobe-tox-molecule/aa39a8b/ : SUCCESS in 43m 09s\n- kayobe-overcloud-centos http://logs.openstack.org/39/673139/1/check/kayobe-overcloud-centos/1be59bd/ : SUCCESS in 1h 01m 20s\n- kayobe-overcloud-upgrade-centos http://logs.openstack.org/39/673139/1/check/kayobe-overcloud-upgrade-centos/a0b3089/ : SUCCESS in 1h 44m 33s\n- kayobe-seed-centos http://logs.openstack.org/39/673139/1/check/kayobe-seed-centos/c125c9a/ : SUCCESS in 20m 24s\n- kayobe-seed-upgrade-centos http://logs.openstack.org/39/673139/1/check/kayobe-seed-upgrade-centos/fea9b90/ : SUCCESS in 25m 59s","accounts_in_message":[],"_revision_number":1},{"id":"df696db3ebe099ce792d150614f7ab485e21382f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-02 10:22:38.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"7badb3b467cfcb9c6cd45b16343d36c66deeefdf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-07 11:43:32.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"e48efcd0af0e471e22be12226f40867d72887d2d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-14 17:12:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e15c44806f88193013ea17daa4bb6fb20587ced","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 17:12:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42ba8d9b22188fd1596da54a923709b7bc2d87df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 19:12:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e3febd944a644605b7d6461f84320364 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffec8f962e5d498e8f720b5a26845673 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6c4ae6aea47413abc9391b5c8dda690 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/87b079b8bcd24ac5b9f5f183466a9874 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af71feede44943ddb38fb779c1fea08e : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2d96619a54b4d61806b56fb5145a154 : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a0fd3361552c44c4a76a85b87ce4b074 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1932af188f4c444e945a11c11a84e7e0 : SUCCESS in 9m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef40219ca251413dacb8e9ca7369682c : SUCCESS in 44m 50s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/04f6d2a3c8044c93bfdb3a296faec4bc : SUCCESS in 1h 04m 49s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/36d53df0bf3345cdb9504fb9e268296a : SUCCESS in 1h 53m 42s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/70ba9d13f33e409db410789ca902ad55 : SUCCESS in 22m 49s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a1aa2840ebca4057aba54c1d5fb5ba4b : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":1},{"id":"feaba44bbe58888e8cbaeb343bb7d265d19dcbc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 19:12:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffb7f96e16e57020130bddf2371d70966b572d4f","revisions":{"ffb7f96e16e57020130bddf2371d70966b572d4f":{"kind":"REWORK","_number":1,"created":"2019-07-27 14:00:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/39/673139/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/673139/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/673139/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/673139/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/673139/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/673139/1"}}},"commit":{"parents":[{"commit":"d1b51adb3d9638a393277b7100406a52d421fbfe","subject":"Merge \"Ignore --limit for localhost plays\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d1b51adb3d9638a393277b7100406a52d421fbfe"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-27 13:58:43.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-27 13:58:43.000000000","tz":120},"subject":"Add support for custom CloudKitty configuration","message":"Add support for custom CloudKitty configuration\n\nChange-Id: I0c0589387da53458f41df280dff0179715448aaa\nStory: 2006292\nTask: 36003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffb7f96e16e57020130bddf2371d70966b572d4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffb7f96e16e57020130bddf2371d70966b572d4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
