)]}'
{"id":"openstack%2Fkayobe~674262","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I104d6af48ddd916460ee454e17a50a2475de6bff","project":"openstack/kayobe","branch":"stable/rocky","hashtags":[],"change_id":"I104d6af48ddd916460ee454e17a50a2475de6bff","subject":"Ignore --limit for localhost plays","status":"MERGED","created":"2019-08-02 10:52:00.000000000","updated":"2019-08-15 09:49:34.000000000","submitted":"2019-08-15 09:49:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674262-1565862574648-10c1e968","meta_rev_id":"26e08621c798da2a96098651277537eab833b98a","_number":674262,"virtual_id_number":674262,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-08-15 09:49:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-08-07 15:26:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-15 08:01:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-15 08:01:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 15:26:41.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-08-15 08:01:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-08-15 09:49:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bbdb5f59be72389623c1c08fce5bb83dfb33ff64","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-02 10:52:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93b9c065b121a2a67f2d8a96a62b4eee85701263","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 12:36:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://logs.opendev.org/62/674262/1/check/build-openstack-sphinx-docs/a587705/html/ : SUCCESS in 4m 23s\n- openstack-tox-docs https://logs.opendev.org/62/674262/1/check/openstack-tox-docs/5fd81ec/html/ : SUCCESS in 3m 40s\n- openstack-tox-cover https://logs.opendev.org/62/674262/1/check/openstack-tox-cover/6a6a251/cover/ : SUCCESS in 4m 14s\n- openstack-tox-py35 https://logs.opendev.org/62/674262/1/check/openstack-tox-py35/52a74d3/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://logs.opendev.org/62/674262/1/check/openstack-tox-pep8/8ae0beb/ : SUCCESS in 3m 32s\n- openstack-tox-py27 https://logs.opendev.org/62/674262/1/check/openstack-tox-py27/8be726b/ : SUCCESS in 4m 41s\n- kayobe-tox-ansible-syntax https://logs.opendev.org/62/674262/1/check/kayobe-tox-ansible-syntax/bd66889/ : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://logs.opendev.org/62/674262/1/check/kayobe-tox-ansible/49fe96e/ : SUCCESS in 8m 17s\n- kayobe-tox-molecule https://logs.opendev.org/62/674262/1/check/kayobe-tox-molecule/da34d00/ : SUCCESS in 38m 12s\n- kayobe-overcloud-centos https://logs.opendev.org/62/674262/1/check/kayobe-overcloud-centos/c3fad2f/ : SUCCESS in 57m 32s\n- kayobe-overcloud-upgrade-centos https://logs.opendev.org/62/674262/1/check/kayobe-overcloud-upgrade-centos/580266b/ : SUCCESS in 1h 35m 18s\n- kayobe-seed-centos https://logs.opendev.org/62/674262/1/check/kayobe-seed-centos/95b657e/ : SUCCESS in 17m 38s\n- kayobe-seed-upgrade-centos https://logs.opendev.org/62/674262/1/check/kayobe-seed-upgrade-centos/6c7ce03/ : SUCCESS in 28m 33s","accounts_in_message":[],"_revision_number":1},{"id":"bb2a306c331bfdc5128d96296b71dc17fe2fa22c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-08-07 15:26:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"982967c8e1e619badcae745bfe61ff5c45b680cd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-15 08:01:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3af385a06ca45b871ae734f1899a6ed269c37a6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 08:01:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8b056bae9f922cdc412a1b25cef1876ea54d853","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 09:49:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"26e08621c798da2a96098651277537eab833b98a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 09:49:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/35e2a1e3138d45cba4b33bf953f76e13 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55926d38c6854510b31304c062e41613 : SUCCESS in 3m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c1695d1d0175401bb0413fe711f721fe : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86ce93fc650d4a5f82f12861c9872391 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/225320635d844f88b908895445665d30 : SUCCESS in 4m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/57b35d68b0d3437e86c68db3aa5668a3 : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/65d952723a384dd0a2d1632d3c621c27 : SUCCESS in 9m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d3c0903da19d4306925c550713d1c111 : SUCCESS in 33m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9f7d8aa71e1a42eb85419608cfc78846 : SUCCESS in 1h 00m 38s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/0c306a22c2444f46817f90183ad9d95e : SUCCESS in 1h 41m 56s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8d0beb86906645f3859de31e723fd554 : SUCCESS in 17m 54s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8ccdee07e83e4180b0d3e05af0d2d1ab : SUCCESS in 27m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c1df29e60e3e6df440b93a0561dbcab4428a848","revisions":{"9c1df29e60e3e6df440b93a0561dbcab4428a848":{"kind":"REWORK","_number":1,"created":"2019-08-02 10:52:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/62/674262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/674262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/674262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/674262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/674262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/674262/1"}}},"commit":{"parents":[{"commit":"5df4f5cb2fb2816c63b0b2db774d6a3af66f531b","subject":"Merge \"Provide keystone.conf to kolla-ansible by default\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5df4f5cb2fb2816c63b0b2db774d6a3af66f531b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-21 11:37:13.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-02 10:51:50.000000000","tz":60},"subject":"Ignore --limit for localhost plays","message":"Ignore --limit for localhost plays\n\nIf using the kayobe --limit \u003chosts\u003e argument, only those hosts matching\nthe limit will have an inventory generated for Kolla Ansible. This can\ncause problems for Kolla Ansible, even when using a similar limit via\n--kolla-limit \u003chosts\u003e, since it gathers facts for all hosts, and expects\nall host variables to be defined.\n\nThis change adds a new \u0027ignore_limit\u0027 argument to the playbook running\nmethods, which allows us to override and ignore the --limit argument\nprovided by the user, ensuring that we always run these plays.\n\nChange-Id: I104d6af48ddd916460ee454e17a50a2475de6bff\nStory: 2004805\nTask: 28960\n(cherry picked from commit 0b4b6cb79deb96f5cd4ada4212037032e0f79e00)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c1df29e60e3e6df440b93a0561dbcab4428a848"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c1df29e60e3e6df440b93a0561dbcab4428a848"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
