)]}'
{"id":"openstack%2Fkayobe~676643","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I89a973c0b600abece79bddcba5a46cc28a4f1df9","project":"openstack/kayobe","branch":"stable/rocky","hashtags":[],"change_id":"I89a973c0b600abece79bddcba5a46cc28a4f1df9","subject":"Disable chrony container by default","status":"MERGED","created":"2019-08-15 08:39:21.000000000","updated":"2019-08-23 13:05:15.000000000","submitted":"2019-08-23 13:05:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676643-1566565515158-66c16398","meta_rev_id":"210e2b6444162d8c12435eea257fe52095a325d5","_number":676643,"virtual_id_number":676643,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-08-23 13:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-08-23 11:14:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-08-22 10:26:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-08-23 11:14:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 10:26:35.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-08-23 11:14:39.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-08-23 13:05:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6aed4b9dbe7ec4aa9a4997d9859d73544158083","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 08:39:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3924e1dd8e25238deb230cf2f162f77541f1cf1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 10:07:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1df8b0b693e64552bca4b82434d7fc76 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab1abe75f1a4d32ae4ff7b5c4cebb3d : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e590a25e77b4d388c1cf11e2714fd6f : FAILURE in 3m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5d5bc471ffd04acb91ab928ad08e3f9d : FAILURE in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ecbb03785e541bb8e3b9af38f3d061a : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5158b2092ed249c3bbf6944ed8c05373 : FAILURE in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/219d395e2f354fa48229aaba16351ee3 : SUCCESS in 2m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a02cefcff564aaf96f2a5306f004709 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f01520d107b64ce1a1ba98af7221f929 : SUCCESS in 8m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa192dd82713475599e8a7cf4755fd66 : SUCCESS in 35m 49s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f54b0b71a1ec426e8541b8f7b6d2e973 : SUCCESS in 56m 01s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/910116dc2b664dc7b82d2fbf200fccb0 : SUCCESS in 1h 24m 41s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/750f943213f54c26988ed310e38d77b5 : SUCCESS in 18m 43s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b0ef3930a19f4ac09d41096c29688df6 : SUCCESS in 24m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8d94664f53b5eb8a7564bc62c52000cf70b20aa7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 11:19:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"acd475ef9ab20853eb405b42252b7b756c276ed1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 12:56:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b189106afaae4f9dae05ac7b93d5017f : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98547d053e604c8994c4d2d258933024 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64b8da963bb94060aad4d90e7a2ae3c8 : SUCCESS in 4m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ccefc4b6fde94254b2bd240b71d23186 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b10b59e4ca754cd48dcf2cc8d3c88c20 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3cac2ceb2bf04f539d34508eab034dd3 : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7acc440a942a4b62b70e67293c18a1d4 : SUCCESS in 2m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85b473d6c4574f94905db933573f4e30 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c20e4627e320425aaca5dacd463b3cdd : SUCCESS in 7m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/462940aa916c4662a990a6ddea58ed78 : SUCCESS in 37m 57s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/4fa5cb648429436ba2b46ff7201b6b67 : SUCCESS in 59m 54s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/6c77cd1b35554d888f0c62d44101eab7 : SUCCESS in 1h 31m 55s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/94178da6037541a9818602c7da1a3646 : SUCCESS in 17m 22s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ebc3e4c7d69d4d27aea8caa6de78a4a4 : SUCCESS in 26m 48s","accounts_in_message":[],"_revision_number":2},{"id":"2424ba6f087742d2efa8c72423e1c5d32d40570a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-22 10:26:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a7cacf6b6ad751a42ab483aca0ed578bcf63d2ca","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-08-23 11:14:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e3ac42bcf87dbf21636aea80f7e1e033a17c26d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 11:14:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"034cc0a5b981b26e12a75da582aea2b5fddcf21a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 13:05:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/dd0a1f1b655840f6aa6b834ef4631c05 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa5b8744cc6e4138af9e58beb02b7762 : SUCCESS in 3m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4b5659da384e4c1abdc348d8fe69f160 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77d7baa759e94eefab1e7cac78c7acd0 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a7ddde1abc84de78010034705931962 : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8579ae749f81433bbd06385d3eeeca10 : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c63791d290246e49177c4ef8eca3446 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9e99c171f191440881334ec3f77129e8 : SUCCESS in 9m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e6cd84f3285643438aab9781bd44112e : SUCCESS in 37m 50s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9c8bc1dd72b74063b88d3bbedb4755b4 : SUCCESS in 1h 02m 32s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a9165b659cb940e88fe61d37bdf731db : SUCCESS in 1h 45m 17s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/884e3b9c3efb4e7d9e9ec4f4e068f99c : SUCCESS in 18m 30s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3bd3419b02f149ff843e0e4e9b226e3c : SUCCESS in 29m 25s","accounts_in_message":[],"_revision_number":2},{"id":"210e2b6444162d8c12435eea257fe52095a325d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 13:05:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b4235b8ea78fc1f18b5a8b700d2584fc8f8ed126","revisions":{"2251a9e634ff755f221da017a0fe6914f98867b4":{"kind":"REWORK","_number":1,"created":"2019-08-15 08:39:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/676643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/676643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/676643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/676643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/676643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/676643/1"}}},"commit":{"parents":[{"commit":"5df4f5cb2fb2816c63b0b2db774d6a3af66f531b","subject":"Merge \"Provide keystone.conf to kolla-ansible by default\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5df4f5cb2fb2816c63b0b2db774d6a3af66f531b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-22 18:47:10.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-15 08:39:09.000000000","tz":60},"subject":"Disable chrony container by default","message":"Disable chrony container by default\n\nFixes an issue where multiple NTP daemons could be running on the\novercloud hosts, due to Kolla Ansible deploying a chrony container by\ndefault starting with the Rocky release.\n\nKayobe now overrides this default, to ensure that chrony does not conflict\nwith the NTP daemon deployed on the host. To use the containerised chrony\ndaemon instead, set ``kolla_enable_chrony`` to ``true`` in\n``${KAYOBE_CONFIG_PATH}/kolla.yml``. This will also disable the host NTP\ndaemon.\n\nTo ensure that chrony is not running, Kayobe removes the chrony container\nif ``kolla_enable_chrony`` is ``false`` in the following commands:\n\n* ``kayobe overcloud service deploy``\n* ``kayobe overcloud service reconfigure``\n* ``kayobe overcloud service upgrade``\n\nThe play in Kayobe is tagged with ``stop-chrony``.\n\nChange-Id: I89a973c0b600abece79bddcba5a46cc28a4f1df9\nStory: 2005272\nTask: 30122\n(cherry picked from commit 650919f52736b297de50bc48bb7fc71e19cb64a9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2251a9e634ff755f221da017a0fe6914f98867b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2251a9e634ff755f221da017a0fe6914f98867b4"}]},"branch":"refs/heads/stable/rocky"},"b4235b8ea78fc1f18b5a8b700d2584fc8f8ed126":{"kind":"REWORK","_number":2,"created":"2019-08-15 11:19:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/676643/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/676643/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/676643/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/676643/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/676643/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/676643/2"}}},"commit":{"parents":[{"commit":"9c1df29e60e3e6df440b93a0561dbcab4428a848","subject":"Ignore --limit for localhost plays","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c1df29e60e3e6df440b93a0561dbcab4428a848"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-03-22 18:47:10.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-15 11:19:40.000000000","tz":60},"subject":"Disable chrony container by default","message":"Disable chrony container by default\n\nFixes an issue where multiple NTP daemons could be running on the\novercloud hosts, due to Kolla Ansible deploying a chrony container by\ndefault starting with the Rocky release.\n\nKayobe now overrides this default, to ensure that chrony does not conflict\nwith the NTP daemon deployed on the host. To use the containerised chrony\ndaemon instead, set ``kolla_enable_chrony`` to ``true`` in\n``${KAYOBE_CONFIG_PATH}/kolla.yml``. This will also disable the host NTP\ndaemon.\n\nTo ensure that chrony is not running, Kayobe removes the chrony container\nif ``kolla_enable_chrony`` is ``false`` in the following commands:\n\n* ``kayobe overcloud service deploy``\n* ``kayobe overcloud service reconfigure``\n* ``kayobe overcloud service upgrade``\n\nThe play in Kayobe is tagged with ``stop-chrony``.\n\nChange-Id: I89a973c0b600abece79bddcba5a46cc28a4f1df9\nStory: 2005272\nTask: 30122\n(cherry picked from commit 650919f52736b297de50bc48bb7fc71e19cb64a9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4235b8ea78fc1f18b5a8b700d2584fc8f8ed126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4235b8ea78fc1f18b5a8b700d2584fc8f8ed126"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
