)]}'
{"id":"openstack%2Fkayobe~676645","triplet_id":"openstack%2Fkayobe~stable%2Fqueens~Id6e14286b5eb2b767a515e7edfc56741fb8d2c78","project":"openstack/kayobe","branch":"stable/queens","topic":"ip-allocation-stable/queens","hashtags":[],"change_id":"Id6e14286b5eb2b767a515e7edfc56741fb8d2c78","subject":"Stop allocating network and broadcast addresses","status":"MERGED","created":"2019-08-15 08:40:44.000000000","updated":"2019-08-15 12:02:08.000000000","submitted":"2019-08-15 12:02:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676645-1565870528911-0fc9d827","meta_rev_id":"5ea3b82419778fe7c86887b5125056e854ad4ead","_number":676645,"virtual_id_number":676645,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-15 12:02:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-08-15 10:57:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-15 08:42:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-08-15 10:57:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 08:42:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-08-15 12:02:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d67aef470adc4c08666efd5a007f4bf1899d736","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 08:40:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"da46a3021f19553055c6a1732c0c72c0f39f9ce3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-15 08:42:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b739abcda83b3991b866e618efca4f5b0a6ccee4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 10:03:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4ed6137775974b3aa333dbf6791a11a5 : SUCCESS in 4m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b54468817e824983b295f1fb0dd714ab : SUCCESS in 3m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/558d96f2b8fa4d27898252ed578c7eef : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc43b6e7ed9e4b93a3d644a5051d2643 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc9dc510ee574286aa29609e1f90555a : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/197ae1735ef7460096cc2132cd5bccc6 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6f0eb5cc37a94d4ab848d16e9a87f6db : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7f9cdc1551b54f838b56deeb2c0e0480 : SUCCESS in 7m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/898a06368cca4f098b2e926a55dd643b : SUCCESS in 27m 51s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/eedfc4a2ebe1409dbe49033ee1f219cf : SUCCESS in 51m 07s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b4eb93fee0f445508398e40a3118dd65 : SUCCESS in 1h 10m 31s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/69c20296ccab4909acec7216f4b40c70 : SUCCESS in 19m 36s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/52c662f3c3f645f79aa308bd77f00606 : SUCCESS in 26m 00s","accounts_in_message":[],"_revision_number":1},{"id":"bcaaec0d4a914ffa699f2e4cf9f154060a70d3e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 10:57:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"015a0aba897d0ea82aa999c5b0e9a0a6918b2686","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 10:57:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"625a788c43e3201b8470df9d2df31be96e50eaea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 12:02:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1d190f8f37b64e7a8970b2826b0f641c : SUCCESS in 3m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/de95171d41f34914a794625dc5b8d39b : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faeeb220afda457eb3473a63b2bc9b01 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8dff81030f4f4ddc9b1a9ad8b86fe575 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b55cf6f95c845129aa8c84b88c55841 : SUCCESS in 2m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa499e566a174c96b0da141fc0dbc631 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1914899f9ccd4222b55ead1c899e8146 : SUCCESS in 6m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c0939f27e07e4d39add1231b9d99d088 : SUCCESS in 33m 43s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e23f23af258a4b649f3ab7d5cfc7e871 : SUCCESS in 50m 15s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1508961d7a554df4a4e176da82781155 : SUCCESS in 1h 00m 35s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/32dfaf85b3934c09bade75bead4ed80f : SUCCESS in 16m 42s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/9819fa4b2fb74740b14c185214bfd544 : SUCCESS in 22m 04s","accounts_in_message":[],"_revision_number":1},{"id":"afe17957f989e0596c074e694bc4c64d79ef351e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 12:02:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"559408e85c8bafd3a3b1f9bb600ae3837fe840db","revisions":{"559408e85c8bafd3a3b1f9bb600ae3837fe840db":{"kind":"REWORK","_number":1,"created":"2019-08-15 08:40:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/45/676645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/676645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/676645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/676645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/676645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/676645/1"}}},"commit":{"parents":[{"commit":"2b5078b656859f906183cb295f274d78e3700e95","subject":"Ignore --limit for localhost plays","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2b5078b656859f906183cb295f274d78e3700e95"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-02 12:08:34.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-15 08:40:44.000000000","tz":0},"subject":"Stop allocating network and broadcast addresses","message":"Stop allocating network and broadcast addresses\n\nWhen an allocation pool range is not defined by the operator, we should\nnot include network and broadcast addresses in the list of IP addresses\nto give to hosts.\n\nChange-Id: Id6e14286b5eb2b767a515e7edfc56741fb8d2c78\nStory: 2006267\nTask: 35958\n(cherry picked from commit 697e74e41acc7129bb6b0937012209136cd4e01a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/559408e85c8bafd3a3b1f9bb600ae3837fe840db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/559408e85c8bafd3a3b1f9bb600ae3837fe840db"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
