)]}'
{"id":"openstack%2Fkayobe~676647","triplet_id":"openstack%2Fkayobe~stable%2Fqueens~I7ecd6cd3460bea56d0ab9cb221d0ef5b5ec2d0bb","project":"openstack/kayobe","branch":"stable/queens","topic":"story/2006378-stable/queens","hashtags":[],"change_id":"I7ecd6cd3460bea56d0ab9cb221d0ef5b5ec2d0bb","subject":"Restrict ncclient to 0.6.2 to avoid unknown host key issue","status":"MERGED","created":"2019-08-15 08:41:17.000000000","updated":"2019-08-15 22:05:01.000000000","submitted":"2019-08-15 22:05:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676647-1565906701251-5b3f03e3","meta_rev_id":"d0a126e089bfa38f08c807cdf6c872780c8982a8","_number":676647,"virtual_id_number":676647,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-08-15 22:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-08-15 20:16:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-15 08:41:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-08-15 20:16:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 08:41:42.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-08-15 20:16:41.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-08-15 22:05:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3500489dd27557b94759c9f18a989b73a086373","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 08:41:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9afa082c814c2561231483e7ad599c6b6f03a515","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-15 08:41:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d6e63bd167f9f17454d0f7fb1d45a68a840c8aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 10:20:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fc8d868f406141c08f2b60599eb76ffb : SUCCESS in 3m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea31d44a3fcd446ea8ef690329f6dcff : SUCCESS in 5m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bc9801603dab485193ab467dcc01d1a4 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/351dbafb988e436da41966f84eb73d86 : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e53005dd9d2246f9b4b1a6dc98672df0 : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b7728f263e1b44e38360ee0d4db39f9a : SUCCESS in 3m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5130a1366b2242e58d5fae24eafe0407 : SUCCESS in 9m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6cf58d4c4722444eb0a6d2c55cdada86 : SUCCESS in 34m 48s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8d5645e210484db99569fe2238a6eedc : SUCCESS in 51m 14s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/bf727b018e2d4a3fa90c0affc38e9c4c : SUCCESS in 1h 04m 46s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/97c763da45e843b481339a207df7e939 : SUCCESS in 16m 13s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/eab6b88c506f43df926cc0b5b4db2365 : SUCCESS in 23m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5d30ca9afc5964221d07b722aa64b78229e7159f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-15 20:16:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3062c697aa13c89a9be827b324347d13c40acf32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 20:16:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9dd85dd5a1f438890887a17796045603bf08231f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 22:05:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c1028198d1f74676a9970749514fae34 : SUCCESS in 5m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5bd1aa73c1f34a0793cc7b863d220ef5 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/259c4fa667604fa69dec42eda824aa28 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b7a0b47a5e56458395651ba2ec445a94 : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2f2d860a6fd54bd2a62e8ae4d1706342 : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/80e54c186f0a45d884771c99e1e30ed1 : SUCCESS in 7m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/252a5b62a8c74c92810ec0d1ce26178a : SUCCESS in 32m 16s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/104da747445641808239f793b1627227 : SUCCESS in 49m 02s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/94d19379885b4c4d933369db1d33b3b0 : SUCCESS in 1h 34m 06s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3c03e2fd7184478abd24120b56c8343a : SUCCESS in 19m 50s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/7bf3ad06fc2f4e5fa49ef47e39d2336e : SUCCESS in 31m 32s","accounts_in_message":[],"_revision_number":1},{"id":"cc3b6db00f2b7d3da123ad2cf06a026349593320","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 22:05:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b91946ddb32329e09585acb2a806d0cb0b971878","revisions":{"b91946ddb32329e09585acb2a806d0cb0b971878":{"kind":"REWORK","_number":1,"created":"2019-08-15 08:41:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/47/676647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/676647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/676647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/676647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/676647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/676647/1"}}},"commit":{"parents":[{"commit":"2b5078b656859f906183cb295f274d78e3700e95","subject":"Ignore --limit for localhost plays","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2b5078b656859f906183cb295f274d78e3700e95"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-08 13:20:57.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-15 08:41:17.000000000","tz":0},"subject":"Restrict ncclient to 0.6.2 to avoid unknown host key issue","message":"Restrict ncclient to 0.6.2 to avoid unknown host key issue\n\nJuniper physical network device configuration fails when running the\nfollowing command:\n\nkayobe physical network configure --group switches\n\nThe error is as follows:\n\nfatal: [switch]: FAILED! \u003d\u003e {\n        \"msg\": \"Unknown host key [\u003ckey\u003e] for [[\u003cIP\u003e]:830]\"\n}\n\nThis is due to a bug [1] in ncclient.\n\nThis patch avoids the issue by pinning ncclient to version 0.6.2 or\nlower.\n\n[1] https://github.com/ncclient/ncclient/issues/302\n\nChange-Id: I7ecd6cd3460bea56d0ab9cb221d0ef5b5ec2d0bb\nStory: 2006378\nTask: 36174\n(cherry picked from commit 2a17775514009754175cb94a963bf44ea410e1ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b91946ddb32329e09585acb2a806d0cb0b971878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b91946ddb32329e09585acb2a806d0cb0b971878"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
