)]}'
{"id":"openstack%2Fkayobe~676927","triplet_id":"openstack%2Fkayobe~master~Ifc5a36743c98f32a9d841636119bc0d33621cef3","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ifc5a36743c98f32a9d841636119bc0d33621cef3","subject":"Update release notes for stable/stein","status":"MERGED","created":"2019-08-16 12:52:35.000000000","updated":"2019-08-21 14:16:35.000000000","submitted":"2019-08-21 14:14:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676927-1566396844629-c41f93e9","meta_rev_id":"535b051e3d263e7761f723eed4ab841c73a8f456","_number":676927,"virtual_id_number":676927,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-21 14:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-08-21 14:01:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-16 13:35:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-08-21 14:01:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 13:35:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-08-21 14:01:57.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-08-21 14:14:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48fc6ee218bfcb7a62f5b29d291e9561c72f69df","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-16 12:52:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9176cbb0a2d7508a7fd3da869c5acde0ff3bdbb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 13:01:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/867d3d0282834dbc93f0dd19f5806faa : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e2a4e82cf5e42f396e3f8f2dc3746be : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5eb7dcac835b4154ba732550b19a0fa3 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6e1a57ed949499154582287e0fedf0c875af738e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-16 13:14:35.000000000","message":"Patch Set 1: Code-Review-1\n\nI think you need an :earliest-version: tag, as the page generated for Stein includes release notes for 5.0.0.0rc1 and 4.0.0.0rc1 as well.","accounts_in_message":[],"_revision_number":1},{"id":"c5d2984c8398125a3e8e3eaf895858afdc6c2cc3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-16 13:20:40.000000000","message":"Patch Set 1:\n\n\u003e I think you need an :earliest-version: tag, as the page generated\n \u003e for Stein includes release notes for 5.0.0.0rc1 and 4.0.0.0rc1 as\n \u003e well.\n\nYes I saw that. For rocky we have earliest-version 5.0.0, but if I use 6.0.0 here it complains because the tag does not exist. Perhaps I need to use 6.0.0.0rc1?","accounts_in_message":[],"_revision_number":1},{"id":"d225ec972ce18a2426d0092880f2039c9f971f13","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-16 13:31:15.000000000","message":"Patch Set 1:\n\n\u003e \u003e I think you need an :earliest-version: tag, as the page generated\n \u003e \u003e for Stein includes release notes for 5.0.0.0rc1 and 4.0.0.0rc1 as\n \u003e \u003e well.\n \u003e \n \u003e Yes I saw that. For rocky we have earliest-version 5.0.0, but if I\n \u003e use 6.0.0 here it complains because the tag does not exist. Perhaps\n \u003e I need to use 6.0.0.0rc1?\n\nYes, it appears to work with:\n\n:earliest-version: 6.0.0.0rc1","accounts_in_message":[],"_revision_number":1},{"id":"bdd2a723dbb21ba66215404116c2d6a2d8a9e8d1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-16 13:32:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"33bc074b957313d2c418fc0e692e6bf22dc104f0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-16 13:35:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e22a123c16fc137d2efcacb74dac146ac32df45a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 13:43:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae334cf4f62145dfbce13e86f7513c6f : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c830b33471144764910a2b9d9523cd0a : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bf64fbf5c2a45c980c9081ecf536b71 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9202d18f72c860c86a71ed8149873d94a88ad7a6","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-21 14:01:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nhttps://3a529b9c3a74f1543cc3-752ab92f1df702672d910e68345a54f5.ssl.cf2.rackcdn.com/676927/2/check/build-openstack-releasenotes/7bf64fb/html/stein.html\n\n^ Looks good.","accounts_in_message":[],"_revision_number":2},{"id":"5c242df9f7025c8174217a7cbb33bde7e789eab4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 14:02:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a752333103d046bf84b19b912a3772172c52fb78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 14:14:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/846cf754b9784cbdba3cfc69a351a781 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4303515402184550a4507d8c8e5692c1 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d90e3c52b90494e8f72a645aa78d9a6 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"7d58f84966c04d90edae50fa55585b892bdbfd1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 14:14:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"535b051e3d263e7761f723eed4ab841c73a8f456","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 14:16:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70e431d491e94f1bbdfffdc3f11daec3 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd973178ea0fe016ccf69aadd761fb8a09a19a9d","revisions":{"b31294fa77294e31440c88fd5fd362026ab197b9":{"kind":"REWORK","_number":1,"created":"2019-08-16 12:52:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/27/676927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/676927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/676927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/676927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/676927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/676927/1"}}},"commit":{"parents":[{"commit":"93170249303bcfe0ee78350c4dbed02bf5895bee","subject":"Merge \"Generate openrc files during control host bootstrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93170249303bcfe0ee78350c4dbed02bf5895bee"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 12:50:14.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 12:50:14.000000000","tz":60},"subject":"Update release notes for stable/stein","message":"Update release notes for stable/stein\n\nChange-Id: Ifc5a36743c98f32a9d841636119bc0d33621cef3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b31294fa77294e31440c88fd5fd362026ab197b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b31294fa77294e31440c88fd5fd362026ab197b9"}]},"branch":"refs/heads/master"},"cd973178ea0fe016ccf69aadd761fb8a09a19a9d":{"kind":"REWORK","_number":2,"created":"2019-08-16 13:32:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/27/676927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/676927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/676927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/676927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/676927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/676927/2"}}},"commit":{"parents":[{"commit":"93170249303bcfe0ee78350c4dbed02bf5895bee","subject":"Merge \"Generate openrc files during control host bootstrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93170249303bcfe0ee78350c4dbed02bf5895bee"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 12:50:14.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 13:32:19.000000000","tz":0},"subject":"Update release notes for stable/stein","message":"Update release notes for stable/stein\n\nChange-Id: Ifc5a36743c98f32a9d841636119bc0d33621cef3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd973178ea0fe016ccf69aadd761fb8a09a19a9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd973178ea0fe016ccf69aadd761fb8a09a19a9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
