)]}'
{"id":"openstack%2Fkayobe~676941","triplet_id":"openstack%2Fkayobe~master~Ie7f588aab59e57efe240a58a3bd0a6f09fbcefb7","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ie7f588aab59e57efe240a58a3bd0a6f09fbcefb7","subject":"Revert \"Use OpenStack Stein release\"","status":"MERGED","created":"2019-08-16 13:28:16.000000000","updated":"2019-09-03 18:56:10.000000000","submitted":"2019-09-03 18:50:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676941-1567536658511-8928f7a0","meta_rev_id":"2c8c40fc820367981a94b32c11fa1ca70b801a9e","_number":676941,"virtual_id_number":676941,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-03 18:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-08-22 12:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-08-28 07:45:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-22 12:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2019-08-28 08:22:09.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-28 07:57:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 10:22:20.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-08-28 07:45:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-28 07:57:31.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-09-03 18:50:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1738c2fdd8b6e0f4dd7821e0faa524c8c460b2ea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-16 13:28:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"374eabc54ae066ec0974854cf2f1668f40ed5d66","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-16 13:36:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c340d9966c2f7643e94a4af1ec50b235c243063","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 15:31:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cbae210ae994c7686e10bbe5c145ead : SUCCESS in 4m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c693aa9fd2784ad58483b32927d79cfa : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a868815dbd54441598c4f3f329bcce0b : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/390eaa4605054d3691915f096ddd5e3c : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/65d92a088c934f60b59b66b36dc7d883 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aaffde3669d4fe69596897fcd785347 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0069331368264b76bc3cc2749e02ff28 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb0dee60d88e4948a261234491d34df7 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c0e572aaa63f45439071cc2b868c8df4 : SUCCESS in 8m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c2ad21e893a3410bb64ba5c66d8f7e1f : SUCCESS in 45m 47s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/01a8c61875344808bc2b7433a210955f : SUCCESS in 1h 00m 55s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/2c57f22097e146599e3f09860b2c3fea : SUCCESS in 1h 57m 35s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/434d2fb9b65f4f6fa7ec596445a1c6fe : SUCCESS in 22m 13s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/7876495d4a7c4ae08165099d90e55f24 : SUCCESS in 25m 16s","accounts_in_message":[],"_revision_number":1},{"id":"59d02445529960b2e0073ef298709c214aeeeccf","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-22 10:22:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e95362531596da33efbdb68e09da7bf07962fb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:22:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb29b3ede0e3aa893ca8cb90acc3728c7b3d5d7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 11:41:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a8afc6444c1e4923873efed53445dffa : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/874240fd842d49b8a108ffa0add03b7a : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a588b60be2e45069ebd1f44dbb25018 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fe00edefa9849f8bc657875c05d370a : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52b9b6e6bb104be9924288ad45e8edf4 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ce0a221c2344d84afd58464b1881ad0 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/97f50349aa8f40ee8a3a493ede43191b : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/00198f55320c479894b264a3fbf9d326 : SUCCESS in 9m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a96efc471ad4dd08960cb977734a0a9 : SUCCESS in 43m 48s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/4b383e21f55c4913915c3c5bd2b342b4 : FAILURE in 26m 38s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3bb066bfd5c349d694d7e9f7a8e3d0c8 : FAILURE in 1h 17m 01s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c38e62e7012941bb869e87b23cf7ce26 : FAILURE in 13m 22s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/1c453676688846caa0fe85d3fc302d5d : FAILURE in 27m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b58681824360647a74d605108db4fffd8372cc10","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-22 12:52:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"fdb812011bad40610265350f481f2df50ded8562","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 14:49:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6fee1a3510b4ff48e5ec503ad9ee57e : SUCCESS in 4m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/69ad4ca1027740ab8c0125f772811f2b : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8797db33d1b462a88541d54f6dc770f : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6cd168be82f42039a186752ba26812e : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/061dd89003ac435db51c54473eca07a3 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5f2966604d8416e8fe0ecb995a094eb : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4942b6a392684265b8ecfb85bce3a957 : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29abd6b418fc4bce8c295ac06681ee30 : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6c2cbbebf82e45bc94204b3d375c9cf3 : SUCCESS in 9m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa4a89ed25634f1995227b99ec481fe8 : SUCCESS in 37m 27s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/c94c405c719340f0b0be1c9e7ff9df5a : SUCCESS in 1h 08m 36s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/6e2d0e3bb0c645f286333831588e7aaf : SUCCESS in 1h 45m 52s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/619b118cd66c49fc9e7d9e6c181fb40f : FAILURE in 9m 45s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a45a1a030e0e4a228b30607770b0f6c3 : SUCCESS in 27m 14s","accounts_in_message":[],"_revision_number":2},{"id":"ff515292aad4a1b2f9571c09ddc3b767cacfe526","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-22 15:42:03.000000000","message":"Patch Set 2: Workflow-1\n\nBlocked on https://review.opendev.org/#/c/678047","accounts_in_message":[],"_revision_number":2},{"id":"7b259c6d2d063ac2a3725f0abc2f9255a1b5e93e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-27 10:38:33.000000000","message":"Patch Set 2: -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f36f2aab80754f2fbb45806bba204e7e169195e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 12:33:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb361517dcb4445fa72e49c667b93918 : SUCCESS in 4m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6dee1d4c5e124072b1c8685c47957725 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8dcdeac9b164b4b849d3b74baae7baf : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36a272fc050c4fb98e29c4a53f1079a7 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41cca38cb9564fd5bf48f8e821233107 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c75929fa0964008abe28ec6ba3bd0e3 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd8692b6bea947b4be57392c51c7582e : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ee3a0e7334b84439979201bf65462b00 : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/96041f5b6c684883a928cb97830e4560 : SUCCESS in 10m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7c2e14b5a50845b2b033652bc537e756 : SUCCESS in 46m 45s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/3bbe502e3b5b4012aaac4c8382478f2f : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/acab2400cac045c5ba5a21c68ca2d9ed : SUCCESS in 1h 49m 00s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/cd51112d474c45b099b37b1d2be42fba : SUCCESS in 20m 19s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/968a1dfa63e14370ac267f9ff85f59e1 : SUCCESS in 27m 31s","accounts_in_message":[],"_revision_number":2},{"id":"7d68f20ae0f76e471aad42aead00ae957507b651","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-28 07:45:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"87a1273c32eef6098e5fb0866c89ef0cc124dbf7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-28 07:57:31.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"275c6659c1eacc7dbf5edaa5bd97cf281f694e27","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-28 08:22:09.000000000","message":"Patch Set 2:\n\nDepends on https://review.opendev.org/#/c/678009","accounts_in_message":[],"_revision_number":2},{"id":"dc5a0cda0589ebf7f2251f374b9a264feb9bc4ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 16:43:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f5f0af30f1fac61eb685520bfdc36d576c2820ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 18:50:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2035cb240384b6d84e09de04500e194635212fb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 18:50:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ff009868d7ce4ec7a972a108d3c1f522 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a29dc371f0ef4b49bd38577dc1fdd70f : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56be75b74782481caf485b99c2699ac1 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f310d4054244412901c77b5dfdeee44 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9035fb18f63b4245a6981d59dfd7ff14 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94d235e471cb43b2b117f7c3b2b2e863 : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c2031102a4be465d8a74c61540d716f7 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db1b2fc0d89840f38b5253449a1fd289 : SUCCESS in 9m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/480e0e5545864ad5adbe14df27f961ae : SUCCESS in 41m 31s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/01d9a9a8bd8c46b8b4525adb901319f2 : SUCCESS in 1h 08m 40s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/7cc3f20a6c914f5ba2b8023df64c4bd0 : SUCCESS in 1h 41m 36s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8bad32b44ec844c893edc14e6762c1a0 : SUCCESS in 22m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/1ea5ff226bb243c7b0583a870f9f98fc : SUCCESS in 27m 09s","accounts_in_message":[],"_revision_number":2},{"id":"2c8c40fc820367981a94b32c11fa1ca70b801a9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 18:56:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0ab2de612dc4beeac0f131030ff6a60 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ebe13df6f7516594e7bf663e98acda716c393ebb","revisions":{"c875f80dafc89c6ce360b0c20b7b90c4cbff8229":{"kind":"REWORK","_number":1,"created":"2019-08-16 13:28:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/41/676941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/676941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/676941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/676941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/676941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/676941/1"}}},"commit":{"parents":[{"commit":"93170249303bcfe0ee78350c4dbed02bf5895bee","subject":"Merge \"Generate openrc files during control host bootstrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93170249303bcfe0ee78350c4dbed02bf5895bee"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 13:26:50.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 13:28:01.000000000","tz":60},"subject":"Revert \"Use OpenStack Stein release\"","message":"Revert \"Use OpenStack Stein release\"\n\nThis reverts commit b46b2b72bbec2a076a8175fc0aff96cbdefcabff.\n\nSwitch dependencies and upper constraints back to master branch for\ndevelopment after cutting stable/stein branch.\n\nChange-Id: Ie7f588aab59e57efe240a58a3bd0a6f09fbcefb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c875f80dafc89c6ce360b0c20b7b90c4cbff8229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c875f80dafc89c6ce360b0c20b7b90c4cbff8229"}]},"branch":"refs/heads/master"},"ebe13df6f7516594e7bf663e98acda716c393ebb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-22 12:52:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/41/676941/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/676941/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/676941/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/676941/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/676941/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/676941/2"}}},"commit":{"parents":[{"commit":"48c643abbe254209b2576096e9ec209edd6ad578","subject":"Include branch name in URL for IPA on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48c643abbe254209b2576096e9ec209edd6ad578"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 13:26:50.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-22 12:52:42.000000000","tz":60},"subject":"Revert \"Use OpenStack Stein release\"","message":"Revert \"Use OpenStack Stein release\"\n\nThis reverts commit b46b2b72bbec2a076a8175fc0aff96cbdefcabff.\n\nSwitch dependencies and upper constraints back to master branch for\ndevelopment after cutting stable/stein branch.\n\nChange-Id: Ie7f588aab59e57efe240a58a3bd0a6f09fbcefb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebe13df6f7516594e7bf663e98acda716c393ebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebe13df6f7516594e7bf663e98acda716c393ebb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
