)]}'
{"id":"openstack%2Fkayobe~676985","triplet_id":"openstack%2Fkayobe~master~Ic3c2c1c780973e46450c93bee83a59b72892ddd7","project":"openstack/kayobe","branch":"master","attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-08-27 00:17:25.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic3c2c1c780973e46450c93bee83a59b72892ddd7","subject":"WIP: Only generate relevant configuration in ml2_conf.ini","status":"NEW","created":"2019-08-16 16:39:01.000000000","updated":"2023-08-27 00:17:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"68f20c682468e7e79220ce860a4e6e4cf514c816","_number":676985,"virtual_id_number":676985,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-03 17:18:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 17:52:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-27 00:17:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"dae0983f3d79d0982f609ac46d4181765879b07b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-16 16:39:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73640ae4f0a84ec854182f4b85a35516ab338f48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 17:52:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29e14d710f8e4fe5a35660809c53cf59 : SUCCESS in 4m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dc3bee30225b461792f83f97fb97b58b : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77f32c93aa404f5d8987e6cd63ac15f2 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f37a53df4694e8f820c678a76606a43 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe44ce68b1984d8c852b75f9230a9ad8 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bebeba9683b4cebb381b6a3f8466d7e : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17a8ecc800004d989df6af5ede703cf1 : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0e8c67997b404b208a99d64355633c50 : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e78e5f2b04ad4e89afafce40a148f045 : FAILURE in 43m 31s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/81b2569888854f37939dd91acc75120e : SUCCESS in 1h 05m 28s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/85f4c62bc64f48e98bcb258bc4a6a472 : FAILURE in 1h 01m 59s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/39a44c7f99e84c749e35a98f5b9d4ab7 : SUCCESS in 21m 46s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d8ac6eb572674b71b8d739870577ea8f : SUCCESS in 28m 55s","accounts_in_message":[],"_revision_number":1},{"id":"83814d23bd635543ef9d492cc4a4be9b21b6e87c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-22 14:40:36.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0cdb04694491489ae57fe52df2c734a463056be0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 11:47:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ad3e242337c83e8f273a5146ce33b73b5b00db6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 12:02:47.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/441ba4940f4a45feaaa22ffbf38bc68a : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73e834f53e194d36ac60294efdf69e7e : FAILURE in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e50abed68c5747cfb58061e0359a04e2 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e38023f98a2a4556b0f41db3019a6b31 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aec3f5c1253340b9b85c5fc0de9abbe0 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e77c73398ae438dae3186e673a0d5bb : SUCCESS in 6m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f0aa694525704ee1b311772dddabce1a : FAILURE in 10m 05s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"73921e4cecf7fddfada18488b7d3c2b78bcc4d3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 12:33:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cc3d181d571e13a6368804884aa0d815da02151a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 13:17:11.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58ef1d42b9ed4b73a8bbec35602bcf01 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a5329de0c514eadb601fa6a53c25adf : FAILURE in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8557133e01df4efb800a1b15445034ee : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7f4dfd6c4c94e999db87f1d5605520e : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b824fb1666d433c985f5cfb65ab2dd3 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bae867158c1477eac003ef6ed91f4dc : SUCCESS in 5m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/173bcc30c7994e8d93700e2433b2f488 : FAILURE in 15m 50s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"f954081a834d3f78e987f9ef920352afa3941266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 14:18:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b4a60369d9b5520b96b59819ab6b15d4df6155e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 14:44:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/909e3ecb45d14bf784ba89757ad9a902 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cba9ff6895bc4407b13331b6d6b9c6a7 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a731d561a484e87ba4cc06557213922 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/422a426e5707452aae116140323e65ad : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/568695ba6b9c4de882f3cff9790a3a1e : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb5fa4d319ac406492f888ba8ab6d4ed : SUCCESS in 7m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/05a6e90644fa42828cef29a0f2741a40 : FAILURE in 18m 12s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"e5aba61bf5d24ce09321ef193cf160de8727d768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 15:55:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2bb15fa70eb6c131ba0e342e1fe064ebbed8a26f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 16:30:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d109bf516cfb4736a5b34c14cfb21548 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec85c209deb34721907dda767aa10a11 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5498a42b220240348b98e15759cb6937 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb6e17d08b5b4b7aa1ce5f142a873d3f : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9db79f32b8764d17906d36b092df2827 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588f7251710c48ae9fb95b9998423c72 : SUCCESS in 7m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/628279571092413b868aa50754e2866f : FAILURE in 19m 41s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"81c16daf5951a5dff279cd8c7e85ee04ae470c15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 16:51:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9d0ebe54b652c4141869f971cb148084a6e87143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 17:18:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f69e8c1fb9c458d9b481aaf654e237e : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eda52df5c694400aa936ada9f9f824b : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae0bfb02290c4d46a10778ea8bd547a0 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12d0239d486e426ab75d07ce240c9b7a : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21eb09b3c1ae429abd1226a55c8dc227 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5c6919005dc4d9c8707b0503ff79f1f : SUCCESS in 5m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/54000dd428824093a99e04f75707966e : SUCCESS in 18m 03s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":6},{"id":"5490e9680c8176ab5bf9ec8895634f27f246cb0d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 17:36:50.000000000","message":"Patch Set 6:\n\nHmm, perhaps it\u0027s not required any more - kolla-ansible doesn\u0027t use ml2_conf.ini for agents.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7d3b61ce81120f1d069cbfe627e5b26e02271b64","revisions":{"c8c837c579955443a20b46f68f7bb53427d091e3":{"kind":"REWORK","_number":1,"created":"2019-08-16 16:39:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/85/676985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/676985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/676985/1"}}},"commit":{"parents":[{"commit":"93170249303bcfe0ee78350c4dbed02bf5895bee","subject":"Merge \"Generate openrc files during control host bootstrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93170249303bcfe0ee78350c4dbed02bf5895bee"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 16:37:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 16:37:24.000000000","tz":60},"subject":"Only generate relevant configuration in ml2_conf.ini","message":"Only generate relevant configuration in ml2_conf.ini\n\nOn compute nodes we end up with some neutron-server configuration in\nml2_conf.ini. This change adds logic to only generate it when necessary.\n\nTrivialFix\n\nChange-Id: Ic3c2c1c780973e46450c93bee83a59b72892ddd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c8c837c579955443a20b46f68f7bb53427d091e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c8c837c579955443a20b46f68f7bb53427d091e3"}]},"branch":"refs/heads/master"},"fc4a9d9481cfed1953467cd94851c6ceddc00a66":{"kind":"REWORK","_number":2,"created":"2021-02-03 11:47:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/85/676985/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/676985/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/676985/2"}}},"commit":{"parents":[{"commit":"586dccc236d1466e114f371573b39e5a2129a995","subject":"Merge \"Remove extra space characters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/586dccc236d1466e114f371573b39e5a2129a995"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 16:37:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-03 11:47:09.000000000","tz":0},"subject":"WIP: Only generate relevant configuration in ml2_conf.ini","message":"WIP: Only generate relevant configuration in ml2_conf.ini\n\nOn compute nodes we end up with some neutron-server configuration in\nml2_conf.ini. This change adds logic to only generate it when necessary.\n\nTrivialFix\n\nChange-Id: Ic3c2c1c780973e46450c93bee83a59b72892ddd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc4a9d9481cfed1953467cd94851c6ceddc00a66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc4a9d9481cfed1953467cd94851c6ceddc00a66"}]},"branch":"refs/heads/master"},"78a043843c160acd8156e19443482233a12adb6d":{"kind":"REWORK","_number":3,"created":"2021-02-03 12:33:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/85/676985/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/676985/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/676985/3"}}},"commit":{"parents":[{"commit":"586dccc236d1466e114f371573b39e5a2129a995","subject":"Merge \"Remove extra space characters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/586dccc236d1466e114f371573b39e5a2129a995"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 16:37:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-03 12:33:30.000000000","tz":0},"subject":"WIP: Only generate relevant configuration in ml2_conf.ini","message":"WIP: Only generate relevant configuration in ml2_conf.ini\n\nOn compute nodes we end up with some neutron-server configuration in\nml2_conf.ini. This change adds logic to only generate it when necessary.\n\nTrivialFix\n\nChange-Id: Ic3c2c1c780973e46450c93bee83a59b72892ddd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78a043843c160acd8156e19443482233a12adb6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78a043843c160acd8156e19443482233a12adb6d"}]},"branch":"refs/heads/master"},"3accdd087f23607f85fdd28bbe7e3c7e887c598f":{"kind":"REWORK","_number":4,"created":"2021-02-03 14:18:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/85/676985/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/676985/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/676985/4"}}},"commit":{"parents":[{"commit":"586dccc236d1466e114f371573b39e5a2129a995","subject":"Merge \"Remove extra space characters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/586dccc236d1466e114f371573b39e5a2129a995"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 16:37:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-03 14:18:08.000000000","tz":0},"subject":"WIP: Only generate relevant configuration in ml2_conf.ini","message":"WIP: Only generate relevant configuration in ml2_conf.ini\n\nOn compute nodes we end up with some neutron-server configuration in\nml2_conf.ini. This change adds logic to only generate it when necessary.\n\nTrivialFix\n\nChange-Id: Ic3c2c1c780973e46450c93bee83a59b72892ddd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3accdd087f23607f85fdd28bbe7e3c7e887c598f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3accdd087f23607f85fdd28bbe7e3c7e887c598f"}]},"branch":"refs/heads/master"},"b0dc8d65e1647d943584da713949892126c29252":{"kind":"REWORK","_number":5,"created":"2021-02-03 15:55:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/85/676985/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/676985/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/676985/5"}}},"commit":{"parents":[{"commit":"586dccc236d1466e114f371573b39e5a2129a995","subject":"Merge \"Remove extra space characters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/586dccc236d1466e114f371573b39e5a2129a995"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 16:37:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-03 15:55:12.000000000","tz":0},"subject":"WIP: Only generate relevant configuration in ml2_conf.ini","message":"WIP: Only generate relevant configuration in ml2_conf.ini\n\nOn compute nodes we end up with some neutron-server configuration in\nml2_conf.ini. This change adds logic to only generate it when necessary.\n\nTrivialFix\n\nChange-Id: Ic3c2c1c780973e46450c93bee83a59b72892ddd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0dc8d65e1647d943584da713949892126c29252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0dc8d65e1647d943584da713949892126c29252"}]},"branch":"refs/heads/master"},"7d3b61ce81120f1d069cbfe627e5b26e02271b64":{"kind":"REWORK","_number":6,"created":"2021-02-03 16:51:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/85/676985/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/676985/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/676985/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/676985/6"}}},"commit":{"parents":[{"commit":"586dccc236d1466e114f371573b39e5a2129a995","subject":"Merge \"Remove extra space characters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/586dccc236d1466e114f371573b39e5a2129a995"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 16:37:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-03 16:50:58.000000000","tz":0},"subject":"WIP: Only generate relevant configuration in ml2_conf.ini","message":"WIP: Only generate relevant configuration in ml2_conf.ini\n\nOn compute nodes we end up with some neutron-server configuration in\nml2_conf.ini. This change adds logic to only generate it when necessary.\n\nTrivialFix\n\nChange-Id: Ic3c2c1c780973e46450c93bee83a59b72892ddd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d3b61ce81120f1d069cbfe627e5b26e02271b64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d3b61ce81120f1d069cbfe627e5b26e02271b64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
