)]}'
{"id":"openstack%2Fkayobe~677933","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I8d2956f95bbc44b5a9e88e7569372048a62f12f5","project":"openstack/kayobe","branch":"stable/stein","topic":"story/2005923-stable/stein","hashtags":[],"change_id":"I8d2956f95bbc44b5a9e88e7569372048a62f12f5","subject":"Update packages in virtualenvs","status":"MERGED","created":"2019-08-22 08:28:15.000000000","updated":"2019-08-27 18:43:27.000000000","submitted":"2019-08-27 18:41:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677933-1566931312945-6c2ca45c","meta_rev_id":"9a3f45f35c2b876334ddac623aea42f8686ecfe3","_number":677933,"virtual_id_number":677933,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2019-08-27 10:39:14.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-27 18:41:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-08-22 15:33:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-08-22 08:30:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-08-22 08:29:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-08-22 15:33:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-08-23 09:15:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 08:29:55.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-22 15:33:59.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-08-23 09:15:01.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-08-27 18:41:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f6322bdb4757ab5faedcf034a91cec8c6f6beff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-22 08:28:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"307bb14a53ebb8d47c4345f7387f674669ee90ea","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-22 08:29:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1408c502f9cb6d947f6b5170578ba9d4e9fe9517","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-22 08:30:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a4c0e9863ac3ef2de67d082dd5bf365f31a585c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 10:26:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0af84a919f014bd3bb42566011896569 : SUCCESS in 6m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6e8064718b2243d2899b140a2a6d7d08 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12c3d372f65b4f09a3784ddf55f9eb6a : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/888163addc30453f8d8920ec5cc54d08 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/65bfe86ff4f04301b2490920cd9b4f36 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86ce46093ab4e21a393b1fee0a2b2c8 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9944b4a57e4743a9ab8c73c96ec160de : SUCCESS in 4m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/142962cdec5e4edea66624b36677b02c : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6e85b00d9e2d423abad24606b8e7c06b : SUCCESS in 8m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f57e82112064ba1875650ec163b0840 : SUCCESS in 33m 49s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ef31d9dad12241d3ae759f94daad6a39 : SUCCESS in 1h 05m 17s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d0aa91e9d1f44eaabc6cc77c4f42ef08 : SUCCESS in 1h 51m 21s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/48847bba51d645d0bf9fa73e12e4e006 : SUCCESS in 21m 20s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e4e04583307944929e71dd143c73a6c1 : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c0eb1058dcc333114391e462f6569e7bb8f5950e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-08-22 15:33:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d1c48de80d74df760e117e0a6c992664f08281fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 15:34:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2386a6ba33881bb0caca1112bc32057b2056719f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 17:33:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d7a813b0b2a84176ac33cfabe7f3d62e : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29f3f1c4882d4c01bad43e7a1251a942 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81d4b7fd02c74a37b5f34209c0a158e7 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3241f3039f534309a47c244e0d2b19eb : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fc2d1c858b24726916758bf8e41a632 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7588bbb0b6994253b24eba95ae15277a : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e1ef4bc6444d4b6da6c1def14e6c0b11 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8618581e3c22479bbfc05bbf2410bb5d : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/761ff4c42bae475088911f43e3c90208 : SUCCESS in 46m 58s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/887da93e0ec2457c815737b42efacd12 : FAILURE in 10m 00s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4b145699c68044eba026c76500de26a7 : SUCCESS in 1h 52m 41s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/12510cfcdfc8492ca66fa65035a1f704 : FAILURE in 12m 07s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c2c8a79367ef4e90b7c89c16fccf3807 : SUCCESS in 29m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1c3bead14eb2d6917f3b0f4b38fa07d09ffed5ad","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-23 06:50:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7927fcced7a40679a3eafa3215392695551c6366","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 08:56:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd2ac26a02cc490c98aa16e371c94fe2 : SUCCESS in 5m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/48bcde5616d047c198f0157edc7e05ba : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4f8288dd1034c999b36eaf8343cc796 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bebfac1eeef745a6b695b220e5dffd64 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85fb7848657247969b6c82a451265a09 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa6918bbdf94d97bdcfe8a265d9b54b : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83eebc0a6cb44489a3be3cddd5b47b7f : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e82648cb93ed4d3b99acfbd955e753ad : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/45cc5bed535744d2991c08cfa4d16eef : SUCCESS in 8m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f588ae346e5e4209bba0b55aa7aa6c92 : SUCCESS in 44m 25s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e992c084620b4adb98e7714adb35438b : FAILURE in 1h 57m 13s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/54feddf20d9e482297d68a4df6180da4 : SUCCESS in 1h 53m 10s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/cfab6dcb973b4f6a8490ef63495dc6d1 : SUCCESS in 20m 01s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/52a9427ac1d44baea69a3835df64ddba : SUCCESS in 28m 10s","accounts_in_message":[],"_revision_number":1},{"id":"3b2576deea3547bb9b3d9b54b0321252980f87fe","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-23 09:15:01.000000000","message":"Patch Set 1: Workflow+1\n\nrecheck\n\nDownload failed, but seems fine.\n\nTASK [baremetal : Install yum packages] ****************************************\nfailed: [controller0] (item\u003ddocker-ce) \u003d\u003e {\"changed\": false, \"item\": \"docker-ce\", \"msg\": \"https://download.docker.com/linux/centos/7/x86_64/stable/Packages/containerd.io-1.2.6-3.3.el7.x86_64.rpm: [Errno 14] curl#7 - \\\"Failed to connect to 2600:9000:2137:4e00:3:db06:4200:93a1:","accounts_in_message":[],"_revision_number":1},{"id":"4a470a6d58bc3723f0d8c6d0a5b7bea80910710c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-08-23 11:05:48.000000000","message":"Patch Set 1:\n\nreheck: https://review.opendev.org/#/c/678082 now landed","accounts_in_message":[],"_revision_number":1},{"id":"9534e04fd090d6a8f8dbcbde3956d4f923a28007","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 11:29:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2254df6817c24a12ae4648b1936ec326 : SUCCESS in 4m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ae7198494851421fb93485db39a06a73 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0ed07ec649149edbb9298f973219c77 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/266b02344d424545ac21f3272f57ff27 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f203e55bf03e4c27a61134bc26fa9e9e : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04261b6147524487873967ed70bd9eeb : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b39f746c4144573b5dd206b1c66cdc7 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/944b3af069824fa9a30a3129e788aac0 : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7692aff7dc9641f28a28b1497c7fc121 : SUCCESS in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7387916ff4ec423db6cce412e6040b0a : SUCCESS in 45m 48s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9ce237e14dd94b3e93edc8c2c4ff8c85 : TIMED_OUT in 2h 02m 05s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/e26e249b4d594f5ab72b6dc05c15ce60 : SUCCESS in 1h 58m 14s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/35877a6bfe5348fba826b28520d4a775 : SUCCESS in 20m 50s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/bc9d6d96fb1b4aa09a18439b108e47c2 : SUCCESS in 27m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b351bc8d7cb0af4d8ef3167fbd669eef8f1ea9e8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-23 11:34:46.000000000","message":"Patch Set 1:\n\nrecheck\n\ntimeout","accounts_in_message":[],"_revision_number":1},{"id":"a2060e3e6559d72e736cac05fba55a3a657d3233","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 13:39:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c05a6642ea3e4f24ac792a1d8a080db1 : SUCCESS in 7m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1e5ed172b0e24c43804719ba94881146 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91f18db05a214169ba367aab38832481 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a5bd4e2c7784c86ade3bfee9875c427 : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/376ecd0bbb3247fd971dbada1523f2ce : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b025671a634f8d944d5c6b4a0fd922 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0ee9928cf894cc4b3aa546164532e22 : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c248e142c52417194978873af5f0fc3 : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0395ccb971424ef481443b7975a3b5e2 : SUCCESS in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2f5a47f3d95247df8b808b60efa34dbd : SUCCESS in 34m 33s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/2702b87c51fb41ba82310cda16f8dd59 : TIMED_OUT in 2h 02m 58s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/19d9b7c5a97943ad8344fe17649b6be0 : SUCCESS in 1h 29m 02s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1cc140da7b574ef2b74afad95622d363 : SUCCESS in 18m 54s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/2ae25885e64b4b27bd335fb6d52133c6 : SUCCESS in 26m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6ed74b2ac687d90a2e8c0dd2146d83290a7d013d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-23 17:07:28.000000000","message":"Patch Set 1:\n\ntimeout seems permanent","accounts_in_message":[],"_revision_number":1},{"id":"f4d05bf8d0a10cf8ba05faa50e2beb0c859e7ef3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-23 19:15:36.000000000","message":"Patch Set 1:\n\nrecheck\n\ngood luck","accounts_in_message":[],"_revision_number":1},{"id":"c5961ca715315585e67af36a28af17faec7be985","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 21:04:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de16a146df6a42f482cd7c5307ef7c4a : SUCCESS in 5m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/88097c749a8843959ff14d43f458029f : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a83cce946c14e12a883439121d9c4c9 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/330835d23c0347e0a072fb4dbf3ea314 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a4db2a4acc9c42929e21f579b8a873a6 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c4db1475b7c4e1db5e3800aead833c8 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bd65fe94a8d4533ad48efa54e7932ae : SUCCESS in 4m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b7e3a3ed7db54061bf4505b31d13b1e9 : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/78b8765a2e6f4e46935837416f510abf : SUCCESS in 9m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0105c1b896934d81adb2b2b87b450706 : SUCCESS in 45m 07s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d7c6f5f767dc47728dcc490b5c5ba6d9 : SUCCESS in 1h 05m 36s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d55a9201ecc94873a354adcf5e51aee0 : SUCCESS in 1h 42m 35s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c7d74e0292df4460b682a97844b49960 : SUCCESS in 24m 03s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ef4555c495684dab9b117ecb0ce26f51 : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":1},{"id":"965556d784bc6a1ce72d431777c0cd4cd407d345","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 21:05:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84e3da1f7498f11b9c3b35020011780687973f53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 23:12:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e1efd5bb05114583bcaac1a337d0e351 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9468f785836c462e944991cbe8280789 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ff0835ebb864e75aac11c2166dcb182 : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/757a9b81cfac41c08979b7551c99fd92 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a886c463985440b4914c41523cb9ec9d : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/777438bd81194dfcbcb492468575cc3e : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b4105ea57f494b6cb6a9110caa506e13 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a839726118740f782c4359dfbf86ae7 : SUCCESS in 8m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d7ff5b87a46e4cd5a5d8d8958b6114f9 : SUCCESS in 43m 30s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/fc6fe464d79c49148a01126bc279bee3 : TIMED_OUT in 2h 03m 04s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/77a7b8f542fc4ce6ae42a2240ba8ca07 : SUCCESS in 1h 49m 18s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ddd6fb761ec349a7b5cae03b385a4ef0 : SUCCESS in 20m 17s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3bedb13d17ec448cb7bc996e6147289e : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":1},{"id":"bdac01bd164f4ce2bb9448d73a3749228c31934e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-27 10:39:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d8856fb0c77d78044637fac9abeedffca9400a43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 12:37:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59eac836114b4cf7b1cb766dc605ff1d : SUCCESS in 5m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cf328e29dde547808e6c9e7115210915 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb1b4499cd5d41bba6fd68874968bb8f : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a75faea5a33a47758f35ffe508282882 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95a7db688aef4c2c9519ce4bf1e811c8 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a0a54d295a491f840f9449c8cf97b2 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c85ab633972b4eaa9dd8688a08fd9eae : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f07b1b4c1b147e396997feacda30c89 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d0888d0c2764beca0b38272e4042b47 : SUCCESS in 9m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08f51d9182684cbb95e03f9daeac1587 : SUCCESS in 45m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8934217022ab4d29a6a597eb320d109c : SUCCESS in 1h 05m 37s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/bd623247088647888be60e220ba18444 : SUCCESS in 1h 48m 17s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/5bca971b073d44db895b04bbddfd20b6 : SUCCESS in 21m 24s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8b72753435fd4dd69c82a4fdb20645f4 : SUCCESS in 27m 06s","accounts_in_message":[],"_revision_number":1},{"id":"039a0b279aaa8cddb5ea36b0ca2f078776a51eb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 12:37:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"050335d38c968a71d2ab4aa45084a0e8f1c01254","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 18:41:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/accd7e4f93ad413ab97600764127720a : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a0c0e52c01a4dc29fa8da42249ca4fd : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cdd7a0870a24d5cb785ad32466ae78d : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5265a263dfa14286a3382fa29f56251a : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de293eee4a6f472ba2cc39374b51f3eb : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bdd965630894438bdf936cd988b625d : SUCCESS in 4m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5feaee4f59354caab5b0b8ab463c7c19 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9bcec6258b574c8caa0c08c240e4ea2f : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/073c0b1a6ea647feb55fb0b6398c900c : SUCCESS in 44m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/a2c433ca61a144bda68d95c05cc7722b : SUCCESS in 1h 07m 00s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/6f1f27e360c8495fb8032906f186e471 : SUCCESS in 1h 30m 58s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c7bc976594024bcdacce52462ade34b6 : SUCCESS in 22m 13s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/389877755fff45bd88bc4fb36eb8eeb8 : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a8c599a490319f36a800b98c6ddba0760c86a224","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 18:41:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0d459b01cf2ba102e1abc958bfdb4e8d97daa434","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 18:43:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb69af2d72434a6e9d1b40d0870d8929 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75ba4458a06a6880d2b156d0eed14b3848af0dd6","revisions":{"75ba4458a06a6880d2b156d0eed14b3848af0dd6":{"kind":"REWORK","_number":1,"created":"2019-08-22 08:28:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/33/677933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/677933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/677933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/677933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/677933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/677933/1"}}},"commit":{"parents":[{"commit":"5f0146b46e457b7f55a7475006a40ff0354d6ba9","subject":"Update .gitreview for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f0146b46e457b7f55a7475006a40ff0354d6ba9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-06-20 16:10:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-22 08:28:15.000000000","tz":0},"subject":"Update packages in virtualenvs","message":"Update packages in virtualenvs\n\nKayobe uses a number of virtual environments on the remote hosts for\npython dependencies such as shade, python-openstackclient, docker, etc.\nBy default these are stored in /opt/kayobe/venvs/. Typically we do not\nprovide version restrictions when installing these packages, so over the\ncourse of time they may become stale and incompatible.\n\nThis change installs the latest version of packages allowed by OpenStack\nupper constraints.\n\nIt also adds a new variable, \u0027pip_upper_constraints_file\u0027, to set the\nupper constraints file. The existing variable\n\u0027kolla_upper_constraints_file\u0027 now defaults to the value of\n\u0027pip_upper_constraints_file\u0027.\n\nChange-Id: I8d2956f95bbc44b5a9e88e7569372048a62f12f5\nStory: 2005923\nTask: 34193\n(cherry picked from commit ddfd6b6202a2a97d7e0d3372624f959b922828b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75ba4458a06a6880d2b156d0eed14b3848af0dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/75ba4458a06a6880d2b156d0eed14b3848af0dd6"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
