)]}'
{"id":"openstack%2Fkayobe~679129","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Ie59f0c729381e5bea112ab7f5d0c738a9488b772","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"Ie59f0c729381e5bea112ab7f5d0c738a9488b772","subject":"Set provisioning and cleaning networks as non-shared","status":"MERGED","created":"2019-08-28 19:34:54.000000000","updated":"2019-08-29 11:28:52.000000000","submitted":"2019-08-29 11:27:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679129-1567078038894-1d71bfeb","meta_rev_id":"cf957f58ff20bf72689fed780e6a44c1c6b151aa","_number":679129,"virtual_id_number":679129,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-08-29 11:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-08-29 09:46:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-29 09:53:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-29 09:53:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-29 09:46:19.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-08-29 09:53:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-08-29 11:27:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6166bac64fb5429259be54be95293e5dbc89e1b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-28 19:34:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3ff6e3510d35b4c9c91e1ffd058418fed50d0033","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 21:40:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ee4d35146294a7dafaebc0bc3ca8a3a : SUCCESS in 5m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/93d958cd72bb44bdae6905706b5dfe56 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b75a0557e4384f5ca82f86204d9a64e6 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99e3d0095abe487db7883b9a80f39054 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/68095b2a678a47f7824420b861d09982 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f9f4f9d2bd54ca2963833c607d78858 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/227303080ce44d68ba7a59338f3f4c26 : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/48facc487d474bb1a2e0e1ef4634ead9 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d29b0cba875e4d4b80f3fa43a4d74d9c : SUCCESS in 9m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/edf81a3e038d4a5fbffceef9cb0bb3e5 : SUCCESS in 46m 22s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7c34208aa86b46da9136415a18e5bc49 : TIMED_OUT in 2h 02m 24s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/26f8bbe2287a4c419c60110758fddf30 : SUCCESS in 1h 45m 24s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b7b3538aba4e4bbf88627aff4827f92b : SUCCESS in 21m 28s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b558addc4c1949cbbb0e8d13b5b770fe : SUCCESS in 35m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9dc1925dba2b45ac25993b8314a16a8264a24643","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-29 06:47:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d623035aef2b2d0af7b7f6cee511fdc7d301d73e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 07:52:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/250a5a7ee94d4e469f77bf48513af4fc : SUCCESS in 7m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5118fa7adce644ac874f4ce87930cc01 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b52e0aaa8ae445e5b02daac085140daf : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1333298c2c04607a2f43031c6eb9aa8 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e98c848572f2495eb952e0c3b519bc67 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8263f9b06f68419a9f8ff86d43725dcf : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f6849eb7aa24f37ae59f549eb37e56a : SUCCESS in 4m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8331f69eb68c4bb1a3a5d2e2520593e4 : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/599afaa4081a4c81aa359ba4883ff8cd : SUCCESS in 9m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2e26c5c53e7f464a9e191bb1d2ed282a : SUCCESS in 37m 35s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9b8f9f8c70044360af6bdeee359f6915 : SUCCESS in 1h 02m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b1fde9ba57ff4995a6ddf739ef5a01b8 : RETRY_LIMIT in 3m 15s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b424799ff4004ce0817b9a1409c22bc8 : SUCCESS in 21m 36s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/0fabd9b29a544ef2ba8ce6a50dd205dd : SUCCESS in 28m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2bb702c73c9340dfa02a6e5f9d0757bf80095c67","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-29 07:55:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b53bf7ecacd53ecfd3cb13e1c97e3fda1d9dc1db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 09:42:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0eb15367e6c40ec90c55190e218cca2 : SUCCESS in 3m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ecf6651b4262459f9a92a50e43a4d5a4 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/838d4490ad204008a4a082bff5a6c6c8 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b133a74d582414c9053693dceb6ca12 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf6312f4305247abaec40462274b3db0 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5e99cbc67604302a0961fc0da2a8fed : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bb85832cdec41c3822d528142b80c57 : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bd635bc6d5174a98a38b7b6acb440014 : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ff3def638d8d48f1b0ca3f2d01f9601e : SUCCESS in 11m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/651e8387a01b4b04b4f6f152cc27f14b : SUCCESS in 45m 12s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/73f7769e6fdf43bb9214d341682d3180 : SUCCESS in 1h 04m 59s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/9a43a613a67c40edaef5594b25e43afb : SUCCESS in 1h 42m 55s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0fbdc56119b34d7dae21fe31c5f41b7a : SUCCESS in 25m 39s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/fb3a9509ac5343758d4d0946ef10bb97 : SUCCESS in 27m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bb51656c41be61f4c989241d33144d43a60c221f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-08-29 09:46:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"13f0d016b67bf228efc89fbc54dd2ddf474858d8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-29 09:53:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ebacbb01224e5b415575bb207aeab161bcd724f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 09:53:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"432b244a510dcb4edee43768230928e1bae4215b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 11:27:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e9dc5003f55e4df2add23fdabaed9375 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2551261fa39431286102d26d8e45d77 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a02741880e74a0ba9d5f9223b9382f4 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f78ee0a145594db6ab19ad7551e07ae9 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e1efbd7a28e4d5fb9776964d22750ca : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1b7f2e6c65f4a1f83d2b8adb979b4ed : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e3dda98bc92a4e4c98928d9fa327f08a : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d8f78419abf44bc19a598b4daaa6f842 : SUCCESS in 8m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/39f6f660881f44a8bbce31364d37c349 : SUCCESS in 45m 55s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d8bcd9d649cd46b682e691df54643b88 : SUCCESS in 1h 01m 05s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1e11d8bac65c49e2a5161c160f158b8c : SUCCESS in 1h 32m 09s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/22f1b2e876cf410a9a1a963942890164 : SUCCESS in 21m 46s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a80bd38593234fdc99c1e2c078a7dd3c : SUCCESS in 24m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b3372de91df13893df952f29a705f01e3c955798","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 11:27:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cf957f58ff20bf72689fed780e6a44c1c6b151aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 11:28:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3af8f36e31b4ceca17847ba3a8059ce : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3a6c24bc492abfcfba0459a094b390508840f85","revisions":{"d3a6c24bc492abfcfba0459a094b390508840f85":{"kind":"REWORK","_number":1,"created":"2019-08-28 19:34:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/29/679129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/679129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/679129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/679129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/679129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/679129/1"}}},"commit":{"parents":[{"commit":"0da672abfd591af17f6e8f4d9f59952ae6e635f9","subject":"Add support for Arista switches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0da672abfd591af17f6e8f4d9f59952ae6e635f9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 11:08:25.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-28 19:34:54.000000000","tz":0},"subject":"Set provisioning and cleaning networks as non-shared","message":"Set provisioning and cleaning networks as non-shared\n\nModifies provisioning and cleaning networks in multi-tenant ironic\nenvironments to be non-shared. Flat networks remain shared.\n\nTo apply the change to an existing environment, run \u0027kayobe overcloud\npost configure\u0027.\n\nChange-Id: Ie59f0c729381e5bea112ab7f5d0c738a9488b772\nStory: 2006409\nTask: 36295\n(cherry picked from commit f91cf43876afa967bcf26c133c7973f84f02341f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3a6c24bc492abfcfba0459a094b390508840f85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3a6c24bc492abfcfba0459a094b390508840f85"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
