)]}'
{"id":"openstack%2Fkayobe~681689","triplet_id":"openstack%2Fkayobe~master~I6f56ba2d6e3c5e5f91e83bd12d8a09c04845cd62","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I6f56ba2d6e3c5e5f91e83bd12d8a09c04845cd62","subject":"CI: Remove rocky pin from molecule requirements","status":"MERGED","created":"2019-09-12 09:46:40.000000000","updated":"2019-10-02 17:47:44.000000000","submitted":"2019-10-02 17:44:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681689-1570038298522-fe29816b","meta_rev_id":"163b5b2381667e8bf01bb4a77a078bc9ee813a92","_number":681689,"virtual_id_number":681689,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-02 17:44:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-10-02 15:36:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-18 15:47:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-10-02 15:36:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-09-18 15:46:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 15:47:47.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-10-02 15:36:22.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-10-02 17:44:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5e666f03ced56ec4e29fcaad500a8362d28dde9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-12 09:46:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a67483abf6a0b1144767e23c7707172425ba03e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 11:30:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a017dcbb21284c61be747cde30c62cc1 : SUCCESS in 4m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c72f5961d8164e0fa741f14fda1ad06b : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1306495642544359b6c70979524bbe3 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ff145dad6e486aab177ac13ee0b1ce : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa9400b3c79245dbb0a5c5022521a7a2 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a68308ab5234ac7845372440ed1b98e : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/322a92107e4c4ac89ef1da0bbb9367cc : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e93ab8ba73724afd94c6119337dbaa25 : TIMED_OUT in 30m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad03830fb36c4f22aedf666eb86c7fb4 : TIMED_OUT in 1h 00m 54s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f71e8ced58b34f83bfd9a51e9baa8eaf : SUCCESS in 1h 10m 52s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/6f899da6e37940abaefa3cbc31bf4640 : SUCCESS in 1h 37m 32s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9b8ebb03de484624aa4413f90c914d13 : SUCCESS in 23m 42s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/bdfd0bed07f74f8fabeda394709fb241 : SUCCESS in 28m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7e36dede07b73d2026a71ec797f5347ef98cb76f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-12 15:06:27.000000000","message":"Patch Set 1:\n\nrecheck\n\nCI weirdness","accounts_in_message":[],"_revision_number":1},{"id":"4ce9a2520b2d82a34c88f053a91eaf6daa0c2ee2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 16:59:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b20ce19a57f8495e8e85031183e25cdb : SUCCESS in 3m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/019e8f0647244c5aa58b740fc1be2501 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c8a31aaa0214e028dedbb509d284908 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97b5d0ec0ed1414893a19f9daf45a599 : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4502d929a61347c896fe201bc5bfd7c4 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/732b8af668e54d21895023e21105aaec : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb88e52555454a5ca5a79735ff8191cd : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71cfb71f2514411491ce26c56b4ae300 : SUCCESS in 9m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dca5db51675845d883629df939a06a13 : SUCCESS in 49m 21s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/4e94af40437b45d09cc19d0b9abb28af : SUCCESS in 1h 08m 36s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/ab431957d18044efabc29b1a7c57c5f1 : SUCCESS in 1h 47m 23s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f30a94ff6f4347acb00b75f92fd0301e : SUCCESS in 23m 41s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/170ba95174bd43cf8e4138adc29104e9 : SUCCESS in 34m 27s","accounts_in_message":[],"_revision_number":1},{"id":"aa35c3afc33e7aa90a52e7fcdece447c6ba5a3e3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-18 15:46:40.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"2bd186e4a1cf242288aace3dafafdb1d237b37cd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-18 15:47:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9ff60d0713cb7b3c5f1d80b5316d658ed8394c37","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-02 15:36:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"061e5f09118d4ed05d65d8d51c23c5763497596b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 15:36:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3382d209303301b4e29332be8a4b435c66c1ce6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:44:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/916f00a9a36d48edb79be0a601523b9a : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3afa4ff700224d3b9851da08cc978ff0 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7bd9905000741c8a4df1c24bf33d978 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7d75cd4c7204ce6a3d76a6b59025eb0 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dfd6b01b7624a918a788f7c58ffc581 : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c82c157dba84d4fba0be84d43a30359 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9c3bfeccd5ce4255bab233cb75d56f53 : SUCCESS in 9m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e736f48690748c2a5ebc6c1d38c319f : SUCCESS in 44m 21s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/91986d55eb2f4392b9a498444b7c8765 : SUCCESS in 1h 08m 07s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/0fa2b48e89b64e0e9179d1490733fda2 : SUCCESS in 2h 01m 33s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/85fce80c624f4cd4a3835f2356c58a68 : SUCCESS in 21m 57s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/acff316c4af442fe860bdc71f4aba78d : SUCCESS in 38m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b9cbc6e1b89a0f77272762c88412637010c1634d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:44:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"163b5b2381667e8bf01bb4a77a078bc9ee813a92","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:47:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feffb45463ba49dd99c5f9a282bc9a16 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0765c09368f2d011221ae86b246e96074463bb89","revisions":{"0765c09368f2d011221ae86b246e96074463bb89":{"kind":"REWORK","_number":1,"created":"2019-09-12 09:46:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/89/681689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/681689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/681689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/681689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/681689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/681689/1"}}},"commit":{"parents":[{"commit":"fc0d0907f7975c9152bf2a34d6f87ce1d56b5817","subject":"Merge \"Fix typo in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc0d0907f7975c9152bf2a34d6f87ce1d56b5817"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-12 09:45:42.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-12 09:46:22.000000000","tz":60},"subject":"CI: Remove rocky pin from molecule requirements","message":"CI: Remove rocky pin from molecule requirements\n\nChange-Id: I6f56ba2d6e3c5e5f91e83bd12d8a09c04845cd62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0765c09368f2d011221ae86b246e96074463bb89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0765c09368f2d011221ae86b246e96074463bb89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
