)]}'
{"id":"openstack%2Fkayobe~681700","triplet_id":"openstack%2Fkayobe~master~If96437b73b9b5c58600ba1b004f53ee0c1f14398","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"If96437b73b9b5c58600ba1b004f53ee0c1f14398","subject":"Fix kayobe-overcloud-centos CI job flakiness","status":"MERGED","created":"2019-09-12 10:17:17.000000000","updated":"2019-09-18 08:18:57.000000000","submitted":"2019-09-17 21:39:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681700-1568756354411-4f1731f4","meta_rev_id":"c3eacf16663d6469e2c6a3b934ad389bb3e61062","_number":681700,"virtual_id_number":681700,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-17 21:39:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2019-09-17 15:58:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-17 15:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-09-17 15:58:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 15:50:41.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-09-17 15:58:12.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-09-17 21:39:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f8e351858c8c8ce3d359caa26b1c1d23c87fc4ad","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-12 10:17:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"570f6a21fb56fd59dbe727c10d2effc48394059b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 10:53:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos-1 https://zuul.opendev.org/t/openstack/build/d27afdb97e874b9dbefb9595dcedc2ba : RETRY_LIMIT in 33m 53s\n- kayobe-overcloud-centos-2 https://zuul.opendev.org/t/openstack/build/452a7141b352406aaebac7284ea8c693 : RETRY_LIMIT in 33m 49s\n- kayobe-overcloud-centos-3 https://zuul.opendev.org/t/openstack/build/9562ce2154b74301b562be6625b207d2 : RETRY_LIMIT in 33m 46s\n- kayobe-overcloud-centos-4 https://zuul.opendev.org/t/openstack/build/fb316dfc31904273b2f0866b0734caf7 : RETRY_LIMIT in 24m 42s","accounts_in_message":[],"_revision_number":1},{"id":"200cabdb0a8a09826df3c09d1710941778629905","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-12 11:18:51.000000000","message":"Patch Set 1:\n\nrecheck\n\nCI weirdness","accounts_in_message":[],"_revision_number":1},{"id":"fb37d1457b0beeb811397327e001930159d6a0ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 12:46:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos-1 https://zuul.opendev.org/t/openstack/build/5c206c1bef0d4253a625fdb8667f7ea9 : SUCCESS in 1h 21m 25s\n- kayobe-overcloud-centos-2 https://zuul.opendev.org/t/openstack/build/6553142ab702446ca6530873d90457bb : POST_FAILURE in 1h 12m 30s\n- kayobe-overcloud-centos-3 https://zuul.opendev.org/t/openstack/build/a134af35b24a4e73b223708118f07aa8 : SUCCESS in 1h 03m 25s\n- kayobe-overcloud-centos-4 https://zuul.opendev.org/t/openstack/build/182df4c83f9b4622846a42a00e75ad76 : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":1},{"id":"000c2d2992ab9e9ddf6aa5f12d9236917696ba7f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-12 15:07:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d07dbe7659008f22cbd08b7f40a346e37a35abba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 17:22:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos-1 https://zuul.opendev.org/t/openstack/build/14e1ed130a944bc785e5dc6c0a5e8868 : TIMED_OUT in 2h 02m 00s\n- kayobe-overcloud-centos-2 https://zuul.opendev.org/t/openstack/build/c64b29ba0e034be0a963eae0f01032aa : TIMED_OUT in 2h 01m 48s\n- kayobe-overcloud-centos-3 https://zuul.opendev.org/t/openstack/build/e837ea1933c6474187b205f47df1604d : TIMED_OUT in 2h 02m 25s\n- kayobe-overcloud-centos-4 https://zuul.opendev.org/t/openstack/build/bf7b4bb107bf46b2adecde4aeb7f9d33 : SUCCESS in 1h 08m 34s","accounts_in_message":[],"_revision_number":1},{"id":"463b0855288e3e9e2c3badce938b757f15fdbfb6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-13 09:51:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5fa8591ef7e898de60a54ec1255e2ecbd58093f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-13 09:57:15.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d8b5ccc58f15ede71ac5be22c9a099799fdbc57a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 11:09:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos-1 https://zuul.opendev.org/t/openstack/build/bb5697387a734ce8b18d2405d823d726 : SUCCESS in 1h 05m 41s\n- kayobe-overcloud-centos-2 https://zuul.opendev.org/t/openstack/build/f570693fde4f4e2c91398bdf518cbd47 : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-centos-3 https://zuul.opendev.org/t/openstack/build/dde322f5bc1e4b64b13d17d2b239c5f7 : SUCCESS in 1h 06m 45s\n- kayobe-overcloud-centos-4 https://zuul.opendev.org/t/openstack/build/c241fda3cf2645dcb8b174386c07016d : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":3},{"id":"ec1ed6250355cab1769520f951dae35d4d14cbf6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-13 13:07:27.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"ed049379d2fc3f57a4239609058bc09335a70fb0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-13 13:09:39.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"b2fee00c3aa9b4e0b14780be1fa966d89fc18d0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 14:30:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos-1 https://zuul.opendev.org/t/openstack/build/2b710eeeee524068a7a522260c5171e0 : SUCCESS in 1h 04m 57s\n- kayobe-overcloud-centos-2 https://zuul.opendev.org/t/openstack/build/7d504d1fa65145969327603d3765927d : SUCCESS in 1h 09m 21s\n- kayobe-overcloud-centos-3 https://zuul.opendev.org/t/openstack/build/c9a96fa7b80f43ce9edcd83d8de065bf : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-centos-4 https://zuul.opendev.org/t/openstack/build/0b81456e2737423a8f9dfdaa4e73029d : SUCCESS in 1h 05m 58s","accounts_in_message":[],"_revision_number":5},{"id":"fde337b7e3cdfdafff432c221aab65c4d2b8337e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-14 16:58:53.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"f5654909ea852a53f46e9d952c2ebd56d6ee1203","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 19:19:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos-1 https://zuul.opendev.org/t/openstack/build/46b9346f50fd482d8cb7e5fc328644e3 : TIMED_OUT in 2h 01m 56s\n- kayobe-overcloud-centos-2 https://zuul.opendev.org/t/openstack/build/a69a852ea4f3421a8cb1e5164fb5de48 : SUCCESS in 1h 08m 18s\n- kayobe-overcloud-centos-3 https://zuul.opendev.org/t/openstack/build/8a42e29433614f16ad45aca75dda2d55 : TIMED_OUT in 2h 02m 07s\n- kayobe-overcloud-centos-4 https://zuul.opendev.org/t/openstack/build/6d77efd8824b44ee841387fe8f9c0c8d : TIMED_OUT in 2h 02m 02s","accounts_in_message":[],"_revision_number":5},{"id":"08122add7712ebc834e8b7e3b0255dc8ff1f13e4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-17 11:28:18.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b37328d610ecd8b0f9b1a3b67e391061cb946374","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 12:00:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"49f22713d80daf9f34d00dfbcdebe42a9db0d3a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 13:27:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos-1 https://zuul.opendev.org/t/openstack/build/87e9d2852cf14880bb24dd70de4227c3 : SUCCESS in 1h 08m 57s\n- kayobe-overcloud-centos-2 https://zuul.opendev.org/t/openstack/build/4f5275a137aa49658ba234c4cf3f863e : SUCCESS in 1h 12m 07s\n- kayobe-overcloud-centos-3 https://zuul.opendev.org/t/openstack/build/7e96359298aa435ea4e93f1efe6b6643 : SUCCESS in 1h 13m 11s\n- kayobe-overcloud-centos-4 https://zuul.opendev.org/t/openstack/build/78003e3735d34c0eafdc4d4130963f9b : SUCCESS in 1h 10m 06s","accounts_in_message":[],"_revision_number":6},{"id":"f8ae1f7fe963cecc5193b18325c314b64963ddf0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 13:39:00.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"8acd1860c6ad1c02a0fd9565a518e47b4de6480c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 15:03:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos-1 https://zuul.opendev.org/t/openstack/build/a4b91b0c98104c2487045598e0627258 : SUCCESS in 1h 19m 01s\n- kayobe-overcloud-centos-2 https://zuul.opendev.org/t/openstack/build/c53c88e13de94ef186f74da974278c50 : SUCCESS in 1h 19m 22s\n- kayobe-overcloud-centos-3 https://zuul.opendev.org/t/openstack/build/3afae593b4f341ecbe109d07e2e31eb3 : SUCCESS in 1h 11m 35s\n- kayobe-overcloud-centos-4 https://zuul.opendev.org/t/openstack/build/8007c69150d942008a66cc479a88eb99 : SUCCESS in 1h 08m 11s","accounts_in_message":[],"_revision_number":6},{"id":"568fe6c83da1a6cbc5fc2c620f39f2396a2958f2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 15:45:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4f6afc705884c40044d88bfeb46cbced7d8eb9f2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-17 15:50:41.000000000","message":"Patch Set 7: Code-Review+2\n\nGreat job finally finding a solution!","accounts_in_message":[],"_revision_number":7},{"id":"66e048f07b1b532fb864a1005a7e252e6ad15471","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-09-17 15:58:12.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nIndeed, good find!","accounts_in_message":[],"_revision_number":7},{"id":"bc07280eb28a8930575bbd12d69379d8d09dccd6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 17:45:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/942431213f324b90b284d1257bcc3c20 : SUCCESS in 3m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b91f38d7e2b9421d9a1926e5dee06637 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e09967a39b40403fa30719549d9c2bbe : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfc8d60184a04d848865f4ea8dbc4ebd : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae36f24d62fe46b880d20dd703dfaea1 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae3027ac1de3476191dba9c50a6cabe4 : SUCCESS in 6m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/165db0c4f07d40759d87b4a798880d8a : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/45f65b22ce144ced8cfef5294b52304a : SUCCESS in 10m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1122d6d56ed7499ab81339219fd1d930 : SUCCESS in 50m 14s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/02893057f37f474e96ffa7c3a0c8b846 : SUCCESS in 1h 39m 51s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/dae3fb3ead784bf8aaba57aec71261db : SUCCESS in 1h 54m 04s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/13754b7b29a6469988232b089142fa1b : SUCCESS in 24m 42s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/cf5f1f2701ce4f4bb5f421a93bce1c2a : SUCCESS in 29m 06s","accounts_in_message":[],"_revision_number":7},{"id":"05f4d72339bba5ed061b9dd0a966a7f439e07d4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 17:45:24.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"b60fd01e8b8a9da1adc0341fcfadc6bbf6f4fbd6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 21:39:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"b825c957b7caaec7cca4d01b73dc828a19ba2301","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 21:39:14.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f9b7ec8273354bd4be973942e3a3a81d : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6f33c84da154d079cf6fdbdf6db1d6d : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/409956fb90da4d8abafdcd5842d7ca41 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b808228df3154239a7067e5eec46d90c : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fed2c89fdc14d5c851b83ca427262f5 : SUCCESS in 5m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c5cd3b867dc64ce1b993db9664819c17 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/98be9c9623b5485f921e51a8d359ba13 : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/68c8dfe1059846c8bdcaf94e8a4da6a2 : SUCCESS in 46m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/050d59ebc89d4c26a9b52fcd69d95be6 : SUCCESS in 1h 13m 10s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/109be5c7a1544c87b1a23c73d2a243a9 : SUCCESS in 1h 59m 39s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/a7d9f52dcfc247b2b0a7efb592ed2845 : SUCCESS in 22m 22s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/109d12a5b7e84eab833f3142dc125f64 : SUCCESS in 25m 54s","accounts_in_message":[],"_revision_number":7},{"id":"ec4f9a004a74aa4c743b298667f2f5118048783e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 21:40:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf45c50de16e4949b06b2e83bedf1e33 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":7},{"id":"c3eacf16663d6469e2c6a3b934ad389bb3e61062","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-18 08:18:57.000000000","message":"Patch Set 7: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit ad474b9d4fe23fea99e8b5cca0c2fa1aa33cd04c","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4b180502a5bcfb9cee5e0c2e267d2dbc53d4b358","revisions":{"53d6739657ab5a6207510e16a10ea749e6f3863e":{"kind":"REWORK","_number":1,"created":"2019-09-12 10:17:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/681700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/681700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/681700/1"}}},"commit":{"parents":[{"commit":"fc0d0907f7975c9152bf2a34d6f87ce1d56b5817","subject":"Merge \"Fix typo in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc0d0907f7975c9152bf2a34d6f87ce1d56b5817"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-12 10:16:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-12 10:16:22.000000000","tz":60},"subject":"WIP: Fix kayobe-overcloud-centos CI job flakiness","message":"WIP: Fix kayobe-overcloud-centos CI job flakiness\n\nChange-Id: If96437b73b9b5c58600ba1b004f53ee0c1f14398\nStory: 2006534\nTask: 36590\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53d6739657ab5a6207510e16a10ea749e6f3863e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53d6739657ab5a6207510e16a10ea749e6f3863e"}]},"branch":"refs/heads/master"},"8b755626b9e6ad936cd80cb24e74cbb994a6fd12":{"kind":"REWORK","_number":2,"created":"2019-09-13 09:51:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/681700/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/681700/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/681700/2"}}},"commit":{"parents":[{"commit":"fc0d0907f7975c9152bf2a34d6f87ce1d56b5817","subject":"Merge \"Fix typo in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc0d0907f7975c9152bf2a34d6f87ce1d56b5817"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-12 10:16:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-13 09:51:05.000000000","tz":0},"subject":"WIP: Fix kayobe-overcloud-centos CI job flakiness","message":"WIP: Fix kayobe-overcloud-centos CI job flakiness\n\nChange-Id: If96437b73b9b5c58600ba1b004f53ee0c1f14398\nStory: 2006534\nTask: 36590\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b755626b9e6ad936cd80cb24e74cbb994a6fd12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b755626b9e6ad936cd80cb24e74cbb994a6fd12"}]},"branch":"refs/heads/master"},"1ae236029676ff3275bb73f56654a2f496917e78":{"kind":"REWORK","_number":3,"created":"2019-09-13 09:57:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/681700/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/681700/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/681700/3"}}},"commit":{"parents":[{"commit":"fc0d0907f7975c9152bf2a34d6f87ce1d56b5817","subject":"Merge \"Fix typo in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc0d0907f7975c9152bf2a34d6f87ce1d56b5817"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-12 10:16:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-13 09:57:11.000000000","tz":0},"subject":"WIP: Fix kayobe-overcloud-centos CI job flakiness","message":"WIP: Fix kayobe-overcloud-centos CI job flakiness\n\nChange-Id: If96437b73b9b5c58600ba1b004f53ee0c1f14398\nStory: 2006534\nTask: 36590\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ae236029676ff3275bb73f56654a2f496917e78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ae236029676ff3275bb73f56654a2f496917e78"}]},"branch":"refs/heads/master"},"bdbb1596c6da87549c3747381394935c6c8df8a6":{"kind":"REWORK","_number":4,"created":"2019-09-13 13:07:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/681700/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/681700/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/681700/4"}}},"commit":{"parents":[{"commit":"fc0d0907f7975c9152bf2a34d6f87ce1d56b5817","subject":"Merge \"Fix typo in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc0d0907f7975c9152bf2a34d6f87ce1d56b5817"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-12 10:16:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-13 13:07:24.000000000","tz":0},"subject":"WIP: Fix kayobe-overcloud-centos CI job flakiness","message":"WIP: Fix kayobe-overcloud-centos CI job flakiness\n\nChange-Id: If96437b73b9b5c58600ba1b004f53ee0c1f14398\nStory: 2006534\nTask: 36590\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bdbb1596c6da87549c3747381394935c6c8df8a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bdbb1596c6da87549c3747381394935c6c8df8a6"}]},"branch":"refs/heads/master"},"29efcaf650d6b4190ceeeaf16d71ff79bdbc51f2":{"kind":"REWORK","_number":5,"created":"2019-09-13 13:09:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/681700/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/681700/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/681700/5"}}},"commit":{"parents":[{"commit":"fc0d0907f7975c9152bf2a34d6f87ce1d56b5817","subject":"Merge \"Fix typo in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc0d0907f7975c9152bf2a34d6f87ce1d56b5817"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-12 10:16:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-13 13:09:35.000000000","tz":0},"subject":"WIP: Fix kayobe-overcloud-centos CI job flakiness","message":"WIP: Fix kayobe-overcloud-centos CI job flakiness\n\nChange-Id: If96437b73b9b5c58600ba1b004f53ee0c1f14398\nStory: 2006534\nTask: 36590\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/29efcaf650d6b4190ceeeaf16d71ff79bdbc51f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/29efcaf650d6b4190ceeeaf16d71ff79bdbc51f2"}]},"branch":"refs/heads/master"},"73854aa01c0e81d14a502b3f7446332f53d01aa5":{"kind":"REWORK","_number":6,"created":"2019-09-17 12:00:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/681700/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/681700/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/681700/6"}}},"commit":{"parents":[{"commit":"7321205c0678b42d41d774a1ff14bec9a4a9c8e1","subject":"Update docs after repo rename x/ to openstack/","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7321205c0678b42d41d774a1ff14bec9a4a9c8e1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-12 10:16:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-17 11:59:53.000000000","tz":60},"subject":"WIP: Fix kayobe-overcloud-centos CI job flakiness","message":"WIP: Fix kayobe-overcloud-centos CI job flakiness\n\nChange-Id: If96437b73b9b5c58600ba1b004f53ee0c1f14398\nStory: 2006534\nTask: 36590\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73854aa01c0e81d14a502b3f7446332f53d01aa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73854aa01c0e81d14a502b3f7446332f53d01aa5"}]},"branch":"refs/heads/master"},"4b180502a5bcfb9cee5e0c2e267d2dbc53d4b358":{"kind":"REWORK","_number":7,"created":"2019-09-17 15:45:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/681700/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/681700/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/681700/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/681700/7"}}},"commit":{"parents":[{"commit":"7321205c0678b42d41d774a1ff14bec9a4a9c8e1","subject":"Update docs after repo rename x/ to openstack/","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7321205c0678b42d41d774a1ff14bec9a4a9c8e1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-12 10:16:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-17 15:42:09.000000000","tz":60},"subject":"Fix kayobe-overcloud-centos CI job flakiness","message":"Fix kayobe-overcloud-centos CI job flakiness\n\nDocker CE has added a default DROP policy to the FORWARD chain.  When\nnova-compute runs on the controller, kolla ansible sets the\nnet.bridge.bridge-nf-call-iptables sysctl to 1, which causes iptables to\nprocess frames forwarded across bridges.\n\nCurrently, the kayobe-overcloud-centos job is failing quite frequently\nwith timeouts when deploying bare metal compute. Experimentation with\niptables hasn\u0027t revealed why this only happens sometimes, or exactly\nwhat traffic is being blocked, but opening up the firewall does seem to\nfix the issue. We won\u0027t see this in production since control and compute\nservices are on separate hosts.\n\nThis change updates the iptables configuration used in CI to forward all\nframes on the bridge, breth1.\n\nChange-Id: If96437b73b9b5c58600ba1b004f53ee0c1f14398\nStory: 2006534\nTask: 36590\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b180502a5bcfb9cee5e0c2e267d2dbc53d4b358"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b180502a5bcfb9cee5e0c2e267d2dbc53d4b358"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
