)]}'
{"id":"openstack%2Fkayobe~682371","triplet_id":"openstack%2Fkayobe~master~I8efee18cf28344a6af44acbe1ca1d4470f7258fe","project":"openstack/kayobe","branch":"master","topic":"fix-rename","hashtags":[],"change_id":"I8efee18cf28344a6af44acbe1ca1d4470f7258fe","subject":"Fix after x/kayobe rename","status":"MERGED","created":"2019-09-16 14:26:54.000000000","updated":"2019-09-16 19:17:31.000000000","submitted":"2019-09-16 19:14:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682371-1568661271567-4f1731f4","meta_rev_id":"c3d0bd57c82ac3642247031e13a4db1acb53cc14","_number":682371,"virtual_id_number":682371,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-16 15:30:04.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-16 19:14:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-16 15:58:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-09-16 15:58:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 15:58:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-16 19:14:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"811c3c686d5f852c0f363133673a835db2a1c2ec","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-16 14:26:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9392c2a68c2bc84820590013a6a6810ed60535aa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-16 15:30:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"884c2dac13cce42e57db270ddea0295b82332f35","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-16 15:58:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1e8f8460312d8184af16574ffc4638e28386a26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 17:27:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2ba0cc00ee841f5a7d6892f1fbbfd49 : SUCCESS in 4m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/53d6519a176348d88d84c417194c70df : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/289313d1b34348ae9d4c546c1865cb51 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2d2a0a3f44e45c2a36450c3628d549e : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8ce13734c6843d9ac6d71aad882859a : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ccebe7140247cc819398cba8986e3b : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2d318506aee140eead121bb6edfacd07 : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3e5962ce7c854572a3a29e74743c10bd : SUCCESS in 7m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e3b7b5e987b54d22a8f38ab700f22b9c : SUCCESS in 36m 45s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/92e83670474b41b78296782b444c10f2 : SUCCESS in 58m 51s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1c9d92a12d424442b96106927b0d4235 : SUCCESS in 1h 44m 31s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/582214116361487d8153930aa0cd32e5 : SUCCESS in 20m 06s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4150b26e250c463e888d17adc4a88908 : SUCCESS in 38m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5f4c8334864ff1720c6ff8444cf2900a49828b18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 17:28:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df7dda9825ba25828bd4e945da4b0ee0ada82712","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:14:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/279c4ac3ebd045d98bc5f6d9f7a550a8 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a40616563534c7aa8b5b7dfe4f0ae66 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee437ab2ffc44716bc5b1b45a5344aa4 : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ed93b89f8534687b63c3a63c4be17b4 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb3ac600571445cb463fa0fb3972996 : SUCCESS in 4m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3940e06d3a1a45baa211f8a2a1f72bdb : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/45c4be5c8a204d97a3a78264f1355e62 : SUCCESS in 8m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cdb06045fb2541e6a18ee922b5f97ab4 : SUCCESS in 41m 33s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/a8d3ae1fdb084084b8172e127dddb3d8 : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/9a40a13a2d6246609dd965629fdac241 : SUCCESS in 1h 40m 52s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/dab997c1deb24282baa33130ea883806 : SUCCESS in 23m 11s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/80d4562b1d17431897762cf665ee41dd : SUCCESS in 25m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1ce2b87c4cc7ba600f70a5abb8a41ccba539ae34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:14:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7c0f1b5d268d04e4a3e16d128ac53ec36dfcca61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:17:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f4e817b61e4facb8882ab2ec24b16c : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aea7117764ee3ced0ccb7477c5c962569fa1c926","revisions":{"aea7117764ee3ced0ccb7477c5c962569fa1c926":{"kind":"REWORK","_number":1,"created":"2019-09-16 14:26:54.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/71/682371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/682371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/682371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/682371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/682371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/682371/1"}}},"commit":{"parents":[{"commit":"fc0d0907f7975c9152bf2a34d6f87ce1d56b5817","subject":"Merge \"Fix typo in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc0d0907f7975c9152bf2a34d6f87ce1d56b5817"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-16 14:26:27.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-16 14:26:27.000000000","tz":120},"subject":"Fix after x/kayobe rename","message":"Fix after x/kayobe rename\n\nUpdate .git review and zuul configuration.\n\nChange-Id: I8efee18cf28344a6af44acbe1ca1d4470f7258fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aea7117764ee3ced0ccb7477c5c962569fa1c926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aea7117764ee3ced0ccb7477c5c962569fa1c926"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
