)]}'
{"id":"openstack%2Fkayobe~682408","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I8efee18cf28344a6af44acbe1ca1d4470f7258fe","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"I8efee18cf28344a6af44acbe1ca1d4470f7258fe","subject":"Fix after x/kayobe rename","status":"MERGED","created":"2019-09-16 16:39:08.000000000","updated":"2019-09-17 08:19:31.000000000","submitted":"2019-09-17 08:15:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682408-1568708130290-a663f06a","meta_rev_id":"7e57ecefbf223e932f018be34117de95ef168b83","_number":682408,"virtual_id_number":682408,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-17 08:15:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-09-16 19:24:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-09-16 19:24:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-09-17 04:42:17.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 04:42:17.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-09-17 08:15:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"40aa5f8393b97416ff6b3aa49f874a2fdabb661d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-16 16:39:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fb32b8e37ab138feceb4e7abfc2ceb4ce7a9c85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:13:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9341c9ef27044b4869e349c474e9328 : SUCCESS in 4m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/619d415b20dc4a019f2cc032ae072a66 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/445f7b35afda4a3bafbf5f15842e98b5 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/482e90bdf36a419ca66b71c5009109fe : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbb76068bac04579b4e1e225ec1a6c94 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f039bfa58f1c48a8be724b225f448708 : SUCCESS in 5m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0bf37966d32945d599917f302155841e : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d52124efb554442ea3abe8580541c09f : SUCCESS in 10m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/17238e8dbbca426ea3b1c42b0e75f3fe : SUCCESS in 47m 16s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/3b8291704a234e419347e2bb7d66168b : SUCCESS in 1h 14m 39s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1cede6e3769a44049a3014825a6d170d : SUCCESS in 2h 18m 40s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/82132e763f244f40a2c42e79f7d842d8 : SUCCESS in 20m 08s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5461e307398c4650a6adcf8e32b684c9 : SUCCESS in 26m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9f8928e01b9e825d2c8fffdef40a1f05fecc66e3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-16 19:24:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56f51754dc0f5a943bea3e7fd218501b4cab7b30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 19:24:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ee96e34788b17163f0c6d7683db52d6efe776b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 21:30:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/37fe303a61eb4a3e8225870cd144e08b : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/297eef7682064c6e8dd2c095b2d4a795 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32b42e4453a84dd586e048ecfe12a232 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5b9b5adaf204b76bd3bea920f753bc8 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f1cc097cead4b5ba0fc237c958aa19d : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9761eb410422458aa6435caa8deaa142 : SUCCESS in 3m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0d76e41808714dd6a9b453092ea2d280 : SUCCESS in 7m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d25c014a80434317b06f07adfa8756ea : SUCCESS in 35m 55s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/224cacff6b8e40e1b4876522f3e5d7d3 : TIMED_OUT in 2h 02m 30s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/86657d41939b42b8818a903e463aece6 : SUCCESS in 1h 49m 27s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/08afeaeda52644c28165075da5827a71 : SUCCESS in 29m 33s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/30e2ae42b052493db029a5d83b429dfe : SUCCESS in 34m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e7070c0d7d6c1891d3366a9851d4c91b0e3880b6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-17 04:42:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9331b96b99288d0734df83301164a8e4de6da07a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 06:29:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fde4cd9edc34717a1a0af67c6f929aa : SUCCESS in 3m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f752b6bc90f64c6797bc299373a80c36 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5720c23989ce4f8ba569f7942451c0c3 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4c1d66d0f774dfc9a0659e1178dc70c : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bc47dd3432ef4e71bd339a0533d2d396 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abd63d747ce24b36a8593afb0298a721 : SUCCESS in 4m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3dc16597d42a4761bb28f11cd3a06a41 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c426f6533ab4627a06fd71663051d67 : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/394b4a5f97f8470697b15eb3de55da26 : SUCCESS in 42m 44s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/461f0579ab544258b8c70bec119c6a0d : SUCCESS in 1h 04m 37s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/108b5d03d93a4121bca20fd3189af62f : SUCCESS in 1h 46m 19s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/01fff032435b4af699adfccfe868782d : SUCCESS in 20m 35s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/440fdd5f78cf455d8a4c810f9835475b : SUCCESS in 25m 27s","accounts_in_message":[],"_revision_number":1},{"id":"98a581ea370c15ab6d2a4dedc8e91a65e3d9d70f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 06:30:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48e6f8a540327af3b559ce15cae076dd0f0693b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:15:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3771781d4ea9411696073b49f303595b : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65f1e9863cf0484b8f6eb6bb0571b5ae : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae05a366c2bb416f96fe695603908b2e : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9783ef892fc4b2faa7cda2555e2dce6 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d433870be6084691a7d421b527d63dfd : SUCCESS in 4m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3577329a112f4b04b2e3044cc4af2d12 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d9a120462c414084a66bb91374c98949 : SUCCESS in 8m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5fd763cc990643bb97e9ad01f4e37198 : SUCCESS in 42m 25s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/6071ca05fa7f4089a9e5850c092a05ee : SUCCESS in 54m 33s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/6477788f2ac24115b71a438b80412b41 : SUCCESS in 1h 39m 39s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/899544371623453fa213343101de0973 : SUCCESS in 26m 54s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/74b8927d78a640bc92c31019ef7d9a04 : SUCCESS in 27m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fc508b9209d1e31c8fe212f25bdc25a61f70e983","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:15:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7e57ecefbf223e932f018be34117de95ef168b83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:19:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0df653d27bb741fdb13ecba3f7bffcb1 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"127bd8be6cedf129a1a8f048f3de46964bdcd0ce","revisions":{"127bd8be6cedf129a1a8f048f3de46964bdcd0ce":{"kind":"REWORK","_number":1,"created":"2019-09-16 16:39:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/682408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/08/682408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/682408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/682408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/682408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/08/682408/1"}}},"commit":{"parents":[{"commit":"d3a6c24bc492abfcfba0459a094b390508840f85","subject":"Set provisioning and cleaning networks as non-shared","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3a6c24bc492abfcfba0459a094b390508840f85"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-16 14:26:27.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-16 16:38:57.000000000","tz":60},"subject":"Fix after x/kayobe rename","message":"Fix after x/kayobe rename\n\nUpdate .git review and zuul configuration.\n\nChange-Id: I8efee18cf28344a6af44acbe1ca1d4470f7258fe\n(cherry picked from commit aea7117764ee3ced0ccb7477c5c962569fa1c926)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127bd8be6cedf129a1a8f048f3de46964bdcd0ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/127bd8be6cedf129a1a8f048f3de46964bdcd0ce"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
