)]}'
{"id":"openstack%2Fkayobe~682701","triplet_id":"openstack%2Fkayobe~master~Id8cf5f8e50af438e2d8f0077225c38175e38b662","project":"openstack/kayobe","branch":"master","topic":"wipe-vd-disks","hashtags":[],"change_id":"Id8cf5f8e50af438e2d8f0077225c38175e38b662","subject":"Allow Kayobe to wipe virtualized disks","status":"MERGED","created":"2019-09-17 17:14:08.000000000","updated":"2020-01-28 16:17:33.000000000","submitted":"2019-10-25 13:02:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682701-1572008577250-b61b701d","meta_rev_id":"982984e79aadb85297b4b95a5fa145a65eee1f6f","_number":682701,"virtual_id_number":682701,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 13:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-09-25 15:47:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-09-18 10:11:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-10-25 11:18:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-25 15:47:17.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-10-25 11:18:49.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-25 13:02:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd0d39e2ba8656bf35edecaf8260332211a5fc49","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-09-17 17:14:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fd7ab481ce674bf16aae0ea282a119b11890c68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 18:53:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0132b9619fd24a38aca09a47950a7a5c : SUCCESS in 3m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b6487db4f90141deb277b811d03a96fd : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06362eeeb9094f0fb5e262a03bb5e596 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4496d81cd17451b90547c3c9c5ebd8d : SUCCESS in 7m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e831d1fcf18a460194155b49fd545d96 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56eecb342a01461484393a544cdd3f01 : SUCCESS in 4m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c6062a4176244fb9dde6986378074ef : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ce5230d961c4f0ead4fc15ca7e368f9 : SUCCESS in 7m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a25ee6dd8074ae98239cd201272c4f1 : SUCCESS in 46m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/0ae3e5dd24ef4aee95c6060bb05e8fc6 : SUCCESS in 55m 27s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/abdc28857e1e4b70addbb2f05e53ab30 : SUCCESS in 1h 32m 19s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c5e5e292909b4a9da4f2aef520dc7f54 : SUCCESS in 23m 43s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/95566f2401cc4cb69b82fad756d4574b : SUCCESS in 30m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5123f59b94686d9796dc45b58b8b57695e703462","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-18 10:11:56.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"625a87e3888cf5b30684323619acb2372b5034d3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-09-25 15:47:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f91ab2a95c656b111276d73cd086a0210bbaa17e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 11:18:49.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f43dd0670f8271e80882cbad6407711fb011803","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 11:19:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a81ce9f43f13df63b77a8b4791bad771dd729a64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:02:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b9e8673a842f2ffbf821d63232147ec54785f902","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:02:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/78c1d5a0ad994b2aa57e455a73e6e770 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/388be4c4dc404804932cf02992e704ef : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7116816d364427691dd455c660dbb99 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46865610428f43749971b4a29caa1398 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/809c27fc7045472cb46b9abdb2cd8ab5 : SUCCESS in 8m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b8b06bc686c466a842ebe05ce937f1f : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/83db586d1784400d8cebfaf369b31739 : SUCCESS in 7m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/01dc771c8744485e86621418cc07b94b : SUCCESS in 42m 25s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/6ce24d7007a746d08618958cb931465c : SUCCESS in 1h 04m 35s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5d9a5efc89d349c4a9ed01185237243e : SUCCESS in 1h 42m 34s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c01e455514bf404296eb43868f03d733 : SUCCESS in 20m 30s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ecbd7912dc784b21a15a6247d53705fc : SUCCESS in 35m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a6e58d31f613ecbdc9bf79b716520a6e7f595b47","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:04:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a765c3442b9a4182befcff2bec60b67a : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1},{"id":"425f417ad2326dd1bc48a61da03ddd9bc4ebf35e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 16:17:17.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 57a6d42c100d2aa6006bfdfe707da700d24ab3f6","accounts_in_message":[],"_revision_number":1},{"id":"477c07834904336125ee6f95671d5b3c6963fcdd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 16:17:33.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 8963bb21597bc602d9e502c5751b403c8075ed7b","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea7fd93692a92b3958136673c79f0c2b54bd8fc7","revisions":{"ea7fd93692a92b3958136673c79f0c2b54bd8fc7":{"kind":"REWORK","_number":1,"created":"2019-09-17 17:14:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/682701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/682701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/682701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/682701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/682701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/682701/1"}}},"commit":{"parents":[{"commit":"7321205c0678b42d41d774a1ff14bec9a4a9c8e1","subject":"Update docs after repo rename x/ to openstack/","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7321205c0678b42d41d774a1ff14bec9a4a9c8e1"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-17 17:13:56.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-17 17:13:56.000000000","tz":120},"subject":"Allow Kayobe to wipe virtualized disks","message":"Allow Kayobe to wipe virtualized disks\n\nChange-Id: Id8cf5f8e50af438e2d8f0077225c38175e38b662\nStory: 2002046\nTask: 19692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea7fd93692a92b3958136673c79f0c2b54bd8fc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea7fd93692a92b3958136673c79f0c2b54bd8fc7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
