)]}'
{"id":"openstack%2Fkayobe~684212","triplet_id":"openstack%2Fkayobe~master~I90662a9816c4f4af5a4a481d0e815bf6a398f8d3","project":"openstack/kayobe","branch":"master","topic":"constraints","hashtags":[],"change_id":"I90662a9816c4f4af5a4a481d0e815bf6a398f8d3","subject":"Update the constraints url","status":"MERGED","created":"2019-09-24 07:47:45.000000000","updated":"2019-09-24 09:28:25.000000000","submitted":"2019-09-24 09:26:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684212-1569317217813-844a795a","meta_rev_id":"83f01d7a4d0269c1f6bc3bf9930397da68891f60","_number":684212,"virtual_id_number":684212,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 09:26:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-09-24 08:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-09-24 08:07:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 08:07:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-24 09:26:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c5f6688705e29e737eb1d80a1249de4724c48d8","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-24 07:47:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a494349da030c848831aa690ec895ed2734f156","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-24 08:07:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"393044f4c8620dbcd24f259aeaa10d22ed099abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 08:37:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5315e90ece2d47d38c894e47a0c188f7 : SUCCESS in 4m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8207a890ff104e1e8ee1d91f0bb2c0a5 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c356e14ca39f47ad8f3197a7ca4fe640 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674006cd912f48e3b33f5f83d254bb34 : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f823638adea74e50b50fc01f86f6dcaf : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f30582ceb06a459887a3486bcd90b716 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e02186570426431ba6d7fcb4d5f74447 : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/30bca2895ef7456c83198ac83dd72070 : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/61165a466ab144d38f1985ce039d154f : SUCCESS in 8m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7dddf29ff22642eb8fc1b4d528036652 : SUCCESS in 47m 18s","accounts_in_message":[],"_revision_number":1},{"id":"896bec3ca35762029bb77053b81d9f8868256314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 08:37:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8bc71572cd08f224b83d79688baaf2a1dc7e4462","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 09:26:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9e9c3ecbd7ad440cb78b5bcb16eba79a : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a64a918b1e4444e8f9b3d7f6729b365 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb4617e5d5740b79d2764be1218d265 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4938273304754648975f63a7bc320e16 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57950fd280614bb1a69a704e2d30c0b8 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/713097c06ef64d67b356284ce4d4955e : SUCCESS in 5m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b756de42f9684aa184f26ade12cf9704 : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/31cd7ab9e6ca443d86279ff3b507ce1c : SUCCESS in 9m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eed95a3230b24495b8414ee3b43b7f3b : SUCCESS in 47m 20s","accounts_in_message":[],"_revision_number":1},{"id":"76ebc2c80e05081664429baae0428d6659cdbab8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 09:26:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9944f95c18660d262d6c3b16473ee03a980e10ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 09:28:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1dee43baf754759aa934c22a5482768 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4588aa80e49e4960aadd8812a1d5ce3f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aabf913c9b2928f67f4d2d244abe9e7cfc764fe1","revisions":{"aabf913c9b2928f67f4d2d244abe9e7cfc764fe1":{"kind":"REWORK","_number":1,"created":"2019-09-24 07:47:45.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/12/684212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/12/684212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/684212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/684212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/684212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/12/684212/1"}}},"commit":{"parents":[{"commit":"4b180502a5bcfb9cee5e0c2e267d2dbc53d4b358","subject":"Fix kayobe-overcloud-centos CI job flakiness","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b180502a5bcfb9cee5e0c2e267d2dbc53d4b358"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-24 07:47:39.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-24 07:48:27.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I90662a9816c4f4af5a4a481d0e815bf6a398f8d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aabf913c9b2928f67f4d2d244abe9e7cfc764fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aabf913c9b2928f67f4d2d244abe9e7cfc764fe1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
