)]}'
{"id":"openstack%2Fkayobe~685701","triplet_id":"openstack%2Fkayobe~master~Ia48ec33076b46d2ee186823e8257ba0fc8cdc92a","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ia48ec33076b46d2ee186823e8257ba0fc8cdc92a","subject":"DNM: Test kolla-ansible fix for kolla_toobox","status":"ABANDONED","created":"2019-09-30 12:35:07.000000000","updated":"2019-10-10 13:54:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b2de98859d1ddf1c9456b637f16a8b69b89f109e","_number":685701,"virtual_id_number":685701,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-30 12:46:20.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-09-30 12:43:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-30 12:43:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ec6796359d9b1c9b2ba3f1ac34110b2d2c3260a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-30 12:35:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6725ca57bec15bef63493ac4c7605aad9f6f115b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 12:43:48.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"87e10a7d07437ff2841e1b06e3f74a4fd4c49e9e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-30 12:46:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a559a107493fb2108914bd26b70d2984cb1bd82d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 15:00:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb58519c5b4d4bc383d0f49ef7923d1f : SUCCESS in 4m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/07c1e985e42740e0b57ea07a4bd82e31 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54ee2e2ee1474d55971db4bdda8a3282 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05a2b3335c894235992f05e88bbb7320 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82dedfa832264ca5b89dfc87b7c6655b : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a657d1b0d3ea406a9d4f0b42fdf616cf : SUCCESS in 4m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46e1a834278d4cce894f12c341da8af4 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ffa3e2b5ce74576a36945a488dbab80 : SUCCESS in 9m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/78360fe6a6594c3ea9aa37dec9c560f1 : SUCCESS in 49m 27s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f21dadd66add4612819f485b5a60a4df : TIMED_OUT in 2h 02m 40s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4e7601ec8e804cd99c95c4bf07f0ff46 : SUCCESS in 1h 31m 51s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3f004755123b42ce9fe780ef09d960fe : SUCCESS in 25m 30s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/788279f467a84986b7e989b96c83e0ae : SUCCESS in 44m 39s","accounts_in_message":[],"_revision_number":1},{"id":"b2de98859d1ddf1c9456b637f16a8b69b89f109e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-10 13:54:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25de21ca170eeb8163484e621ff4481e4a774f1d","revisions":{"25de21ca170eeb8163484e621ff4481e4a774f1d":{"kind":"REWORK","_number":1,"created":"2019-09-30 12:35:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/685701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/685701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/685701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/685701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/685701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/685701/1"}}},"commit":{"parents":[{"commit":"aabf913c9b2928f67f4d2d244abe9e7cfc764fe1","subject":"Update the constraints url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aabf913c9b2928f67f4d2d244abe9e7cfc764fe1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-30 12:34:29.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-30 12:34:29.000000000","tz":60},"subject":"DNM: Test kolla-ansible fix for kolla_toobox","message":"DNM: Test kolla-ansible fix for kolla_toobox\n\nDepends-On: https://review.opendev.org/685700\nChange-Id: Ia48ec33076b46d2ee186823e8257ba0fc8cdc92a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25de21ca170eeb8163484e621ff4481e4a774f1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25de21ca170eeb8163484e621ff4481e4a774f1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
