)]}'
{"id":"openstack%2Fkayobe~686141","triplet_id":"openstack%2Fkayobe~master~I75ace6878d36cb523c38a33daa9eecd62826fe43","project":"openstack/kayobe","branch":"master","topic":"custom_inv_docs","hashtags":[],"change_id":"I75ace6878d36cb523c38a33daa9eecd62826fe43","subject":"Fix customised inventory section","status":"MERGED","created":"2019-10-02 11:55:27.000000000","updated":"2019-11-14 11:51:05.000000000","submitted":"2019-11-14 11:49:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686141-1573732182360-fe519b9b","meta_rev_id":"3b16a0509599526b28010fc096c610dc0480db39","_number":686141,"virtual_id_number":686141,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 11:49:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-11-14 10:41:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-14 10:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-11-14 10:41:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 10:31:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-14 10:41:12.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-14 11:49:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5cdb55abe1f54d5007ca2ece44d5467fcb2ba85","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-02 11:55:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4c99af5dfe280bf8274712a480bbec30a683a2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 12:05:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d065a40a3e42bb85676e39cfecc2e8 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff979792aefe43b9a586b7ed1da1b5da : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9e2ab801b93914f287f223f41def355e1aa5259c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-02 12:20:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"12d454fd215ad1c09464856ab276159d31a41f0d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-02 15:10:55.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"eb20b85eb33d00d1bf759fd5deb19c99bcc4cc19","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-02 15:13:24.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2d945d272ef0e9568401277f14db8a9f8e1867b6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-02 15:13:59.000000000","message":"Patch Set 3:\n\nThanks ;-)","accounts_in_message":[],"_revision_number":3},{"id":"8e71b75b721c4cfb5c7f565f2d5e3034b9c028cc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-02 15:14:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"74d778adc3836d239fc00bb1b866acd7555cd96e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 15:21:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e7b01e808c94413b209d8308a2b7aac : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19c04e51b6e5415b9eaf7034221c8a0d : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":3},{"id":"0f91746b6990115fec2a9023a3875cf0e572b314","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-03 08:42:19.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nThanks","accounts_in_message":[],"_revision_number":3},{"id":"35407cf20da3deddc46d781360ce8b12aeafe891","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-03 09:52:56.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"0d9169bbf4cde139bd5ce41f567a0df6b16549fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 10:03:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb9dc679f29e41d09cd0efa7e6bc7047 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e86c478697479983ee3cb9147b005c : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":4},{"id":"026a2c75efe5efe3988b97168e2536932666607b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-03 10:10:40.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":4},{"id":"ab5b69fcbaa33006ac5c4c1f3bfdf58ca5cb9b95","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-03 10:48:33.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)\n\nI think Mark\u0027s wording is easier to understand.","accounts_in_message":[],"_revision_number":4},{"id":"138777bd635a288c25406c4573c335ae7c138942","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 11:18:41.000000000","message":"Patch Set 4: -Code-Review\n\nbump :)","accounts_in_message":[],"_revision_number":4},{"id":"8c5a47eeb1dd891a9aba166a2b882aed456c948d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-13 14:14:37.000000000","message":"Patch Set 4:\n\n@Michal: can you please update this patch?","accounts_in_message":[],"_revision_number":4},{"id":"94d1a05686bbcdefe3345280613c6f51de724bf9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-14 10:16:00.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"472f635e58ea9f9bea80d247b788f6bc5190e800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:22:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/725333f4b3de47a086b9c094ea437187 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81d5be50a45949d383b493eeabf69256 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":5},{"id":"313118d4906486a03943d847930b4825c0bb149a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-14 10:31:43.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"305ac15db9d91a2250dfbefc980f0a366d206fe8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 10:41:12.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6cf9706d0e17d2fe9b3aa3f1c2c0cbfc5c2f9929","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:41:47.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fe4891f000e5556466bb53a0696d2d65f9742c88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 11:49:42.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/097f1a4926b54712bf3c735e86406c83 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fd6850e9a994378a783b9a29fcdd5b7 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":5},{"id":"a492bf6a4d59f2199ac44b026b7dc4c79298bdec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 11:49:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"1516df56cc27ffa39d2737c14d60ae401730d953","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 11:51:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/434706de6ee54419a90c628e23da1b35 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ed8e1a63da62b350547e52232d8142ce14dc7487","revisions":{"639bf32fce87861b8b66c69a1d7fef3498c30bec":{"kind":"REWORK","_number":1,"created":"2019-10-02 11:55:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/686141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/686141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/686141/1"}}},"commit":{"parents":[{"commit":"185e93d6537176062e1719b6847db775fccfcf7e","subject":"Merge \"CI: set previous_release to stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/185e93d6537176062e1719b6847db775fccfcf7e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-02 11:54:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-02 11:55:24.000000000","tz":120},"subject":"Fix customised inventory section","message":"Fix customised inventory section\n\nAdd a note to base customised inventory on a kayobe kolla-ansible\nrole template.\n\nChange-Id: I75ace6878d36cb523c38a33daa9eecd62826fe43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/639bf32fce87861b8b66c69a1d7fef3498c30bec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/639bf32fce87861b8b66c69a1d7fef3498c30bec"}]},"branch":"refs/heads/master"},"1e713b3d19f866b78080712dd1e8b82332648fa8":{"kind":"REWORK","_number":2,"created":"2019-10-02 15:10:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/686141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/686141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/686141/2"}}},"commit":{"parents":[{"commit":"185e93d6537176062e1719b6847db775fccfcf7e","subject":"Merge \"CI: set previous_release to stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/185e93d6537176062e1719b6847db775fccfcf7e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-02 11:54:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-02 15:10:49.000000000","tz":0},"subject":"Fix customised inventory section","message":"Fix customised inventory section\n\nAdd a note to base customised inventory on a kayobe kolla-ansible\nrole template.\n\nChange-Id: I75ace6878d36cb523c38a33daa9eecd62826fe43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e713b3d19f866b78080712dd1e8b82332648fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e713b3d19f866b78080712dd1e8b82332648fa8"}]},"branch":"refs/heads/master"},"0b2bf7fb7b272c949b1add193f831bdecc7a23c8":{"kind":"REWORK","_number":3,"created":"2019-10-02 15:13:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/686141/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/686141/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/686141/3"}}},"commit":{"parents":[{"commit":"185e93d6537176062e1719b6847db775fccfcf7e","subject":"Merge \"CI: set previous_release to stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/185e93d6537176062e1719b6847db775fccfcf7e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-02 11:54:41.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-10-02 15:13:10.000000000","tz":0},"subject":"Fix customised inventory section","message":"Fix customised inventory section\n\nAdd a note to base customised inventory on a kayobe kolla-ansible\nrole template.\n\nChange-Id: I75ace6878d36cb523c38a33daa9eecd62826fe43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b2bf7fb7b272c949b1add193f831bdecc7a23c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b2bf7fb7b272c949b1add193f831bdecc7a23c8"}]},"branch":"refs/heads/master"},"2ce87fa1ce0e96b90b73a47067ff6b322e2352a4":{"kind":"REWORK","_number":4,"created":"2019-10-03 09:52:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/686141/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/686141/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/686141/4"}}},"commit":{"parents":[{"commit":"185e93d6537176062e1719b6847db775fccfcf7e","subject":"Merge \"CI: set previous_release to stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/185e93d6537176062e1719b6847db775fccfcf7e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-02 11:54:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-03 09:52:52.000000000","tz":0},"subject":"Fix customised inventory section","message":"Fix customised inventory section\n\nAdd a note to base customised inventory on a kayobe kolla-ansible\nrole template.\n\nChange-Id: I75ace6878d36cb523c38a33daa9eecd62826fe43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ce87fa1ce0e96b90b73a47067ff6b322e2352a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ce87fa1ce0e96b90b73a47067ff6b322e2352a4"}]},"branch":"refs/heads/master"},"ed8e1a63da62b350547e52232d8142ce14dc7487":{"kind":"REWORK","_number":5,"created":"2019-11-14 10:16:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/686141/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/686141/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/686141/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/686141/5"}}},"commit":{"parents":[{"commit":"185e93d6537176062e1719b6847db775fccfcf7e","subject":"Merge \"CI: set previous_release to stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/185e93d6537176062e1719b6847db775fccfcf7e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-02 11:54:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-11-14 10:15:57.000000000","tz":0},"subject":"Fix customised inventory section","message":"Fix customised inventory section\n\nAdd a note to base customised inventory on a kayobe kolla-ansible\ninventory template.\n\nChange-Id: I75ace6878d36cb523c38a33daa9eecd62826fe43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed8e1a63da62b350547e52232d8142ce14dc7487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed8e1a63da62b350547e52232d8142ce14dc7487"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
