)]}'
{"id":"openstack%2Fkayobe~686188","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I9588b58449e8c8f8eab85ce01b4e130adc150ac6","project":"openstack/kayobe","branch":"stable/rocky","topic":"story/2006475-stable/rocky","hashtags":[],"change_id":"I9588b58449e8c8f8eab85ce01b4e130adc150ac6","subject":"Fix container image build with multiple regexes","status":"MERGED","created":"2019-10-02 15:30:36.000000000","updated":"2019-10-03 15:22:55.000000000","submitted":"2019-10-03 15:22:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686188-1570116175552-0bb3ae39","meta_rev_id":"9205fb99999dca2daa54b9c7c60b219914060768","_number":686188,"virtual_id_number":686188,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-03 15:22:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-10-02 15:33:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-10-03 10:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-10-03 10:12:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-02 15:33:02.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-10-03 10:12:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-03 15:22:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f942f84744280f9eb1eba84fe988070f2a27680d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-02 15:30:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9c5acadb541314ffe9a73e3abb6388ad47295a3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-02 15:33:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27e2e14452d78e5f91d64eab8dab83a450dd83a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:27:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3bc0601484c947208af47f0d9d2e742f : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa733f3c0d546b2b41065849f7786e2 : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14ac5c5af0ae41d89b5eded2320fe4d0 : SUCCESS in 4m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a7eef1c09ab14b03a9411c42c547168a : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba3ccef51ee345e6aa9651c705c3a64b : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d146cbfe78bd429e81d7828818ff50e9 : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac1b5d5dfa6d481694736ec436d7b9df : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9545d158cc0b433a886f21be9c555b64 : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2e65d379cdc547e3b09e86001438ee1c : SUCCESS in 8m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2745469c7eb445beaf79631b55a1cd4a : SUCCESS in 41m 11s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1962efc22e83417d964565beadcf3d4a : SUCCESS in 1h 04m 55s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/03c9d6d79dca4cdb9948b8eee148c9c2 : SUCCESS in 1h 35m 54s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/02af2e74362d41169273a02fb41626c3 : SUCCESS in 21m 47s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/065758e01bad42aa9f492987e86977f0 : SUCCESS in 32m 31s","accounts_in_message":[],"_revision_number":1},{"id":"15d9e9dce83f5e11bbaf6e5d1726f5f5bb4dad21","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-03 10:12:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d8ee43537a82a62d5406ea8b18ad441140dadef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 10:12:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7820de6e4ca51c7fc7b5b02167a816365203a254","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 11:25:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/df953a062c8d4615a3b24f0a220423fb : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed9d15f9c1704b0e852291ad13414c95 : SUCCESS in 3m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f91130cd7e004e1db5f74faf4c693ec6 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d18477939496446f8c41daa345b695b7 : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/247258db1006480bbf17f6097ca368e8 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ae95941d35e4d77a3e29c4915ebd156 : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fcc3a555e99e4665a79f7e5ec8d3f0ab : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b0edae5ac255405ca554b3b896729101 : SUCCESS in 7m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/72ed800742d04fa9b5fe7bfcd36637f4 : SUCCESS in 39m 32s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/190966ff36b14937807a6ac54219678e : SUCCESS in 1h 07m 58s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4b34c35aa4c447108ac212f2aa1f346d : FAILURE in 27m 50s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8001db6358bd47b5afbc02e41f893692 : SUCCESS in 18m 41s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/115a8380fcf84e12b9a22fd7349dd874 : SUCCESS in 31m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d5b083d37d973f65de443469b543ccafa42cf2c8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-03 11:30:11.000000000","message":"Patch Set 1:\n\nrecheck\n\nweird timeout","accounts_in_message":[],"_revision_number":1},{"id":"e7e60abbbf7b00aabcf36105e0845b696315c39d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 13:26:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/063910d05183465fb0fc22d9eb4f104b : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1950fdcfc06e40e289900ee3eec0906d : SUCCESS in 3m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25346b0cc0334202a940e3da11a4e884 : SUCCESS in 4m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/001a293a75e84cfabb4322286158d9d3 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21e071a441dc47ad85aefa2bfe91ed50 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/545b1cb6759743e09c8ecc7cdf52c43d : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e88741a2d97a4eed8d21c090cc979b57 : SUCCESS in 3m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c42f589033c448c0896ef62f7b79af95 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9783fa98178b4ef496489986987f4c59 : SUCCESS in 7m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/708091b8933141a6b4edbd786687b936 : SUCCESS in 34m 34s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/3aa9fc2f3ddf4e9f84c90976b42408fa : SUCCESS in 1h 13m 48s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5290984e86e9466d95013ffa55863032 : SUCCESS in 1h 51m 11s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0aa15e0685ee40c8abb702f79c6fab99 : SUCCESS in 18m 31s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/52eda27ec69c4516bddc919def3eabf2 : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":1},{"id":"559159f8f29b0c495203ad767640cbc7e053fc5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 13:26:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf3f13d1d9e7da520203cbf8d5fea2f8034ebd7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 15:22:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"746063772d5f6297871d4648a47c9f331c377f2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 15:22:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/11d413a5932e4f02926d239e66869350 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c88af580d2844ee1bf00cc3f9c7e2919 : SUCCESS in 3m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/145d1befa54f4558b3427541d65dc5fd : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c99ddf552c47dc9e9c4988b1989bae : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/80a35bb20972452bba754ded01e3c714 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13f6324355fe41b1838972187644df97 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a2d8b6bd9f2c4bf485864dec0871c6b0 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eea6d9b7715e4d7193129bc1148ae89c : SUCCESS in 7m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e827ddbd4b504f46ab96f01c9ddf9be1 : SUCCESS in 33m 29s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/4f200e3576934075a47972bac17312c4 : SUCCESS in 58m 10s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4040e246235941ef9d630e5a78a46c1e : SUCCESS in 1h 50m 11s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/04e1683fd8984657ac1b611fd86c9954 : SUCCESS in 24m 46s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e9adacd1791c4d07a0641a66fd59939f : SUCCESS in 35m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"262d856071f95466deb9fd4b9ca00364ce828047","revisions":{"262d856071f95466deb9fd4b9ca00364ce828047":{"kind":"REWORK","_number":1,"created":"2019-10-02 15:30:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/88/686188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/686188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/686188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/686188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/686188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/686188/1"}}},"commit":{"parents":[{"commit":"1e4c002aaff77e87b4ab7643704c07903200267c","subject":"Fix after x/kayobe rename","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e4c002aaff77e87b4ab7643704c07903200267c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-11 14:30:33.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-10-02 15:30:25.000000000","tz":120},"subject":"Fix container image build with multiple regexes","message":"Fix container image build with multiple regexes\n\nWe don\u0027t need to provide additional quotes for passing these arguments\nto playbooks, since this is now done automatically via\nkayobe.utils.quote_and_escape.\n\nChange-Id: I9588b58449e8c8f8eab85ce01b4e130adc150ac6\nStory: 2006475\nTask: 36416\n(cherry picked from commit 1e5fb0075482f2e971a2b8f2e4c6a873b6b12afd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/262d856071f95466deb9fd4b9ca00364ce828047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/262d856071f95466deb9fd4b9ca00364ce828047"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
