)]}'
{"id":"openstack%2Fkayobe~686361","triplet_id":"openstack%2Fkayobe~stable%2Frocky~Ie59f0c729381e5bea112ab7f5d0c738a9488b772","project":"openstack/kayobe","branch":"stable/rocky","hashtags":[],"change_id":"Ie59f0c729381e5bea112ab7f5d0c738a9488b772","subject":"Set provisioning and cleaning networks as non-shared","status":"MERGED","created":"2019-10-03 09:17:24.000000000","updated":"2019-10-18 18:26:54.000000000","submitted":"2019-10-18 18:26:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686361-1571423214627-8b4ac2a1","meta_rev_id":"dfe0c46e23ef5b6a8fc9628c3bed42e146d12dbd","_number":686361,"virtual_id_number":686361,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 18:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-10-03 09:23:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-18 16:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-18 16:31:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 09:23:33.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-10-18 16:31:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-10-18 18:26:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"532fb1a9d71964c60dae164ae443101334214537","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-03 09:17:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c565ec799090b6363e4d37f4ea6c70eb82f4864f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-03 09:23:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bd59aa969a145c60b7d2be0da22790ea197f57f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 11:06:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7946867cc7e14733a3755c019aed64ee : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d6f135bf8e146dfbb155879a0eba00a : SUCCESS in 4m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14f952b90af047589b4e2c22a6ca132a : SUCCESS in 4m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/37eb531bdd714aba978ea71f1dab8fc0 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6181dea29244f968c673c0e410115b3 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e6bf6f270824561b7825fb6554df8b0 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17b5d315ba9443f3a662b8db5f9ff889 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c82b6e761bc14af4b43c5dab7129d056 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71337945a4bc49ff8eda8f460ad25b63 : SUCCESS in 7m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5a19264944684ff3a067034d902b2614 : SUCCESS in 39m 47s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/6dda826ca260450387515b2c8f16d6f0 : SUCCESS in 1h 10m 51s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b7fbf2e756e84c2bb1d8b300933e52ee : SUCCESS in 1h 42m 55s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/17f4068a93234829b7ca1cb88e331f7e : SUCCESS in 23m 09s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d7979b07af9547b4a8616c22ad156410 : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":1},{"id":"8c03040c8ec446bd3973b6d20bc01aba59aeb16e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-18 16:31:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb775b9e5646b06ec12dd630242081043c956ad0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 16:32:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5350a976f67ffac967a3e6d86d8dc52f723673ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:26:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dfe0c46e23ef5b6a8fc9628c3bed42e146d12dbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 18:26:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/38b3b7ed228f44908ba4f4f5b3e4707b : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a17936eb5d741a8aac857901aafefdd : SUCCESS in 3m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/38c8cf77882e427d9f4892f417d630a5 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49211e6c8b2c4229928783388a86aea2 : SUCCESS in 2m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5feb7d4926864006ab211a826d32d9a1 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d84d55654f034672893a30b05b76e434 : SUCCESS in 3m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b9919f6979b64b12a9d3551ccea268e9 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/df4d8193ba27457aa2ca99a55ec8e7c9 : SUCCESS in 7m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/baab3e2be2ec43df9131ed0ae79109a0 : SUCCESS in 32m 57s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/3408fbbe75024b6f84a79ef63791ed8e : SUCCESS in 1h 04m 38s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/ea831f3bdfc3447f96cfb614573cc11d : SUCCESS in 1h 48m 56s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e6bbceb4943640bdafc84e112b1af832 : SUCCESS in 21m 45s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/63c76b0c23494b07b8b9cd63599525a0 : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21c2970c756c7a690c3be8df9b8c45de53002489","revisions":{"21c2970c756c7a690c3be8df9b8c45de53002489":{"kind":"REWORK","_number":1,"created":"2019-10-03 09:17:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/61/686361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/686361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/686361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/686361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/686361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/686361/1"}}},"commit":{"parents":[{"commit":"b01290ed471dcea021657d6a51efd1904bf0a5e8","subject":"Merge \"Add alternative tenks deploy and teardown entrypoints.\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b01290ed471dcea021657d6a51efd1904bf0a5e8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-16 11:08:25.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-03 09:17:16.000000000","tz":60},"subject":"Set provisioning and cleaning networks as non-shared","message":"Set provisioning and cleaning networks as non-shared\n\nModifies provisioning and cleaning networks in multi-tenant ironic\nenvironments to be non-shared. Flat networks remain shared.\n\nTo apply the change to an existing environment, run \u0027kayobe overcloud\npost configure\u0027.\n\nChange-Id: Ie59f0c729381e5bea112ab7f5d0c738a9488b772\nStory: 2006409\nTask: 36295\n(cherry picked from commit f91cf43876afa967bcf26c133c7973f84f02341f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/21c2970c756c7a690c3be8df9b8c45de53002489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/21c2970c756c7a690c3be8df9b8c45de53002489"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
