)]}'
{"id":"openstack%2Fkayobe~686755","triplet_id":"openstack%2Fkayobe~master~Ice26a113f9023b1a25a4af0552d07c5e95e64556","project":"openstack/kayobe","branch":"master","topic":"story/2006662","hashtags":[],"change_id":"Ice26a113f9023b1a25a4af0552d07c5e95e64556","subject":"Fix baremetal serial console autoenable","status":"MERGED","created":"2019-10-04 15:52:36.000000000","updated":"2019-10-31 11:14:07.000000000","submitted":"2019-10-28 14:50:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686755-1572274229195-cc9bba01","meta_rev_id":"142335795f638e5e268f7aff30e9d3649ce53c60","_number":686755,"virtual_id_number":686755,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"date":"2019-10-25 15:17:05.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-28 14:50:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-10-21 13:37:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-28 09:29:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-28 09:29:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 13:37:22.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-10-28 09:29:54.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-10-28 14:50:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f5e3985bcac45cbae20d92f30e2e862fac1fc79","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-04 15:52:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"120fe478aa961ee5e14da8d2c7bb8fff3c9d9e74","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-04 16:36:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7ffb56e59590fafaef230bd12e52e2e484f20fae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 18:04:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09ca4fef40f345acaf117c5143c0989b : SUCCESS in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7283c0a7c9ac4b5289aa9b4b4ef40f7d : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efe1b00d9b964e3486cc49a108187d49 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00b89c057a6f4492ac8be4ece967450e : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a66c65549114676a5f8d3c3930e572c : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f79a54df4b44dcfafc8f2e2b36f9106 : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5e7470317f9f4046b5b4e85d22adf637 : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/620a1de3841e4229846c71cb333d63b4 : SUCCESS in 8m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/03f9ebc406d54ab6966e036dd6215c98 : SUCCESS in 42m 57s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5916479f727b43858d23a8a4b7d0c0c4 : SUCCESS in 1h 14m 32s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5ccc9e53fc764ebcabc467e1cee65706 : FAILURE in 2h 08m 51s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/6e2e7ffd8dbc4e1bbbad4b78df8d6aea : SUCCESS in 23m 04s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/23714e266a2e429d96fe147ed2d83c99 : SUCCESS in 34m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6dad50b3f7dc4162190810b6326ad11bef9ced55","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-07 09:33:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2f15ae738bf7ca7e593c0785dfd6d7cda119119f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 11:41:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d126c0928314c9c93dc48a1619ddabc : SUCCESS in 4m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/02a91bdffd0145be96350466ed1e80d4 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a717d742e8254f19b46e429f74af29f2 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21c13fb1a07545c3a7884c36e4adde54 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb3c8ae01d0142c0b2f7d22498c9ae93 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2f0f6edc734aae83e9a4ee356409e6 : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cf28ab52bc7542b481adddbe46a70fbb : SUCCESS in 3m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6080ed85ac5040bc8ac14be05ad22606 : SUCCESS in 8m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c86c648cf954b05bb485e016243c401 : SUCCESS in 44m 20s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ec4491a51e20401f8c1b96207479468a : SUCCESS in 1h 13m 49s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d1b1f9a4622449ff805a748c79ca2032 : SUCCESS in 2h 03m 17s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/74d19c62bf3e41cf8e1a3f9787faa95e : SUCCESS in 22m 20s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/690aaea20d1543f9ae660b70d99d917c : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b947d6d9dea24999329f4a625d95acb93c893e18","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-18 14:42:49.000000000","message":"Patch Set 1: Code-Review-1\n\nLooks like serial console autoenable was broken without this fix. Should we add a release note and backport?","accounts_in_message":[],"_revision_number":1},{"id":"0c00d9baf5280ce4a68cbcbb428cac37996ac489","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 13:16:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9d3c9abf57cc5a8dd659cdf97d83c8ad9d20c673","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-21 13:37:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"106452529b86433a399547cffb75fe89d8bfb5b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:34:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25ad3a804a2044c9a3bca20fa0cea9c0 : SUCCESS in 3m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/93979c2f39474f81879ed5fc34934b7b : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c68a569f30524cab99d1661015f0d84d : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d3776cdefe24cfaa245d80af6f2455b : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02dafba2cfe34ffc88215dda049e1690 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b722ae3c4ad40a28b627d5b12579c9b : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1a29f40558041949dbfbbad720f3a23 : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6f0e581455de4993917689d4e45c28c0 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/05c0b26e86ae411aaeb66f530e2a6076 : SUCCESS in 8m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5ad3ceb5c18b4feb905a8a6e580c3359 : SUCCESS in 37m 51s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/4dd638417f2a4252b8fb4c62f1b30249 : FAILURE in 17m 35s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/6d036343144e4b85882a1dfc656b3acf : FAILURE in 1h 12m 48s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f95fda765aec4ad4a8e6346d9e9dcfac : FAILURE in 18m 30s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8c54cff7860c4cf7b12d51a14d9e9bf8 : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":2},{"id":"0e822ff1b1da25e952ded8321c8d6db832bc2de1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 14:46:27.000000000","message":"Patch Set 2:\n\nCI failed due to https://review.opendev.org/#/c/687017.","accounts_in_message":[],"_revision_number":2},{"id":"787e2ea31e7ff0f26f2630b1ac9ecafac07b5cf7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 11:17:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b5bba4a034fd169e7ec853f4eeb308f98481e21b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:51:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7451fa2eb57743fb99ee46260f530157 : SUCCESS in 5m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3983f75110d24c5ea9fbe3ade82bd6b1 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6bc5dd977d849928aa5d79bf52ff15a : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5762c1b4bf400594ca2fb6910199e8 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c6561cfaa7340b0b4d99979b9191347 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18896925732548a2989e63ef2e83076b : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9014249ef7242f899b16430a15fedab : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8fbee65f000b4269935105d07be27172 : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d5b0ef1160034a0b935a8b3b5283891d : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f56d36c6d90e4aaa988edfe694b6c43e : SUCCESS in 37m 15s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/49672ebc2d2e46618f6f5b2eb3f6f834 : SUCCESS in 1h 19m 21s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d67084df1fa64877a76af3fd878fcfff : TIMED_OUT in 2h 32m 06s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/960f15b94ed04d308148f34ede241a2e : SUCCESS in 20m 42s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/070677f987f44728a7c9e93b5ca12723 : SUCCESS in 33m 07s","accounts_in_message":[],"_revision_number":2},{"id":"5f5c4c29ecfed190f8645d06a0bc7f602eb27bc1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 15:17:05.000000000","message":"Patch Set 2:\n\nrecheck\n\nironic deploy failed after upgrade","accounts_in_message":[],"_revision_number":2},{"id":"2be00215fcf327dab7e49d151dba89b16fc7437d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:33:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5dc93929f734faca42e0415ee98f3b4 : SUCCESS in 5m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e036bb029b6c401bb457075b8b3b736e : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d7b2c1cb02e4431bb2d68c5700d816e : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe846c2444bf48d29a4ef46f81669fe2 : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ebd98623bca849ba906e638e848c6ec1 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f38b93e8b43474790e1214b452e6ca0 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fd31ded48e54e6c88e09901f621a75f : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7e4901a457fe4bd1b752fa5c978e3262 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7730702fc33c4d67b8c7a16225992ba2 : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a54d35d8f456497d97166bce1442d157 : SUCCESS in 43m 55s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e8ccf4b8523041d398dfafae8a7aaf93 : SUCCESS in 1h 10m 16s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/0275e1c263864a87a15b9fe08d625177 : SUCCESS in 2h 10m 52s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1f2e59037e934231a7cb39080067d8ee : SUCCESS in 23m 28s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/9d521d90758b4cd7ae371e8c149ed9f3 : SUCCESS in 35m 50s","accounts_in_message":[],"_revision_number":2},{"id":"515f5770f0005636207d439ab41b765ed7f23fdd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-28 09:29:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a434dde3e6d805a83810fd9f1a1063084eda9409","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 09:30:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1969a62da01f8d3b39149ebc200514e2f61d0dde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 11:21:34.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1d31be342df249418aedfaebe00a29d4 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b03e6011e2044b49b68bdc81eabbcca : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/519296913c6b4433be1961015039770f : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b90f739cffc463d9fc16251e24499c8 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6383fdebb38149c3a2d6ed4381b755be : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3edfbb9240484ae9be549b74af3793e2 : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca11bb9949e1401fbd7fbfbae1f16530 : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/218b0def861549e9918aedd9716e8fb3 : SUCCESS in 9m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9badf672fa5446a48a693672e36ae315 : SUCCESS in 39m 44s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7079d91f359f41eeb2413705208f90d7 : SUCCESS in 1h 11m 29s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/58bacfa458db465ab8aa28f01be199ee : SUCCESS in 1h 49m 44s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/26e0b949e73447a8af52649d720fb27a : SUCCESS in 21m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e7ae6b7ad6054be1a8d52ba6cce7cabc : FAILURE in 26m 39s","accounts_in_message":[],"_revision_number":2},{"id":"d83a7de11c1d04b85b746dd1b05818c05876e1b0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-28 11:23:49.000000000","message":"Patch Set 2:\n\nyum mirror timeout\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f79c139b8b1d710d2fdf117a0f2039d77050b84a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 12:59:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ad7248c042d4bd4823c4cfbfbad69ef : SUCCESS in 4m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/95658cf6cea94901ae1e91cbb211950d : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd71c84cbdcc48c0a6efb20e8bfdf411 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5f3bba743834f0e836f242306782971 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7fd6934c4c44267b628fa42a69ea402 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13fdfaf375a648de9b3a17844dfab96b : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84d45014da4645f6ab33337a349e41b4 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6ab5e9bd8f91494a8d6b5520b6ed7381 : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c8e6751986db436fb9a9546cca198b67 : SUCCESS in 7m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4fcceec6a42843b585db3d9ee9aed4bb : SUCCESS in 30m 11s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9b2c0a4345ee484d8629020aa3025009 : SUCCESS in 1h 02m 00s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/9a17dc93493c49418ce876f312681172 : SUCCESS in 1h 34m 22s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/69260b9954894b49b948f37901726fff : SUCCESS in 18m 45s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a323eedd04fe49b8a75589fc9d11fd67 : SUCCESS in 31m 10s","accounts_in_message":[],"_revision_number":2},{"id":"848f24c69972ba7250e7d728af9904e3e073e638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 12:59:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fd907d7271e013ff4729b8d931f09c9e671a3b31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 14:50:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f89232e653d647138eaf6e86b511dc11 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2eb00743e4984340bf485008c1544bd7 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d22dacab9c42498ae3f1463857b69e : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75bcdb1356c244d9bc54ef0b6aef7783 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/769808ced88e4c6d87c9f43aff33f58d : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d3df19bca4c47e6ac14b1015f26d4b6 : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/55b15e56c70a4f0ca62058977aa723fe : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ec6d75fcb0548988452a3ae4e14ae1c : SUCCESS in 7m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4bb5b00d6054124b459ee7861b4fa96 : SUCCESS in 31m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/211114acc7a44833ab399185165c0e6e : SUCCESS in 1h 07m 53s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/e360bc31dce44778a7ebfc2767f07816 : SUCCESS in 1h 48m 59s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/30c1e392da4c43dfa72633e6b187d7c0 : SUCCESS in 19m 20s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/47638feccf1046f78d91f9dd5ab188d7 : SUCCESS in 33m 17s","accounts_in_message":[],"_revision_number":2},{"id":"fbc508d81b408be814ad0a4142aaaaf05a4b4e3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 14:50:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"43aa8e0cacaf78dc9935cc8fc7a8239a49cc4e3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 14:51:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67de69f6bdf14b609f240fd6e06a25a3 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b911105169154faab5ff11c578b5d00f : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2},{"id":"8498826e1730656559b113d4546945699c7aee12","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-31 11:13:49.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 1451f67df23934e88190f178027903a2474815dd","accounts_in_message":[],"_revision_number":2},{"id":"a716c86c341cf2f4a15b0773976a134cad3b8773","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-31 11:14:07.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit d90bdaeb4466da58dec39db1e2dfeaf01cffb174","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5977a4411921c1880ee023b945ac2cceb533749a","revisions":{"fe3fc72e2fa5176e0cb2d031634024333bd92827":{"kind":"REWORK","_number":1,"created":"2019-10-04 15:52:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/55/686755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/686755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/686755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/686755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/686755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/686755/1"}}},"commit":{"parents":[{"commit":"886fe4fd7694c365c13c0063fb017ce0425a6664","subject":"Merge \"CI: Remove rocky pin from molecule requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/886fe4fd7694c365c13c0063fb017ce0425a6664"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-04 15:49:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-04 15:52:15.000000000","tz":60},"subject":"Fix baremetal serial console autoenable","message":"Fix baremetal serial console autoenable\n\nIf ironic_serial_console_autoenable is set to true, ironic compute node\nserial consoles are automatically enabled during kayobe overcloud post\nconfigure. However, the baremetal-compute-serial-console.yml playbook\nexpects a variable called cmd to be defined, and set to \u0027enable\u0027 in\norder to enable serial consoles. This patch fixes the issue.\n\nWe also add a conditional to only import the playbook when ironic is\nenabled in the overcloud.\n\nChange-Id: Ice26a113f9023b1a25a4af0552d07c5e95e64556\nStory: 2006662\nTask: 36924\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fe3fc72e2fa5176e0cb2d031634024333bd92827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fe3fc72e2fa5176e0cb2d031634024333bd92827"}]},"branch":"refs/heads/master"},"5977a4411921c1880ee023b945ac2cceb533749a":{"kind":"REWORK","_number":2,"created":"2019-10-21 13:16:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/55/686755/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/686755/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/686755/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/686755/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/686755/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/686755/2"}}},"commit":{"parents":[{"commit":"886fe4fd7694c365c13c0063fb017ce0425a6664","subject":"Merge \"CI: Remove rocky pin from molecule requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/886fe4fd7694c365c13c0063fb017ce0425a6664"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-04 15:49:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-21 13:16:20.000000000","tz":60},"subject":"Fix baremetal serial console autoenable","message":"Fix baremetal serial console autoenable\n\nIf ironic_serial_console_autoenable is set to true, ironic compute node\nserial consoles are automatically enabled during kayobe overcloud post\nconfigure. However, the baremetal-compute-serial-console.yml playbook\nexpects a variable called cmd to be defined, and set to \u0027enable\u0027 in\norder to enable serial consoles. This patch fixes the issue.\n\nWe also add a conditional to only import the playbook when ironic is\nenabled in the overcloud.\n\nChange-Id: Ice26a113f9023b1a25a4af0552d07c5e95e64556\nStory: 2006662\nTask: 36924\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5977a4411921c1880ee023b945ac2cceb533749a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5977a4411921c1880ee023b945ac2cceb533749a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
