)]}'
{"id":"openstack%2Fkayobe~687240","triplet_id":"openstack%2Fkayobe~master~I3caa80da7c44704a3b6559d98f928d95e83289e5","project":"openstack/kayobe","branch":"master","topic":"docs-tags","hashtags":[],"change_id":"I3caa80da7c44704a3b6559d98f928d95e83289e5","subject":"Add ansible tag headers to Host Configuration doc page","status":"MERGED","created":"2019-10-08 08:34:15.000000000","updated":"2019-10-25 16:59:58.000000000","submitted":"2019-10-25 16:57:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687240-1572022679961-34f7ffe6","meta_rev_id":"8de56753aaeb1e1a02a54feb6f19654e9114cac2","_number":687240,"virtual_id_number":687240,"owner":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 16:57:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":2,"date":"2019-10-09 08:37:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-25 16:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-25 16:48:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 08:37:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-25 16:48:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-10-25 16:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04f039935b8f7ce27db7347733a776654461bb55","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-10-08 08:34:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fa52a1a17940ab80590214cbf65a683ca0da4cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 08:44:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93bd99279f874f4eb2f87d5a9751572a : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dee5042ebcd453eb94ac8dfcb116be9 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"57397e0d958b61c90468e5e5c3d6a55b73010cf7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-08 08:46:03.000000000","message":"Patch Set 1:\n\n(6 comments)\n\nLooks like a good start.","accounts_in_message":[],"_revision_number":1},{"id":"5dbd9149b31945ec729e4922c1fbb7b902fc0020","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-10-08 10:16:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"297e1e40c62f8bb1054cbe0d8b93fc6376713e15","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-10-08 10:19:54.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5a02ffe985c8ee66853d9d1b8f2d93a9ea33a6aa","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-10-08 10:22:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"912a2786f6124fe0f334f09831b16f244c372603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 10:50:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be1d8eb2c84e4031aae5e2622299ff69 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c46a79514dc4ac7bf2f1c8922fbcc27 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":4},{"id":"fc168e5a6cc34db9ae39e1d242ea9e084f185281","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-08 13:57:18.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fc8731f6aec92957c7bcb0c4a89212c456e201f3","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-10-08 14:50:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"de40a08afacf5e5f0b34385db4e8be02258ac707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 15:01:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb02fd5bce92443bb6f23847c278797f : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c136ad24936d483fb0671d55d42be4ff : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":5},{"id":"0e0d30a3508e61dccd5b3a65d353da6e2b9db178","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-08 15:56:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f9d6f6d79453eb19f76ddcff69d0de01a41d9c41","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-10-09 08:33:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d3528d85945ec465294ab3970c4f632d8438fcf7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-09 08:37:24.000000000","message":"Patch Set 6: Code-Review+2\n\nGreat, thanks","accounts_in_message":[],"_revision_number":6},{"id":"362ef3471614d44b56661b095992217d19e9fe9b","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-10-09 08:39:50.000000000","message":"Patch Set 4:\n\n\u003e Great, thanks\n\nCheers for the help!","accounts_in_message":[],"_revision_number":4},{"id":"a43f1ca668bbd4e9d142de369f0dc0538119a573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 08:42:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c9f5578bac64e54b646647d5732c891 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/432b128dab2446ff9c97048bdcb14ba0 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":6},{"id":"81fc007a284f33f112fd0fac423b846082916cfc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-25 16:48:27.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"26ae15272691d66fe90e9e30aef8042358519b7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 16:48:38.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"29755561498fec32da84508f21e416cd01229eb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 16:57:59.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b89c00e7456d41658654583547fedd30 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d44f76b65ea443649cce25cada0d0b7a : SUCCESS in 9m 03s","accounts_in_message":[],"_revision_number":6},{"id":"1a990cf24b83203baaede5401b0aa6d98a3532bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 16:57:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"405dac0c9d0899c1b4da90c52319edd423598635","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 16:59:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b32001785fd04b779657a7e421f5ad67 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ee72de447c913e8d843dd94fe7d5facefd0bb91d","revisions":{"dc6c66b840fda4bb8a099cc358f6f29f0cf994ca":{"kind":"REWORK","_number":1,"created":"2019-10-08 08:34:15.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/40/687240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/687240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/687240/1"}}},"commit":{"parents":[{"commit":"886fe4fd7694c365c13c0063fb017ce0425a6664","subject":"Merge \"CI: Remove rocky pin from molecule requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/886fe4fd7694c365c13c0063fb017ce0425a6664"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-10-07 22:24:40.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-10-08 08:33:12.000000000","tz":60},"subject":"Add ansible tag header to Host Configuration doc page [WIP]","message":"Add ansible tag header to Host Configuration doc page [WIP]\n\nChange-Id: I3caa80da7c44704a3b6559d98f928d95e83289e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc6c66b840fda4bb8a099cc358f6f29f0cf994ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc6c66b840fda4bb8a099cc358f6f29f0cf994ca"}]},"branch":"refs/heads/master"},"4bf219bc678a51832de2d1156e3368738e335107":{"kind":"REWORK","_number":2,"created":"2019-10-08 10:16:27.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/40/687240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/687240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/687240/2"}}},"commit":{"parents":[{"commit":"886fe4fd7694c365c13c0063fb017ce0425a6664","subject":"Merge \"CI: Remove rocky pin from molecule requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/886fe4fd7694c365c13c0063fb017ce0425a6664"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-10-07 22:24:40.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-10-08 10:15:14.000000000","tz":60},"subject":"Add ansible tag headers to Host Configuration doc page","message":"Add ansible tag headers to Host Configuration doc page\n\nChange-Id: I3caa80da7c44704a3b6559d98f928d95e83289e5\nCloses-Bug: #2006212\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4bf219bc678a51832de2d1156e3368738e335107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4bf219bc678a51832de2d1156e3368738e335107"}]},"branch":"refs/heads/master"},"8d12e0eabd3b211518d5732c58dc01cd46540a43":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-08 10:19:54.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/40/687240/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/687240/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/687240/3"}}},"commit":{"parents":[{"commit":"886fe4fd7694c365c13c0063fb017ce0425a6664","subject":"Merge \"CI: Remove rocky pin from molecule requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/886fe4fd7694c365c13c0063fb017ce0425a6664"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-10-07 22:24:40.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-10-08 10:18:23.000000000","tz":60},"subject":"Add ansible tag headers to Host Configuration doc page","message":"Add ansible tag headers to Host Configuration doc page\n\nChange-Id: I3caa80da7c44704a3b6559d98f928d95e83289e5\nStory: #2006212\nTask: #35802\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d12e0eabd3b211518d5732c58dc01cd46540a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d12e0eabd3b211518d5732c58dc01cd46540a43"}]},"branch":"refs/heads/master"},"376b66325de84779df21dca636b161f8410b8c19":{"kind":"REWORK","_number":4,"created":"2019-10-08 10:22:29.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/40/687240/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/687240/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/687240/4"}}},"commit":{"parents":[{"commit":"886fe4fd7694c365c13c0063fb017ce0425a6664","subject":"Merge \"CI: Remove rocky pin from molecule requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/886fe4fd7694c365c13c0063fb017ce0425a6664"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-10-07 22:24:40.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-10-08 10:22:19.000000000","tz":60},"subject":"Add ansible tag headers to Host Configuration doc page","message":"Add ansible tag headers to Host Configuration doc page\n\nChange-Id: I3caa80da7c44704a3b6559d98f928d95e83289e5\nStory: #2006212\nTask: #35802\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/376b66325de84779df21dca636b161f8410b8c19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/376b66325de84779df21dca636b161f8410b8c19"}]},"branch":"refs/heads/master"},"8621cfe2e33773a806a305fae04cb95c9d525c58":{"kind":"REWORK","_number":5,"created":"2019-10-08 14:50:12.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/40/687240/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/687240/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/687240/5"}}},"commit":{"parents":[{"commit":"886fe4fd7694c365c13c0063fb017ce0425a6664","subject":"Merge \"CI: Remove rocky pin from molecule requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/886fe4fd7694c365c13c0063fb017ce0425a6664"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-10-07 22:24:40.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-10-08 14:50:02.000000000","tz":60},"subject":"Add ansible tag headers to Host Configuration doc page","message":"Add ansible tag headers to Host Configuration doc page\n\nChange-Id: I3caa80da7c44704a3b6559d98f928d95e83289e5\nStory: #2006212\nTask: #35802\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8621cfe2e33773a806a305fae04cb95c9d525c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8621cfe2e33773a806a305fae04cb95c9d525c58"}]},"branch":"refs/heads/master"},"ee72de447c913e8d843dd94fe7d5facefd0bb91d":{"kind":"REWORK","_number":6,"created":"2019-10-09 08:33:37.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/40/687240/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/687240/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/687240/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/687240/6"}}},"commit":{"parents":[{"commit":"886fe4fd7694c365c13c0063fb017ce0425a6664","subject":"Merge \"CI: Remove rocky pin from molecule requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/886fe4fd7694c365c13c0063fb017ce0425a6664"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-10-07 22:24:40.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-10-09 08:33:26.000000000","tz":60},"subject":"Add ansible tag headers to Host Configuration doc page","message":"Add ansible tag headers to Host Configuration doc page\n\nChange-Id: I3caa80da7c44704a3b6559d98f928d95e83289e5\nStory: #2006212\nTask: #35802\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee72de447c913e8d843dd94fe7d5facefd0bb91d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee72de447c913e8d843dd94fe7d5facefd0bb91d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
